)]}'
{"id":"openstack%2Fmanila-tempest-plugin~862787","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id7e4bf5fb9aaf2dc32a7bebac3a29fb730015f70","project":"openstack/manila-tempest-plugin","branch":"master","topic":"remove_alt_client","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-10-31 19:09:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-10-27 18:06:58.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-10-31 20:45:25.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-10-31 20:45:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7e4bf5fb9aaf2dc32a7bebac3a29fb730015f70","subject":"Remove \"alt_project_share_v2_client\" client initialization","status":"MERGED","created":"2022-10-27 08:44:42.000000000","updated":"2022-10-31 20:45:25.000000000","submitted":"2022-10-31 20:45:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862787-remove_alt_client","meta_rev_id":"04532da5e06e87d701e60e47c1ecea34fbf92762","_number":862787,"virtual_id_number":862787,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-31 20:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-10-31 19:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-10-27 18:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-10-31 19:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 10:30:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-10-27 10:30:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-10-27 10:30:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-10-27 10:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-27 13:48:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-27 14:10:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"c23cd09032cfff1a2e0e061b857f6ed884a5b069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 08:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9de34dee909a99b3745a5135c5434caa471ade15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1caca75f9ca426eb05a37ef601f8db5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9637c2e88b0455cb234a2adf4b92c67 : SUCCESS in 3m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ff4a70fd01844f02abea438d64f49cb0 : SUCCESS in 1h 20m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fe8e7401b53547db9194a7659d4639c7 : SUCCESS in 2h 10m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87623107a5814a7ea1f8e2a7a5ff2d85 : SUCCESS in 1h 20m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/00c900041b524050acfabff48d58fbbb : FAILURE in 1h 30m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d3ba090700154b7dbf296972704c8c0d : FAILURE in 1h 51m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ce8759fe483b4034a688ef58864fc386 : SUCCESS in 1h 28m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44fe9b1fb71248e0a5282f8fedffe7bb : SUCCESS in 37m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6add7ea0b8434c32b6530d0e2db7f642 : SUCCESS in 45m 40s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8e011db12fab4613aef599fdab8dd1e5 : SUCCESS in 1h 25m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f7acc0d455ce47ffbee6bb2a9074f4c3 : SUCCESS in 1h 13m 59s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/84a8cfb326d74a90b744b677770cd70a : SUCCESS in 1h 20m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b2e3ddfe3edf42438e23bbc2fb41f08f : SUCCESS in 30m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d093fd7e7d047de86d234aa28c45316 : SUCCESS in 40m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3713e42a829466181f414d9358e24d7 : SUCCESS in 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c1ad8a233633a8706f9e5a03d1a221b651fd764","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-27 13:48:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/87/862787/1/check/DellEMC_Unity_Manila_dhss_true/d5bd94c/DellEMC_Unity_Manila_dhss_true/1006 : FAILURE in 2h 25m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/87/862787/1/check/DellEMC_Unity_Manila_dhss_false/9666632/DellEMC_Unity_Manila_dhss_false/1070 : SUCCESS in 1h 13m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6ca6ce514c67ce2039b8154ff8afa61c8d0f17f3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-27 14:10:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/862787/1/upstream-check/manila-tempest-plugin-ontap-dhss/fa3e2b9/ : TIMED_OUT in 3h 09m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/862787/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/24b8b70/ : SUCCESS in 1h 20m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"77df2f15f66065f4037ca54781d67d2c3c3ae33b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-27 18:06:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"699a11feb24a818021e04075cc201786f178eecd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-10-31 19:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca5862fc063ae743c4586ad90cebd516baf525e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 19:10:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e655f564e86ffef22bca517b4051613581f1df7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 20:45:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07de56225e6f4d42a388a406272e2c8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93783c5d0d540339661a2b952360ae8 : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e1cdec8946b340a992a0cacb339ea1b1 : SUCCESS in 1h 30m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d233663a1d394910b2208786bf515f40 : SUCCESS in 38m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a76efb361bd41418a228aa04a6c776c : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":1},{"id":"04532da5e06e87d701e60e47c1ecea34fbf92762","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 20:45:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3255045efef5002378c9e1cdd5ddd02c4883bca","revisions":{"a3255045efef5002378c9e1cdd5ddd02c4883bca":{"kind":"REWORK","_number":1,"created":"2022-10-27 08:44:42.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/862787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/862787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/862787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/862787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/862787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/862787/1"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:37:13.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:41:35.000000000","tz":180},"subject":"Remove \"alt_project_share_v2_client\" client initialization","message":"Remove \"alt_project_share_v2_client\" client initialization\n\nIt\u0027s already initialized in the base classes.\n\nChange-Id: Id7e4bf5fb9aaf2dc32a7bebac3a29fb730015f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3255045efef5002378c9e1cdd5ddd02c4883bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3255045efef5002378c9e1cdd5ddd02c4883bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
