)]}'
{"id":"openstack%2Fmanila-tempest-plugin~862800","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I40c45eb6b36688b83cc5803ec99d9a420701a624","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-12-15 20:30:42.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-07 14:59:41.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-11-09 20:52:09.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-02-07 14:59:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40c45eb6b36688b83cc5803ec99d9a420701a624","subject":"[RBAC] Add quota tests","status":"MERGED","created":"2022-10-27 11:49:24.000000000","updated":"2023-02-07 14:59:41.000000000","submitted":"2023-02-07 14:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862800-secure-rbac","meta_rev_id":"3baf42afbb2be089e38b6e5297e53894e1706076","_number":862800,"virtual_id_number":862800,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 14:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-12-15 20:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-06 21:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-02-07 13:02:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 12:57:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-10-27 12:57:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-10-27 12:57:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-10-27 16:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 01:48:13.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-28 15:55:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"0cea3d397eca89d2f531b424e765eb6f5072bf87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 11:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e85fdfe0738fdec0c34736e32b4b4cb8992514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 12:06:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dc104a97a96b0599d199aa1d947049e681a93fe","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 12:07:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"f0d394190563dfb984e59ca2f7309b9a673b0d8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 12:34:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bdfd53a8f1d71c2ba67d70b98d0e7e712d94add9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 12:57:21.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"d6da5c6d148d18e2a2a9b9038e6980fe64c8e658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 13:37:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d28fdd74bd12d63a9569d9cdcaaefeb03bb2e613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:38:55.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4eab637212f4a18a279fe76816dce9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4c94ce86f040379c98c2c8188bb304 : FAILURE in 2m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8da33a4df6f741cdba77bb5fd81d6aff : SUCCESS in 1h 17m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/437a603e04064940b5e481ef3c9bccbf : FAILURE in 2h 53m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da4cd68475704fd1a2054a973ef601d2 : FAILURE in 2h 28m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/868f5e3298354cb5a1586aa8c0244187 : FAILURE in 1h 29m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4fc2cc164e9945b282cc61d5ccd6866c : FAILURE in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d71ef661b7c4432fb1a3f452025e0c35 : SUCCESS in 1h 10m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6d31a6321bb8442c9abcadf3ff42becb : SUCCESS in 26m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/caa9af1fd1e94425be0eeb1121c59c2b : FAILURE in 37m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/12ef036743914ffeab3f9d6174ca359f : SUCCESS in 1h 00m 57s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/28f0110ac84c4170b4f3eae427b1d829 : SUCCESS in 56m 29s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e4650df717ed401e86259732a16768ac : SUCCESS in 1h 08m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7f4f4227446e44d482195eab18ae7f7f : SUCCESS in 22m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa5ef605f9c44033a80de306050c23d6 : SUCCESS in 49m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/792e356b012646b6818b0d51a05a0ec7 : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"882fcdc026e4582f3a5e6e704e9ce5e3c5d1346c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 18:21:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"336de93db1988bfb89556fdf2db54461f335eb9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 20:04:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dae8338325745b8964f17086db3d3f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfc07a176834f40a31834df1a5dee06 : FAILURE in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4dc8e0c7232d47be9de5fbb0b1929df4 : SUCCESS in 56m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6123fac991104e23aec105f6875411e8 : SUCCESS in 1h 41m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5f6d190aa0ed44269d58b2055ae0b07b : SUCCESS in 1h 08m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b37140039e3f42c281b231fb818a8d61 : SUCCESS in 1h 09m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b189c7f19946409dbac246be004358f5 : FAILURE in 1h 25m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43b93b3c08ef4821a1d8a90e45e35af6 : SUCCESS in 1h 01m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2001651d3cb24cbe84677b9f311757d9 : SUCCESS in 25m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab3c2b1a13f64b22bcfc3f3d0faa3723 : SUCCESS in 28m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/26d4b1be04ce4504959e3c43486cec0c : SUCCESS in 53m 33s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6729ea8763934536aeb8bb9a04b3a48d : FAILURE in 53m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ed86f17272084b79ab5ab71c870e9a42 : SUCCESS in 58m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b11b54f163854bae82e7d1836be30d1e : SUCCESS in 25m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f91452499324edf901928e168ffc33e : SUCCESS in 37m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/059dc2ce25174a10a0e8254a5877f326 : SUCCESS in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9938926199e296363171a444fa664acabb886f93","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-28 01:48:13.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/00/862800/5/check/DellEMC_Unity_Manila_dhss_true/aa544bf/DellEMC_Unity_Manila_dhss_true/1010 : FAILURE in 3h 51m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/862800/5/check/DellEMC_Unity_Manila_dhss_false/39eb498/DellEMC_Unity_Manila_dhss_false/1076 : SUCCESS in 1h 20m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"2579bab513952def7dd3d046cfc244f99ec08994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-28 10:52:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"51bf23d18031611b3bade143fc527110e72804c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:44:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346b8d43803a4722812bc9eb7a53d310\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dcfd11020a46c5a94b9cbd022f86ab : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6ee441ff6f549c684060dfc73e7976a : SUCCESS in 1h 00m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d1e3caa64b3f4eb38b71b2cf46deda44 : SUCCESS in 1h 40m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c5df037437004d4386b43fae3f2e843a : SUCCESS in 1h 12m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/061342dea9c640a99b9fa38263310e3f : SUCCESS in 1h 06m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f635ee4921054a7ea20f894b2a1c9ac6 : FAILURE in 1h 50m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ddcc0dc5bdc2442bbbc9613f7eabe8d5 : SUCCESS in 1h 11m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/973b997d1206415bac096cc274f44a4b : SUCCESS in 28m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4949d07886cb4de58b5428b3f57c52c6 : SUCCESS in 29m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7e30a92f7b964e858b8695b0a01004db : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/16490635e0e8436d944246dc59cf0f7a : SUCCESS in 1h 18m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3bb26db6d5a3476e837334151e520409 : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a8bd9a6968be4252a63a987306bd9b6d : SUCCESS in 22m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40e4cc2ff2ef4e7ab8cfe2012ffdea77 : SUCCESS in 46m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a71d0c9dbdd41888d7cc968df39f1e9 : SUCCESS in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eeb618c869c102a7423fad5ce8edfec053501cc6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-28 14:28:19.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/00/862800/6/check/DellEMC_Unity_Manila_dhss_true/68e097b/DellEMC_Unity_Manila_dhss_true/1011 : FAILURE in 3h 35m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/862800/6/check/DellEMC_Unity_Manila_dhss_false/9d9ae21/DellEMC_Unity_Manila_dhss_false/1077 : SUCCESS in 1h 13m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c2f4cac2f237151b848ae9ec32d6d6eadcdc0d0d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-28 15:55:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/862800/6/upstream-check/manila-tempest-plugin-ontap-dhss/883b0cd/ : TIMED_OUT in 3h 09m 19s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/862800/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/85401cd/ : FAILURE in 1h 54m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"673efc68bdcda926eadae1b68d58ae1356416e4d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-31 15:45:38.000000000","message":"Topic secure-rbac-quotas removed","accounts_in_message":[],"_revision_number":6},{"id":"3cb898814fa8a5cb18af86bbca144d3c3f3d192d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-31 15:45:40.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":6},{"id":"100d3e50b1d6854a52d3abd3a67b6289e09cd002","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-11-09 20:52:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5083f806227976a882b85c20a783bdec73565406","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 18:29:52.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"aed634c32a19e4af1cfc689aede70846179043a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 10:53:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5955e4e9564fb6df96585a63dccd3440282f5436","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 10:54:07.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8eadf52b83710151c0aa44cfc24a5836252389fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 10:54:42.000000000","message":"Topic s-rbac-quota removed","accounts_in_message":[],"_revision_number":7},{"id":"7e4c08b062f3610ce2c54396bb1cf6b3b5f877b4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 10:55:01.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":7},{"id":"9016173d35fbf5d7a6d92269dc5ecce6528db07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 12:53:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63f89b4d13844ddc94fee9ba9a96b46a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32a5c21a88124975a19ff58e558927fe : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/424c23f6631f427da825691732ca8570 : SUCCESS in 1h 09m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/28290bbd28a7412da6a51c055b7a6a53 : SUCCESS in 1h 44m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7d2b7c025be4480bac1655adbda6fc58 : SUCCESS in 59m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6dcbfd9fadf8465ca929c1b251f65d0b : SUCCESS in 1h 30m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b01340b967fe4d57a8f58a0165258c91 : FAILURE in 1h 25m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b4ca8ddaae7d4fe58bf2006de3c09f35 : SUCCESS in 1h 03m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/279cf036fecc452a83b6c8801549623a : SUCCESS in 27m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a30d2f6db54840aa91b9b08465879062 : SUCCESS in 32m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6161f5e846584f2eb7cb032e7484aace : SUCCESS in 1h 27m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/27d8e7b59c384aa7b8c2dbd1b41b0741 : SUCCESS in 1h 14m 37s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d9ffef332e1b4288854bc171dfb0dc6a : FAILURE in 1h 34m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d72ba477cc124701997249754c7cbbf8 : FAILURE in 31m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6b28326c7f8c49faa4fbd0308506268f : SUCCESS in 59m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/499475a864844ba1b31ef30eda92d2c7 : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1fb8f1e06d08bd3fa980ad839b48f974b8bc456e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 13:08:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b6ae7e72faccf090de6bcd152bc3982c64ba8f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 14:54:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9278ba02be0e456496015dfad48acc09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1063ae876d74d338f21672c0a71d177 : SUCCESS in 2m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/29364aaf1d3b4c37b09940cb1edf71cb : SUCCESS in 1h 04m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/972c4ab8062646779f148deba6b2f97e : SUCCESS in 1h 45m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d521b02353b14bfaa254b983c81b5cd9 : SUCCESS in 1h 00m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ff8dc2c41c624cd6b7305d9422d50be6 : FAILURE in 1h 28m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e4cbe3ecc6d24eb798ac516573b91f1a : FAILURE in 1h 30m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdc2482d747e4fc789e518924e39aa6e : SUCCESS in 1h 04m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d63268d3ed934c289228442cb856ea98 : SUCCESS in 25m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3c74b572fcfa4c24bd97b1f179048314 : SUCCESS in 32m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/47b2bad19e2a435e868d2f7616530671 : SUCCESS in 1h 05m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cac5db256d9c4c43b5b7d05436a1f1fa : SUCCESS in 1h 00m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/920f0e1dff4942d5b167af139225abe3 : SUCCESS in 1h 10m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dc5363468a2e4c3d931936e401db9a27 : FAILURE in 22m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8d1560d38f9a4dbabb85f82be31f621b : SUCCESS in 42m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3f87a1b88a294469b869aedbf3341a2a : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"380beaf1dc390aad2dbdfe76a65e0cb69e90564a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-15 08:54:06.000000000","message":"Topic s-rbac-quota removed","accounts_in_message":[],"_revision_number":8},{"id":"8d54ae25d437be59a1b3127ef3d470b49097ddac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-15 08:54:15.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":8},{"id":"fcec3e9cc5e51f99cbc46fa1252c1d6f85f53a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-15 08:59:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"891dbe6ec59f1e116828d3d3692254b93dcf31ab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-15 08:59:47.000000000","message":"Topic s-rbac-quota removed","accounts_in_message":[],"_revision_number":9},{"id":"aa46564282a041ba196627b22ec17f41212c499f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-15 08:59:54.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"291e9a3d2b9bc5f9577b7cb16734f1439ed4d19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 11:36:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f4c8de2c924e7bb02c5a8735dfd278\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f374b6cee16480886bd1e640c6ba38d : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2734adac1cb24908a3c2cafa93d032d5 : SUCCESS in 1h 24m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bbe08277016d42b18722b478b6870a04 : FAILURE in 1h 57m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c1a602c5253743aaaad22d513c4870ec : SUCCESS in 1h 14m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/499e766e5b944372918e252e6604cab0 : SUCCESS in 1h 26m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1133835f7470415da504eeeb9a660774 : FAILURE in 2h 31m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0fb5b0b7f7c349a3946224d7123c9597 : SUCCESS in 1h 13m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da23aea2e70547378544a7fc97e18634 : SUCCESS in 36m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b85fe502cf324166ba70c64fb3fddfa3 : SUCCESS in 40m 44s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/952d8f847443432d8fc851e11092ee70 : SUCCESS in 1h 56m 07s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4b08e03a868a40db843d7dfd5fd9bb66 : SUCCESS in 1h 33m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/15fa9040fac44a599a2fe2739f982e57 : SUCCESS in 1h 27m 52s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e3f0b88e5a5146b79af4572347a6a746 : SUCCESS in 37m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9098729c552849718d6aa34857eb60d3 : SUCCESS in 56m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9f3e902884e046e5ab457ae0dde79471 : SUCCESS in 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4cd22309ef18e526bed6e8c2bb286b6888a51df4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-15 13:00:15.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/862800/9/upstream-check/manila-tempest-plugin-ontap-dhss/11289d7/ : SUCCESS in 1h 22m 55s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/862800/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/98063e1/ : FAILURE in 48m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"aa4c13adab2c1a960b77202ed06eead70f27779f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-15 20:30:42.000000000","message":"Patch Set 9: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ecfb92c9a8bffd38adb5f658f39e794b5f1b0e0f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-06 21:17:55.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e4f07f51ae17b8a7d31c13dde268d91b1a85bad4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-07 13:02:54.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"d38da291eb9e902b06e175900e0c4bceaf1151bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:03:32.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"54f1c13dc2444f1e8935ad49a026f52a043f7129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:59:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c6a50ec12c46a1b7dc6ccd194ecf83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daab523cd631471aaada4aed82bf9fba : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2a4a92879d82435a99be58e3dcc0dbb5 : SUCCESS in 1h 18m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/656a712c58374ff9a4c2f0a46e69c180 : SUCCESS in 24m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2ca1f6fb0be94a51a1e5d1ab5af5f761 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":9},{"id":"3baf42afbb2be089e38b6e5297e53894e1706076","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3bd44e8be8d5842da955f0f25b0a882a8e53c9ad","revisions":{"cccecec839e05d1684a80ae82c109d38971e6056":{"kind":"REWORK","_number":1,"created":"2022-10-27 11:49:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/1"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 11:48:13.000000000","tz":180},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cccecec839e05d1684a80ae82c109d38971e6056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cccecec839e05d1684a80ae82c109d38971e6056"}]},"branch":"refs/heads/master"},"822222394766d93243f3f4e620a6f2b9fa8265d8":{"kind":"REWORK","_number":2,"created":"2022-10-27 12:06:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/2"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 12:06:34.000000000","tz":180},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/822222394766d93243f3f4e620a6f2b9fa8265d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/822222394766d93243f3f4e620a6f2b9fa8265d8"}]},"branch":"refs/heads/master"},"9c4f4f1627f15dca73df1d3ee5906a78659e4672":{"kind":"REWORK","_number":3,"created":"2022-10-27 12:34:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/3"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 12:33:58.000000000","tz":180},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4f4f1627f15dca73df1d3ee5906a78659e4672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4f4f1627f15dca73df1d3ee5906a78659e4672"}]},"branch":"refs/heads/master"},"11aaaad071aa664d94acf2396a25277dfe234389":{"kind":"REWORK","_number":4,"created":"2022-10-27 13:37:56.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/4"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 13:37:44.000000000","tz":180},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11aaaad071aa664d94acf2396a25277dfe234389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11aaaad071aa664d94acf2396a25277dfe234389"}]},"branch":"refs/heads/master"},"4b0a7abac7b1a8180548451a696f4d4f743f19b8":{"kind":"REWORK","_number":5,"created":"2022-10-27 18:21:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/5"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 18:21:18.000000000","tz":180},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b0a7abac7b1a8180548451a696f4d4f743f19b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b0a7abac7b1a8180548451a696f4d4f743f19b8"}]},"branch":"refs/heads/master"},"d98fcd8a02f9cb54f1484c074e2616ec17df2951":{"kind":"REWORK","_number":6,"created":"2022-10-28 10:52:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/6"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-28 10:52:29.000000000","tz":120},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d98fcd8a02f9cb54f1484c074e2616ec17df2951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d98fcd8a02f9cb54f1484c074e2616ec17df2951"}]},"branch":"refs/heads/master"},"403b78c1600b2fdb8c745df783bf6fe4f364655e":{"kind":"REWORK","_number":7,"created":"2022-12-14 10:53:58.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/7"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-14 10:53:48.000000000","tz":120},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/403b78c1600b2fdb8c745df783bf6fe4f364655e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/403b78c1600b2fdb8c745df783bf6fe4f364655e"}]},"branch":"refs/heads/master"},"f3f4c6e7680e486c8eb6a6bb10353f9ca533ab68":{"kind":"REWORK","_number":8,"created":"2022-12-14 13:08:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/8"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-14 13:08:07.000000000","tz":120},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3f4c6e7680e486c8eb6a6bb10353f9ca533ab68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3f4c6e7680e486c8eb6a6bb10353f9ca533ab68"}]},"branch":"refs/heads/master"},"3bd44e8be8d5842da955f0f25b0a882a8e53c9ad":{"kind":"REWORK","_number":9,"created":"2022-12-15 08:59:42.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/00/862800/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/00/862800/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/00/862800/9"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:29:57.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-15 08:59:31.000000000","tz":120},"subject":"[RBAC] Add quota tests","message":"[RBAC] Add quota tests\n\nThe tests validate policy of quota actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I40c45eb6b36688b83cc5803ec99d9a420701a624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bd44e8be8d5842da955f0f25b0a882a8e53c9ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bd44e8be8d5842da955f0f25b0a882a8e53c9ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[]}
