)]}'
{"id":"openstack%2Fmanila-tempest-plugin~863031","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5d3471267f508118047bcf8e52afa5a1319ee016","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-03-14 07:24:01.000000000","reason":"Change was abandoned"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-03-14 07:24:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5d3471267f508118047bcf8e52afa5a1319ee016","subject":"[DNM] Check api tests with rbac enabled","status":"ABANDONED","created":"2022-10-31 10:50:51.000000000","updated":"2024-03-14 07:24:01.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31a3630ed46f89c50a9ac63e64a16cf7ee8e25a3","_number":863031,"virtual_id_number":863031,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-01 00:57:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 10:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 14:13:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-31 17:08:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-31 21:31:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"28bb21dfec8fd63ca0bf54032e6299bf73b423ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-31 10:50:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"523b8c9082d358635f3afd888c0d45750c7da1e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 10:51:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila on branch master.  The error was:\n\n  The project template \"manila-tempest-plugin-jobs-using-service-image\"\n  was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n        - manila-tempest-plugin-jobs-using-service-image\n      queue: manila\n      check:\n        jobs:\n          - manila-tox-genconfig\n          - openstack-tox-pylint:\n              voting: false\n              timeout: 5400\n          - openstack-tox-cover:\n              voting: false\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-container:\n              voting: false\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs:\n              voting: false\n          - manila-tempest-plugin-lvm-fips:\n              voting: false\n          - manila-grenade:\n              voting: false\n          - manila-grenade-skip-level:\n              voting: false\n          - manila-rally-no-ss:\n              voting: false\n          - manila-rally-ss:\n              voting: false\n      gate:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n      experimental:\n        jobs:\n          - tripleo-ci-centos-8-scenario004-standalone\n\n  in \"openstack/manila/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"dea7a844d133df12ea9bb25cdf2f3e6d624418ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-31 10:53:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66651d57c5f5409791f050f5a5a27ca68936fbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-31 10:55:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19af2bad4759248ef70b2bdf3d31ae84e2b93067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 11:23:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48e5077e16d49b784fdaeced03e566b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68dcd04761274ee8bc2d5b2174fcce5e : SUCCESS in 2m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b81ed072d9544733855a12862e8d26bc : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":3},{"id":"be519baf7bfc1e25171270417a9be594acc621ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-31 12:03:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"42bb4c80c3e2ff33f944bf5fbebbcb832bee2fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:49:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26108923d56046b3b4f7e8ca144c9961\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8d470c887a4a9d8862180c2919004e : SUCCESS in 2m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bdcdafa00fe44451b0ead6632a5c800f : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":4},{"id":"cdf713521070248619a48b0d978b20f9e03ed511","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-31 14:13:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/863031/4/upstream-check/manila-tempest-plugin-ontap-dhss/720ac2d/ : RETRY_LIMIT in 7m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/863031/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/1e12ff1/ : SUCCESS in 1h 22m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"21a9c1be0af9d8ea52608bf125fee83a4bdb5a1f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-31 17:08:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/863031/4/check/DellEMC_Unity_Manila_dhss_true/0c24f50/DellEMC_Unity_Manila_dhss_true/1014 : FAILURE in 3h 44m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/863031/4/check/DellEMC_Unity_Manila_dhss_false/20cf9a5/DellEMC_Unity_Manila_dhss_false/1081 : SUCCESS in 1h 00m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2d2628b5ae72461fcac2444a5afbd0b122d81ae3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-31 21:31:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0fb0d6a8cc27e574f798ce87d233b6c7f70c38ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-31 21:32:48.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"6131816c52f1227763b7ee79b5c43c415695c46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 22:06:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37ba8579c417483cb887c364a04ad14f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70fb412934d441ac8a9ba6df33470796 : SUCCESS in 3m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/93f9487e9a974ca690f1f9a2961f4f49 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":5},{"id":"af25b24f09d82d5441d9bd7bb4028cd8a94f996d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-31 23:15:28.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"99c414f76d905edcbdfe2c0795f1875d80e64267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 00:57:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fe3675c2dbb45d09082fcbda10c8ce4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af35e9be179452ba4b5d5a5cc16976a : SUCCESS in 2m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10dd30b8b6fe4f4d9a2acbb7f65f0fe2 : SUCCESS in 1h 41m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/38bd662f42b34545b22832e8957b00bb : SUCCESS in 1h 34m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6071e90b02434a82b5ff5015bdaa49b4 : SUCCESS in 54m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e87c08a5ae664549a8339cb02c6d150b : SUCCESS in 1h 10m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a2ac39bfdc9b4c839b7e5c308ef70de5 : FAILURE in 1h 31m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6d0021a9573c4b2e879895ccfd7f304c : SUCCESS in 1h 03m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/16bc2b9b00df4320b3c7f7040007ab81 : SUCCESS in 26m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e301701a808e4ba59331a43bfc48f919 : SUCCESS in 34m 56s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/922a555dc259447ebb29ede8b1c30b03 : FAILURE in 1h 12m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ebd853b93c9d4e938400b0da602663d2 : SUCCESS in 1h 01m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4dd9e62e0c8c47ebb3eb613f79bf2fb2 : SUCCESS in 53m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/84043acf6cf64e25ba18c7924b4acc57 : SUCCESS in 31m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2ef62e2c874b488aaebd4426039cba7a : SUCCESS in 32m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/946bee2a7c60468e8e32e17d2f5ee809 : SUCCESS in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"48f6f2e54d4f3215b49582d84c469435784cb955","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-01 03:14:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/863031/6/check/DellEMC_Unity_Manila_dhss_true/d02aa27/DellEMC_Unity_Manila_dhss_true/1017 : FAILURE in 3h 58m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/863031/6/check/DellEMC_Unity_Manila_dhss_false/1580d91/DellEMC_Unity_Manila_dhss_false/1084 : FAILURE in 1h 10m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"48c2632c139502e3935d9dc9e39c14ebef56050d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-01 04:45:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3f46a1875345358bc30370eea9a64e9a0f01725c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-01 19:16:19.000000000","message":"Topic api_with_rbac removed","accounts_in_message":[],"_revision_number":6},{"id":"5fd67e4255a70724c96bbf2114b486973af7fe47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-01 19:16:24.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":6},{"id":"5a1a95c889a598365738a68f63e3f0a8a54fb7af","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-01 19:21:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"760b2f86e5c461166140a6394eb28957c7aecf09","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-04 10:52:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"951e005c9c61ef8c72a613c522e62cd50ed629fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-04 12:57:04.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f9129543d6043448bbda10d8504c4b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c2e74cd078548a9bcf4219aaaaed122 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f7b247feff14412880bba2d992d8a18f : SUCCESS in 52m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0f2d62f8df52445a9a8df657dc670af6 : SUCCESS in 1h 27m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/946d4dfead544de4b65c50af497bc898 : SUCCESS in 45m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48efbd1ef0bb43688112840d9ad48188 : SUCCESS in 1h 03m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2a2a192b00724bae93220bd9161c8989 : FAILURE in 1h 52m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d4d70d67cf14015be87b1cb418c0d7e : SUCCESS in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cfa66fc15d7f464a8caddd247c6ba121 : SUCCESS in 21m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/671965e13ded4e2f93041f78b0096d38 : SUCCESS in 26m 38s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f889079faad143508e9964b3ddbc4d2e : FAILURE in 52m 45s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f990da565ac04fbd9293cea345ae5d31 : SUCCESS in 49m 07s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8b7e261861344b20b35b14e75c212f9c : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/99adbecebb1c4b0881b046b8d57a1f7f : SUCCESS in 24m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5044c76a9d3b42ccafbaaace7f9e6407 : SUCCESS in 46m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/24b6a4c73ae44a78aa32707e9c751d55 : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31a3630ed46f89c50a9ac63e64a16cf7ee8e25a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-03-14 07:24:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"655648957ea7787f05ffec36e7c5b117114125b0","revisions":{"ae9e80b2d9e3cab4aa2a18c96a0554f1e5b0037f":{"kind":"REWORK","_number":1,"created":"2022-10-31 10:50:51.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/31/863031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/1"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae9e80b2d9e3cab4aa2a18c96a0554f1e5b0037f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae9e80b2d9e3cab4aa2a18c96a0554f1e5b0037f"}]},"branch":"refs/heads/master"},"3a42fc730dee81310d81e95af2bee067e5e3517f":{"kind":"REWORK","_number":2,"created":"2022-10-31 10:53:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/31/863031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/2"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:53:32.000000000","tz":120},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a42fc730dee81310d81e95af2bee067e5e3517f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a42fc730dee81310d81e95af2bee067e5e3517f"}]},"branch":"refs/heads/master"},"0c677de202ddb8f43b9629fa02aadc52868f7ea8":{"kind":"REWORK","_number":3,"created":"2022-10-31 10:55:08.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/31/863031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/3"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:54:59.000000000","tz":120},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c677de202ddb8f43b9629fa02aadc52868f7ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c677de202ddb8f43b9629fa02aadc52868f7ea8"}]},"branch":"refs/heads/master"},"69e5c4360586576273584bf66762d022e6f254bb":{"kind":"REWORK","_number":4,"created":"2022-10-31 12:03:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/31/863031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/4"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 12:03:15.000000000","tz":120},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69e5c4360586576273584bf66762d022e6f254bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69e5c4360586576273584bf66762d022e6f254bb"}]},"branch":"refs/heads/master"},"589af60d23b78159e3d46d5c47a6e33b61dbd208":{"kind":"REWORK","_number":5,"created":"2022-10-31 21:32:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/31/863031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/5"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-10-31 21:32:48.000000000","tz":0},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/589af60d23b78159e3d46d5c47a6e33b61dbd208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/589af60d23b78159e3d46d5c47a6e33b61dbd208"}]},"branch":"refs/heads/master"},"655648957ea7787f05ffec36e7c5b117114125b0":{"kind":"REWORK","_number":6,"created":"2022-10-31 23:15:28.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/31/863031/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/863031/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/863031/6"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-31 10:50:14.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-10-31 23:15:28.000000000","tz":0},"subject":"[DNM] Check api tests with rbac enabled","message":"[DNM] Check api tests with rbac enabled\n\nChange-Id: I5d3471267f508118047bcf8e52afa5a1319ee016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/655648957ea7787f05ffec36e7c5b117114125b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/655648957ea7787f05ffec36e7c5b117114125b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
