)]}'
{"id":"openstack%2Fmanila-tempest-plugin~863253","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I34bed8ac06aad666fe1d4c8eb79e005716970bf0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-11-08 22:01:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34bed8ac06aad666fe1d4c8eb79e005716970bf0","subject":"[RBAC] Fix test tags","status":"MERGED","created":"2022-11-02 11:00:27.000000000","updated":"2022-11-08 22:01:08.000000000","submitted":"2022-11-08 22:01:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863253-secure-rbac","meta_rev_id":"f0b7d05a3d66d0f2775513c8f1393c2edde90007","_number":863253,"virtual_id_number":863253,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 22:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-11-08 21:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-11-02 15:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-11-08 21:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 14:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 14:12:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-02 15:20:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-08 21:04:08.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"f67c47c56ea8b32291f28c436b59f8ee2ae4bb88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-02 11:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e53e8cd8675a1bdac71fe7852d4d886bde9ee3e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-02 11:00:42.000000000","message":"Topic fix-test-tag removed","accounts_in_message":[],"_revision_number":1},{"id":"9036bb6cf995717d155d2eaecd8fa306014bb706","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-02 11:00:50.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"c111d74b833e547b2c8b1ba6aaabe2e791c64bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 14:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6868918ee747435f8c54e0113d652121\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3523cab83df94617a541fabf09763f75 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/376e1b31da9c49118b5b5904645223c9 : SUCCESS in 1h 14m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fd3121d10ebf4b2e946e2a00848fc406 : FAILURE in 2h 54m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7fefb33145244e3fa899585a181e5d7b : FAILURE in 1h 55m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/89028cc1caf542c28ee935f77d3bc6b9 : FAILURE in 1h 48m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8a16e4444c554774b61ee1410e1ec150 : FAILURE in 1h 46m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/134d669e52b4438e909005fb071d08ac : FAILURE in 1h 08m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/405bb7c8d7d643ceb5406385445e7d01 : SUCCESS in 38m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ac3f5c1720264236b262e133ba605f44 : SUCCESS in 40m 58s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/df13f1a3f994482abded7c03ee348189 : SUCCESS in 1h 11m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/34258277b68742d38ddd528791fec9dc : SUCCESS in 1h 16m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6a884aa56b6b4a599f784a4b00d25b93 : SUCCESS in 1h 16m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e4df8294378d40028e48ebce8c3cb081 : SUCCESS in 31m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fcf7b9d766f94ef7b72f4ab068de2420 : SUCCESS in 54m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/24c22588b00146bf9cf671b607f83d88 : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aacc62c077f4d067a20298b5ad46b154b93adad","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-02 14:12:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/863253/1/upstream-check/manila-tempest-plugin-ontap-dhss/42a8148/ : RETRY_LIMIT in 9m 00s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/863253/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/87fc915/ : RETRY_LIMIT in 8m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c267485589a321bfbfbea83194328f9ef02f890c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-02 15:20:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c0fb475d5cc50e5a0ad3e8ee0a3effb034f65ca","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-08 21:04:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d0352c9f7eb2102f2d0fd4e7700bf84f0b185c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 21:04:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46a260700cebf549169e994b08cf73d90c47206a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 22:01:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2cd73244644ff799dbda43a2ea2e70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee58b6196cf42e3963dd069cda6656f : SUCCESS in 2m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6784a2f624f244f9b7f62d1c6312bc0b : SUCCESS in 55m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/844a02c9a88a4729acd49fab0900f335 : SUCCESS in 25m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c325c6a5c574cfea2375f1d9ade75a6 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f0b7d05a3d66d0f2775513c8f1393c2edde90007","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 22:01:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"121f536ac32649fdc9f13dca681ab3fd69e725e8","revisions":{"121f536ac32649fdc9f13dca681ab3fd69e725e8":{"kind":"REWORK","_number":1,"created":"2022-11-02 11:00:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/53/863253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/863253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/863253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/863253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/863253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/863253/1"}}},"commit":{"parents":[{"commit":"256a6b09d4ae8305de8ef9f9e1ae3498d43b4421","subject":"Merge \"Replace period to underscore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/256a6b09d4ae8305de8ef9f9e1ae3498d43b4421"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-02 10:55:00.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-02 10:55:00.000000000","tz":120},"subject":"[RBAC] Fix test tags","message":"[RBAC] Fix test tags\n\nThere are a number of tests whose direction is\nnot marked correctly.\n\nChange-Id: I34bed8ac06aad666fe1d4c8eb79e005716970bf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/121f536ac32649fdc9f13dca681ab3fd69e725e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/121f536ac32649fdc9f13dca681ab3fd69e725e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
