)]}'
{"id":"openstack%2Fmanila-tempest-plugin~865259","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I26303d012cc2c75233cc01cd297f364d4cd0ce7c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-12-13 21:41:09.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-01-03 18:11:55.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-01-12 17:02:23.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-01-12 17:02:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26303d012cc2c75233cc01cd297f364d4cd0ce7c","subject":"[RBAC] Add share type extra specs tests","status":"MERGED","created":"2022-11-22 10:07:31.000000000","updated":"2023-01-12 17:02:23.000000000","submitted":"2023-01-12 17:02:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865259-secure-rbac","meta_rev_id":"d7461073159d55c857db1a6240898bfaf7dbcb64","_number":865259,"virtual_id_number":865259,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 17:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-01-12 12:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-01-12 15:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-01-12 15:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-22 13:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 15:21:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-11-23 13:17:05.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-23 14:16:02.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-11-23 14:16:02.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-23 14:16:02.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"605fe91192d51b2ac6b16ab84a8507e43333864e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-22 10:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f439c5f1f27010314835f4f1d393b8e5d0063d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-22 10:09:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2603b2df4102ee280feaa4cb80646a9fe709fa6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 13:20:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d04b3ab66ff74afb977f18ac63399927\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55c5f1b91b2348b4ba958e8102c7877f : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/83a7c590bfac46fdb0a3bdf776d9b616 : FAILURE in 33m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fa397c0e7a1f44bdb4357a17f03ffb15 : TIMED_OUT in 3h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/161c189fae8b429c9440191945ffdc1d : TIMED_OUT in 3h 04m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/06c755e0e2f64352ac656c42ab7e8c53 : FAILURE in 38m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/248cd50236fa49139fcbedf1aff17a31 : FAILURE in 16m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1acae2f4b1ea4895b728e8adeb832700 : SUCCESS in 1h 18m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6dedf831b89c4a40b6029e04c4189f87 : SUCCESS in 44m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9fd0661b76eb435ab176ce2c64edca62 : SUCCESS in 45m 47s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/478edc4a904441779ae615e15a8a91fa : SUCCESS in 1h 28m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/690e2057f91b490a95da4d96afd00e13 : SUCCESS in 1h 25m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/31d475316df94a84a8d39556123db500 : SUCCESS in 1h 15m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/08850bd2bee14bd5a33cd06ed7dae698 : SUCCESS in 31m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c62ae38b74fd4410b6e140d248fb066f : SUCCESS in 56m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8f25cf8c7c664768a07bc85cc83a59cb : FAILURE in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89963bc9f87b43b95eb0cd3ca28c27d433e7af57","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-22 15:21:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/2/check/DellEMC_Unity_Manila_dhss_true/044c697/DellEMC_Unity_Manila_dhss_true/1036 : FAILURE in 2h 46m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/2/check/DellEMC_Unity_Manila_dhss_false/ef399b1/DellEMC_Unity_Manila_dhss_false/1103 : SUCCESS in 1h 04m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"128f048f929eb94fc45524257bf9a3cc5ee01f52","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-22 18:47:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bc88df5830bcc32db19a3a7c459a9bfd341039f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 20:43:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/773f9462c5eb422099676e879898e9a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dd0085f32b486cb8da5fd2bbe370f7 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/98666fefbf7e42859aa5592a84c0f194 : SUCCESS in 1h 06m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/91ff784657d54875a65943a7795147bd : SUCCESS in 1h 50m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0cb64adc81b54689a0bd5ab6a932f635 : SUCCESS in 1h 09m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/07491250ad034db3903f855858592080 : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/00f31d12df584b248969d02ae7089f4e : FAILURE in 1h 47m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ffb7f6ae0af14d5bb0713cf9282f07ab : SUCCESS in 1h 02m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d7a8088c5d744191841a64521f48bfac : FAILURE in 27m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/85554ea5a052416eb684eb9439517982 : SUCCESS in 31m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/cef97791afc54420bc77c14639f3a5b7 : SUCCESS in 1h 27m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f829e55fc8c548e2925e4ad2e286201e : SUCCESS in 1h 18m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/62a58f4739d44f70b63d901cba15c984 : SUCCESS in 1h 05m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/233c6b765f6d4022abde63a55bd4ae97 : SUCCESS in 23m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6736bb3c37b4149a11a57e0cd09c8b4 : SUCCESS in 53m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0cfc3070cc9c417da5c87e791730151f : SUCCESS in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"552bc410286ab4c5baec4a9def1b91dd437bb9a2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-23 05:27:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf24e3b4e90969e49c58f79d3151b92f3daf1553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 07:59:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1253381653a14a3ebb0071dad4a99775\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5abcaafe514734ad820b9090facde3 : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f95b6f959c864e1b8128c31bf5c5cd82 : SUCCESS in 1h 07m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d2cd352dd0be4002a31ccb3a7a970975 : FAILURE in 2h 26m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f43fea1d4aab48ddabd59747007bca04 : SUCCESS in 1h 12m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3e216d9f02054f11b9c0d2f16600c142 : SUCCESS in 1h 20m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dc50556a0929410aa24a0fc944c4468d : FAILURE in 1h 31m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43be83cc957d474184520277e960e262 : SUCCESS in 1h 04m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0573adc6f0814b15ba63ef91926c2eff : FAILURE in 38m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/388ca3b254944cb58e5abce6feb88db7 : SUCCESS in 40m 58s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4ff9c57ab93649f2a84805235b281878 : SUCCESS in 1h 09m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/049e4076754044a0b9fe0c4efb025b3b : SUCCESS in 1h 02m 36s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/27a75d4346284390990c77d0d955000b : SUCCESS in 1h 05m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/29ea171e00a94084bfdf7e7dbf4e66d8 : SUCCESS in 28m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/30101fc423e743b3959b9f82c05318f5 : SUCCESS in 44m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0bc5cc755c834babac2d8bf3a2029de6 : SUCCESS in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49f79b656f03e52219a6af3ee7cbbcb210eec4d8","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-23 08:14:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3d6d3061c240157e3edbc0fd4b1a1585d67ceb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 11:25:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d323f48144944dafab3f750d068481dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ee9ccda5094d42b1b80e1366f6daab : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7bba54a8ed5f4e66ae924bde48b56ffc : SUCCESS in 1h 07m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2d109324c410407ca946e27a9702940c : TIMED_OUT in 3h 02m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a0c658148754fd2bdf46c892c29ac91 : SUCCESS in 1h 01m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/74d75955a8014c90934895e1c30adb4f : SUCCESS in 1h 21m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8c527c749c064edeb65f51037ec96319 : FAILURE in 1h 48m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b2e98543443f4cb8abe980c25d1de2e4 : SUCCESS in 56m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47cddd6cca9540e9a4ffe355b3670d2f : SUCCESS in 38m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f4f130ed84f94ed789d6d7d3997333b0 : SUCCESS in 31m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7cbcd8f734b744e6988e02fb0b900c09 : SUCCESS in 1h 09m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5940347040c945799383cb34fd6e4518 : SUCCESS in 1h 06m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c62b52d0a20d4466a42904b876d3b619 : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/87ad20afd2f84489a9cb6fde703fbd9d : SUCCESS in 24m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/273362fd2df64a389728515bccbd2d6f : SUCCESS in 56m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/654f5f642c6a468e883fdf0ce0f3eab8 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"133e0ebb9846b11d3ec29694df3461beb68d7265","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-23 13:17:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/865259/2/upstream-check/manila-tempest-plugin-ontap-dhss/bea754c/ : SUCCESS in 1h 50m 07s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/865259/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/48582d4/ : RETRY_LIMIT in 8m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"84a1862501f8a29da8cf6335b8843150b670a38d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-24 03:53:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/2/check/DellEMC_Unity_Manila_dhss_true/6a701eb/DellEMC_Unity_Manila_dhss_true/1047 : FAILURE in 3h 15m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/2/check/DellEMC_Unity_Manila_dhss_false/dda8063/DellEMC_Unity_Manila_dhss_false/1116 : SUCCESS in 1h 13m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fc171ddf0536503862cdff6ca4c00da85d159ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:01:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e1776abd8ee6e3d1429db05d4291efb614b3682","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:04:26.000000000","message":"Topic s-rbac-extra-specs removed","accounts_in_message":[],"_revision_number":3},{"id":"07826147b1eb604a3d6deaa043b36a65164e04d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:04:33.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":3},{"id":"432993b6272d551855d533ebcd10b79cdc9fcf5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:19:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d67210c95401334e11cacde3fbacddb9b830aeab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:19:53.000000000","message":"Topic s-rbac-extra-specs removed","accounts_in_message":[],"_revision_number":4},{"id":"63dcfe88fcca0dcd8b0782c8a7e726da87637e02","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:20:00.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"3411efdb44141ee5eadc356a403f8130540c4ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:35:38.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1be6c3aab92ef5a8bf24e14e4c7f029763793410","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:36:02.000000000","message":"Topic s-rbac-share-group-type removed","accounts_in_message":[],"_revision_number":5},{"id":"d0c745866351649acef4520076c1090a7ddda029","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:36:11.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":5},{"id":"b935af7540d0299bc0d0c0fe0aa2dff12dce99d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-27 14:05:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c224d9d1aca4d07b9517ae93cad8a0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94286c6fe7fd4d919c255fe37011cf69 : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3cd9eebe6c8243b7a20127de6c1ad01f : SUCCESS in 1h 19m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b96dcd8687dc4472bb622faa56ea99d1 : FAILURE in 2h 23m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/756d881758cf4a59b519ddce43053d89 : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/00376cb7ba804c90904546c81e9f1faa : SUCCESS in 1h 13m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b0853cdfbe374e9d9f9913d5e67a689e : FAILURE in 2h 08m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8ba9945460324db2b9b54759acd4cbe4 : SUCCESS in 1h 13m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7168aeae37cd419bb7e7bcb45c43c702 : SUCCESS in 35m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bfe1c90696db4ae88d190212ea37a536 : SUCCESS in 44m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ef80fdf98a554172a6927472b1633cf7 : SUCCESS in 1h 31m 54s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/af4a44b2ddd64f63b4124d8d370e8f65 : SUCCESS in 1h 13m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/45849ab5cc934607a1d23e570aaa4b5b : SUCCESS in 47m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9453582dc9ec48c69c5af424d38b125c : SUCCESS in 19m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4115623b61f545d2b3912ac79aa289db : SUCCESS in 55m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/26b9666f14574aa8b1b2be9af7044173 : SUCCESS in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"503a326cdd2455beeca0ebdcf3d91a4325301c86","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-29 17:09:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/5/check/DellEMC_Unity_Manila_dhss_true/06ca9dc/DellEMC_Unity_Manila_dhss_true/1052 : FAILURE in 0s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/5/check/DellEMC_Unity_Manila_dhss_false/9ca5263/DellEMC_Unity_Manila_dhss_false/1123 : FAILURE in 0s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a8e03daff10cf57802b4a552943c48c0389d8b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-08 12:51:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8b5ab90953d7839ad9ec9c3b8f52e3575a70f969","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-08 15:53:03.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/6/check/DellEMC_Unity_Manila_dhss_true/87bec0e/DellEMC_Unity_Manila_dhss_true/1069 : FAILURE in 3h 01m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/6/check/DellEMC_Unity_Manila_dhss_false/a9047a6/DellEMC_Unity_Manila_dhss_false/1141 : SUCCESS in 58m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ccb593d914711ebc28495f9f5649235ca73a9c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 16:01:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d42af9a40bf3413b9d5904c9430c9c67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad2a2e35eb34885a544ce9f7342cf77 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e392fcedc2a94891881d8deb55b56d26 : FAILURE in 34m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/24eef9c0415f4d42b9dea08b429038fe : TIMED_OUT in 3h 02m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/89e1a8171e2c4c1890600ea3c328b5cc : FAILURE in 2h 37m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7174f6e6b07f43249ee9de10316b25bb : FAILURE in 37m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c5d18b1077544b94a19372d7b4b258a5 : FAILURE in 39m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5db7cb28770445d2a053fe9dbec38dba : SUCCESS in 1h 15m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/477fb8ebbfe9418bb4095c51b830412d : SUCCESS in 37m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/84558a188c414d40a7956891b011d503 : SUCCESS in 38m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/bd79f63e050542349ff9e760b50329c1 : FAILURE in 35m 18s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2bfb20462c2b40e2a080e32ebf13429f : SUCCESS in 1h 07m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ab89339bf87840238dff41439b4f9208 : SUCCESS in 1h 15m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c14bb6ba32e04f81af9627e86a6eee7d : SUCCESS in 30m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bcc543513661437fa134ae50eebc82a6 : SUCCESS in 55m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9983c727a8b14e3187687985f0a884a4 : FAILURE in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"68512e65d740a75e85088c52d236e83c282f94b7","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-08 17:51:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d6f99cbea59ac33cefb47994131618515d502f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 21:02:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/451281a34fee42709fbc8129c8ae60d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774b0f491b3d4d0a8295890e68ab6a7f : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/48538c0ddbc347e3a19e1aefee291693 : FAILURE in 19m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4706d32af9fa43c1b8d91f9314415051 : FAILURE in 2h 22m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0f43400ca61d455897a0c4bc1df0992a : TIMED_OUT in 3h 09m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4872e9a986734d34959d93c9ad9b9bec : FAILURE in 36m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f768244bcdfe4bc8bff7b0ba16cd392d : FAILURE in 35m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7893e0d59cbd43bfb0a795433cdaa82d : SUCCESS in 54m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d380da93ca9448ba84d0ad43754b1281 : SUCCESS in 28m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/19d1a6baf2324f5f9549d41cc2a1a2f5 : SUCCESS in 40m 14s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fa61e3f92bbc41d0886635634ddff486 : FAILURE in 20m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4938af23e72a421a9b8f05ff3c2ace7f : FAILURE in 1h 16m 40s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/bdb07c4859d2421cab3045f12ffc1381 : SUCCESS in 1h 07m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/982b6e7c77a94557b36099e6d179a31a : SUCCESS in 17m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4fdd4ca6018943f7aa57a46b8f917410 : SUCCESS in 55m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c54b08163138449cb518692580c17b72 : FAILURE in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b1d8a9f8310430da6938118f005f9da56d0971a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-11 20:21:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"14eeb1762c21e60aee8e174f5ff9ddb4e4f30b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 22:16:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27dc21c2f93647ef88c5c0304d478cac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaaa189220f24d19bdf007ea6f3c93eb : SUCCESS in 3m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3eec98c9007e4edab10fdcf8d8e18027 : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fb1189abb7d84069991988537f3992fd : SUCCESS in 1h 36m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/649ca5c5b70e42b7a8eec9c295067157 : SUCCESS in 1h 04m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a7c1b3981f474f86a202bc934df8bd10 : SUCCESS in 1h 25m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/12d5234ef246493bb761fb03ef7ba6bb : FAILURE in 1h 51m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b33dcb9fecb43b4be55ede28fa45d86 : SUCCESS in 58m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/23d780ccb7ff4fd3b197e94067128a38 : SUCCESS in 26m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f507f349d7c4874993dabeb2bc5cda0 : SUCCESS in 33m 27s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d4a67f308c4c4ac7990b4e2ce535e2ae : SUCCESS in 1h 34m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cc58df5f8772418f8abffd4e1235c32f : SUCCESS in 58m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7d39d92de75e4d7aba5252e16ab15477 : SUCCESS in 1h 05m 23s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c4f5a7cbdfb845288c2b6f2e9f41b6b9 : SUCCESS in 22m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/82ac37d57b834be4afa2277a0392e92b : SUCCESS in 45m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/520c3c8cd1d14bd5a1a4e1c742eaab69 : SUCCESS in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0b814e70998d3b024c86346e5b780c840971223","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-12 00:15:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/865259/6/upstream-check/manila-tempest-plugin-ontap-dhss/d512ffd/ : FAILURE in 1h 58m 11s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/865259/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/a269505/ : RETRY_LIMIT in 9m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"8b4297b75530d6dffd3c74de849adbc3e6f6d683","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-12 16:15:58.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/6/check/DellEMC_Unity_Manila_dhss_true/3ef8a6a/DellEMC_Unity_Manila_dhss_true/1078 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/6/check/DellEMC_Unity_Manila_dhss_false/9504819/DellEMC_Unity_Manila_dhss_false/1151 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"72a9734ac5f17ee8b914230d87e91a45afdc6444","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 21:41:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"445963c65e2b9d181c5366cca3a2900288420ce8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-03 18:11:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"82dcd4db9edb417e24ce7330ca65e0a70a27ba13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:12:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"12755211eaa4d5f7a5d414b60e67cc0ae629aee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 20:04:03.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fc7814080824295966fcb196e394502\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a546cf63214a9d94605159b5f6f1c3 : FAILURE in 2m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/488e94ef082546498b8df5faf71472c3 : FAILURE in 1h 45m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1d4a2bc942fd44fc8535cb89e61bead5 : SUCCESS in 48m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d85203582134222b54fdc0c857f61f2 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":6},{"id":"54fdf7f8a7b62cc6bd628675de3c05f68515b242","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-05 21:13:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"959d47af8940f833979631739d6fc75423b1afec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:10:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0721699568f54c769b8ad43d2a9c7308\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae2599016bd43cf9fd47209e953a7cf : FAILURE in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/463157f7797f4b0b94e955a9fc97876f : SUCCESS in 58m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3a8653dc061d4fa884a2680bf31f7d74 : FAILURE in 1h 54m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ce628242991d4e0fafbf20bf30bbd24d : SUCCESS in 1h 09m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/32cbdfc3a3894070b171524f363593ab : SUCCESS in 1h 13m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1ea07ff0e48e482b8582e925d18f0932 : FAILURE in 1h 40m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d1e6a3c6ad3542b3b88fd8e510f08730 : SUCCESS in 46m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/155553b3368549bf926253fc817af68b : SUCCESS in 22m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7fcdb6f364164333bb66306664881f42 : SUCCESS in 43m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c3a109551a6d4580925a4e34d35e7c67 : SUCCESS in 58m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ef6790f2e6024fb5872bf3310f939597 : SUCCESS in 52m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a58088cec7be4500bca25630fee25193 : SUCCESS in 54m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5bac1c2c2b814d77a5fcdb612914f39e : SUCCESS in 32m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6336ea476697401691eb32351e442d99 : SUCCESS in 51m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/693ee6b3877647329115be6fae45db37 : SUCCESS in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"23b3bb9d1c3fed3f8a780c4232db86e623af585f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-12 12:54:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5d3f652826a231ff28fb6eda9630aac8a939bb44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 15:14:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7e851a727b48a6ba51aa0f82143b37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cfc076847fc471c9ec302ff96736ae5 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/043d14475b83409383eec927f7871ae8 : SUCCESS in 1h 45m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b493dafb96344b12b3852ef17b4c2aac : SUCCESS in 1h 55m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b1c9d9a4372b4010bcf198e38ef2253c : SUCCESS in 1h 00m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/87b0c2a26de040d2bb6b175ae0518127 : SUCCESS in 1h 07m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e7972d41fcca458fa72fc573e7bd60cb : FAILURE in 2h 09m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/73fef7875e5542be9d076ed5f8ca31d3 : SUCCESS in 1h 24m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e110c0960294465a484f6d5d18e2257 : SUCCESS in 28m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c84f9737d427458b9d6427a79bff6c41 : SUCCESS in 42m 41s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/748252732eea4dd3b0ce531147d68717 : SUCCESS in 1h 11m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0561f78891cc45b99621a55208eec401 : SUCCESS in 1h 00m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1cff89affb2241739a0e5731b0ce0c31 : SUCCESS in 1h 04m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/452701e9fd3640f6a2b57b1434d462b2 : SUCCESS in 21m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7367455384244dda5b85eb883f14fbe : SUCCESS in 54m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ad25ff93e3ab41d9afc3f894523d488f : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"23f838cf35c166f8b112c0ffffa3fe27124035ee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-12 15:40:17.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/59/865259/7/check/DellEMC_Unity_Manila_dhss_true/1b00869/DellEMC_Unity_Manila_dhss_true/1138 : FAILURE in 1h 02m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/865259/7/check/DellEMC_Unity_Manila_dhss_false/0b15546/DellEMC_Unity_Manila_dhss_false/1208 : SUCCESS in 1h 01m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"fa67560837b902010a74153aa8ca1fa3573beb96","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-12 15:58:22.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b2483dbde710312c4ba384af87c3816143225267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 15:58:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1219d3878567c08ce6bd117e367b387bea71ce98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 17:02:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a0be82d52544fbbfb57882aa6a4f6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc9f6f97ebd4d89aac57c093bc3c899 : SUCCESS in 2m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4696917a2f614c01ade03d39c8a2e87c : SUCCESS in 1h 03m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e761e29dd47d4b279b772f7516ce40fb : SUCCESS in 21m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/efc0d2eb11d949b687f26cffe706e343 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":7},{"id":"d7461073159d55c857db1a6240898bfaf7dbcb64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 17:02:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4db10d48acee0418bcbbc6e4026349f2876d1050","revisions":{"242871fcb963f61a112ec4178bd3f2502e77d2fd":{"kind":"REWORK","_number":1,"created":"2022-11-22 10:07:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/1"}}},"commit":{"parents":[{"commit":"2bf48538cc350eb537ffd0a378f15cfc82e1ccfd","subject":"Merge \"Add tempest tests for share snapshot metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf48538cc350eb537ffd0a378f15cfc82e1ccfd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/242871fcb963f61a112ec4178bd3f2502e77d2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/242871fcb963f61a112ec4178bd3f2502e77d2fd"}]},"branch":"refs/heads/master"},"b213e0a8858b34d014176e297e856b5581df817e":{"kind":"REWORK","_number":2,"created":"2022-11-22 10:09:00.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/2"}}},"commit":{"parents":[{"commit":"2bf48538cc350eb537ffd0a378f15cfc82e1ccfd","subject":"Merge \"Add tempest tests for share snapshot metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf48538cc350eb537ffd0a378f15cfc82e1ccfd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:08:51.000000000","tz":120},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b213e0a8858b34d014176e297e856b5581df817e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b213e0a8858b34d014176e297e856b5581df817e"}]},"branch":"refs/heads/master"},"4fc77cdcf81a762c5193720d8f5f0e9212c6214f":{"kind":"REWORK","_number":3,"created":"2022-11-27 11:01:49.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/3"}}},"commit":{"parents":[{"commit":"2bf48538cc350eb537ffd0a378f15cfc82e1ccfd","subject":"Merge \"Add tempest tests for share snapshot metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf48538cc350eb537ffd0a378f15cfc82e1ccfd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 11:01:40.000000000","tz":120},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc77cdcf81a762c5193720d8f5f0e9212c6214f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc77cdcf81a762c5193720d8f5f0e9212c6214f"}]},"branch":"refs/heads/master"},"66facac6e9345fe562e84a6449830ddae3d91165":{"kind":"REWORK","_number":4,"created":"2022-11-27 11:19:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/4"}}},"commit":{"parents":[{"commit":"2bf48538cc350eb537ffd0a378f15cfc82e1ccfd","subject":"Merge \"Add tempest tests for share snapshot metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf48538cc350eb537ffd0a378f15cfc82e1ccfd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 11:19:37.000000000","tz":120},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66facac6e9345fe562e84a6449830ddae3d91165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66facac6e9345fe562e84a6449830ddae3d91165"}]},"branch":"refs/heads/master"},"5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-27 11:35:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/5"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 11:32:02.000000000","tz":120},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d"}]},"branch":"refs/heads/master"},"5c81f94d1995d78065c2eb0b397a4f8dcf7e81ed":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-12-08 12:51:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/6"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2022-12-08 12:51:31.000000000","tz":0},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c81f94d1995d78065c2eb0b397a4f8dcf7e81ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c81f94d1995d78065c2eb0b397a4f8dcf7e81ed"}]},"branch":"refs/heads/master","description":"Rebase"},"4db10d48acee0418bcbbc6e4026349f2876d1050":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-01-12 12:54:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/59/865259/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/865259/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/865259/7"}}},"commit":{"parents":[{"commit":"6238ad6efc88fe99bbb80d86064854fec2321b87","subject":"Merge \"Move general methods under utils file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6238ad6efc88fe99bbb80d86064854fec2321b87"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-22 10:05:47.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-01-12 12:54:44.000000000","tz":0},"subject":"[RBAC] Add share type extra specs tests","message":"[RBAC] Add share type extra specs tests\n\nThe tests validate policy of share type extra specs actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I26303d012cc2c75233cc01cd297f364d4cd0ce7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4db10d48acee0418bcbbc6e4026349f2876d1050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4db10d48acee0418bcbbc6e4026349f2876d1050"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
