)]}'
{"id":"openstack%2Fmanila-tempest-plugin~865535","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I759fe2b5048aea466cab9107cca68d70548bd347","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-12-13 21:48:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-08-10 23:50:27.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-07-19 13:48:54.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-08-11 01:09:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I759fe2b5048aea466cab9107cca68d70548bd347","subject":"[RBAC] Add share group type tests","status":"MERGED","created":"2022-11-24 12:06:04.000000000","updated":"2023-08-11 01:09:15.000000000","submitted":"2023-08-11 01:09:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865535-secure-rbac","meta_rev_id":"6c61cc706cc182c8615b4f55b1b17b165b1c12e7","_number":865535,"virtual_id_number":865535,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 01:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-08-10 23:50:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-07-19 13:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-08-10 23:50:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 13:50:54.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-11-24 13:50:54.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-24 13:50:54.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-11-24 14:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 16:36:15.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-29 19:43:15.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-18 12:48:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9194ad8fc879c5abf2b8645b66bf451673d9fee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-24 12:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9bb89e83e21a65fb1811eabd8b371d3146e3d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-24 12:55:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3315d967df36781a5542a27d95ab3fbd499b610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 14:45:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8330049bccc8496fa89361990f261063\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc197133a9f45f0a7acc08d40914777 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e29351b5a7c460da06cb843b84ad768 : SUCCESS in 58m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2acac92e014d4f07972146b853cf73b2 : SUCCESS in 1h 45m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/60c7688f66bb47fe9b3a234fec22770b : SUCCESS in 1h 00m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0075d4ad75b5457c8b6d5d7610727d97 : SUCCESS in 1h 11m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/80517c5ead234e36b32612b9a0720e6e : FAILURE in 1h 38m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2be8785acb39410c81b8fbd5face2168 : SUCCESS in 1h 06m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43d582c834e240cab03af6fda4edd7ba : SUCCESS in 23m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9547c73ced0d4382b65e2e4d0a033e73 : SUCCESS in 32m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/88b3228304604ada870339cb5c136b11 : SUCCESS in 1h 08m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/faf3818364d048ddb9e76a52e76ea360 : SUCCESS in 56m 40s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/eccce66e025b420bba6658040cc1b5cf : SUCCESS in 1h 00m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/54c58370e9d34b7d8a24971bec45ade3 : SUCCESS in 21m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1d032bb6aff547169f2440b21ccf543b : SUCCESS in 46m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c16ee0bbc34543e8bb4d955cb545b7e1 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a03b25765198650909392f43bb15642408ae3b4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-24 16:36:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/2/upstream-check/manila-tempest-plugin-ontap-dhss/6bc6b32/ : FAILURE in 1h 49m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/597a65b/ : RETRY_LIMIT in 10m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cf5b1d67a44e2aa320002eed9ba2cba1164cfef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:35:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf963d069ec6d04f851611d1fb709ababcd0dc45","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:35:50.000000000","message":"Topic s-rbac-share-group-type removed","accounts_in_message":[],"_revision_number":3},{"id":"fc764e2629d8bca11f51558be28960683f2d07a7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 11:35:57.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":3},{"id":"ea7badfeca868e2a04200f37bdafc140c25e877f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 13:15:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"078cac2e1fb98e9f518f47a6e843ca412507ffe9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 13:15:34.000000000","message":"Topic s-rbac-share-group-type removed","accounts_in_message":[],"_revision_number":4},{"id":"360152697eec2315493c79e4e5c0c4cd9968b3a3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 13:15:40.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"22a0968b4673a3669a1759217e663d36fdc73909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-27 15:35:38.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56737c274c354af991880eff37261fa9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99d32a62900490c8da628b05d925dbb : FAILURE in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ceccdbe96fca4838a22e5b69cce52afb : SUCCESS in 1h 30m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7abbd3ccb66947dbb5f06849f6b43d7b : FAILURE in 2h 15m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/46ac8d121d004f8882f77e01461f4c37 : SUCCESS in 1h 12m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9c363cc59b154fcd81c07a3bb717a7ff : SUCCESS in 1h 06m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e3a57d821694183be2747114d558b0b : FAILURE in 1h 24m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6497cf5256ea4315852fecf2a91da8b1 : SUCCESS in 56m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a3bff6189f584bbd8740a796fff0bd52 : SUCCESS in 24m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2dd3e64155014ac386d80585be9b62a6 : SUCCESS in 31m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/473c046aba73401eb6f1b32e2cc9a3f3 : SUCCESS in 1h 15m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2c77f3d5d2154b8c9a4275b5a34b4d07 : SUCCESS in 55m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/07135a585179451ab6bf9ecb8d87417e : SUCCESS in 54m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6f3f79becec2442e93dcffb1103cac94 : SUCCESS in 21m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d685bf47e7b5483b87377430f4ca11a8 : SUCCESS in 50m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c6beb5276d8a49768907421a00d9e421 : SUCCESS in 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"732d42cbff2815e85322360f20cf904ef8a1af90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-27 17:29:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c87d6ef4367e9569d07a0a3380bcfdb16d1969f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-27 19:29:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2e07da41fb484c8e63268541282cc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b356c0910044af9d34d3629f16761e : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a99658eda2114ff792131c84d31e0fd5 : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ee52e7f14fc9459fb320305782bafc4f : SUCCESS in 1h 49m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1af05bf6e2334368abd17071fb041e39 : SUCCESS in 1h 22m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/42b49763988e4a92b16aa4a02b88e2d7 : SUCCESS in 1h 19m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d7006b4fe74e48d498fd6be4dbc8eab3 : FAILURE in 1h 46m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/687d50f45cd04ff3b9b8654b1beaac3f : FAILURE in 1h 18m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/36634c8bb4494b908dfb0e9cbb9de07e : SUCCESS in 34m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ca32f38d05ac4854bac931b84d19da07 : SUCCESS in 40m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9fce9f7efdca4e85813a515f34431581 : SUCCESS in 1h 26m 57s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c4198a685956491590c232806bc2d144 : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/790a9c76ac3d40f284bf06932505d1f1 : SUCCESS in 1h 15m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/710417d561dd4a698668436258a228db : SUCCESS in 31m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c62be931fb1b4512866427851be6aa57 : SUCCESS in 55m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5e76777402fc487faca4b6c8fd393da6 : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a7de453355840f2cf145e37954a2dbe3b6673f3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-29 19:43:15.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/5/check/DellEMC_Unity_Manila_dhss_true/239495f/DellEMC_Unity_Manila_dhss_true/1053 : FAILURE in 2h 33m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/5/check/DellEMC_Unity_Manila_dhss_false/2cb3dfe/DellEMC_Unity_Manila_dhss_false/1124 : FAILURE in 1h 01m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"11b339f3f6be534e22d6f99c9489426872d571a7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-08 08:26:35.000000000","message":"Topic s-rbac-share-group-type removed","accounts_in_message":[],"_revision_number":5},{"id":"694a40ce4c1c1d1b94c0fe10474ff7b800c870fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-08 08:26:42.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":5},{"id":"cc0cf0acd8ad655f38eacf1447d2d237e1907518","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 21:48:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a81c599a0c7e36ae93da0b75099834aa651281bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 09:10:43.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0d3f6b88da25483da6df18a004f907e69382f0c4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-15 10:29:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/6/check/DellEMC_Unity_Manila_dhss_true/0867284/DellEMC_Unity_Manila_dhss_true/1145 : FAILURE in 58m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/6/check/DellEMC_Unity_Manila_dhss_false/60552e7/DellEMC_Unity_Manila_dhss_false/1215 : SUCCESS in 1h 18m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"74d6750360c0a15cf7756ce0bd7d19a98379b02c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 11:30:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ca43f8108f2456fa08388e443459d93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b05849b383749e4a83ca48d891aeb6b : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6513e7e3d78f4189aad028a5bcfdea0a : SUCCESS in 1h 40m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d909ee0b3e5445378cc8cebfc29ce740 : SUCCESS in 2h 14m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/430c91429d9740f5b718461a6cdedf5a : SUCCESS in 1h 22m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ed369d0765be401a9e3c168cbe371523 : SUCCESS in 1h 21m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/61b76b764b83441086433c6c8f54ea08 : FAILURE in 1h 16m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3190eaa38004d108fbd0875800a0f9b : SUCCESS in 1h 31m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0057e3b90d7497dbc2e2ea8b6c63e79 : SUCCESS in 22m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a34fece3e5614e029d5e4ee9cde2f5fe : SUCCESS in 36m 55s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4f98a50855854e38876f5d201e649893 : SUCCESS in 1h 00m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/19a542bc77c94339ac3d25daca9406c7 : SUCCESS in 45m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/987ac007f5584339bfbdf125068611a5 : SUCCESS in 48m 42s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/003b9996a5ab403dacc2d9e3c583cd1a : FAILURE in 17m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/11af6d074e6f443490bdecbde98d1ab4 : SUCCESS in 38m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/df8226c8e18345018f23759de168bef8 : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7d3250315c36fc59507136766828faa11fc242ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 12:24:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f413753c3248cfa378b0246ddfa920dff991363d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 12:27:06.000000000","message":"Topic s-share-group-type removed","accounts_in_message":[],"_revision_number":7},{"id":"9d87f9793ab35dbdd2137624f6e10806e0b8bcb1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 12:27:22.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":7},{"id":"22722e889983065735d0ef754bfc43f726af05ee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-15 13:24:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/7/check/DellEMC_Unity_Manila_dhss_true/a798121/DellEMC_Unity_Manila_dhss_true/1147 : FAILURE in 56m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/7/check/DellEMC_Unity_Manila_dhss_false/4a1757c/DellEMC_Unity_Manila_dhss_false/1217 : SUCCESS in 59m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4b3906a50abd590d1af7a0382d74ba292db4e22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 14:47:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce46b7b7cbc34c91b165948ae9f64d92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1dfdefae3d340cbac8d3034c7eb0cc1 : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/218ecb8e643c438fb779aa35d44f6f8a : SUCCESS in 1h 15m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b64d559385c04dfd800063f8af8bf0f2 : FAILURE in 2h 17m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3137fa02e7a64f9c9469eb325c8ca888 : SUCCESS in 48m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e91de78259bb45e89f69d28670faeae8 : SUCCESS in 1h 22m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2a8073d8f5c44f459ecd727946a77160 : FAILURE in 2h 06m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cbfe26b96eac4f44a530e489547113c1 : SUCCESS in 49m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e9cf81882b21493793eb9460cc957d15 : SUCCESS in 35m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/252def67050748c8b8155c9078fd6ec9 : SUCCESS in 26m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0456426bdad149ef8144ae7c94759908 : SUCCESS in 56m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d6f83539469a4e05b87c1afd3b87a621 : SUCCESS in 48m 13s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/121a9cc545cd4e599f5756dc34ed9b28 : SUCCESS in 56m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/91739a5960bc41cdb954062404ae81bf : SUCCESS in 19m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a41b2eef3a44921ad6c5e95fa4d13f8 : FAILURE in 55m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/caca090606c0434abd55aa7e8f262798 : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4bba71aba45cb99e311f7745f4b08171fe2ea53","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-15 15:45:44.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/7/check/DellEMC_Unity_Manila_dhss_true/e276edf/DellEMC_Unity_Manila_dhss_true/1148 : FAILURE in 56m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/7/check/DellEMC_Unity_Manila_dhss_false/2b55683/DellEMC_Unity_Manila_dhss_false/1218 : SUCCESS in 57m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"9b8aa975fa589983e41d71d8e5a4d963ddee3928","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-15 16:27:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/7/upstream-check/manila-tempest-plugin-ontap-dhss/3f70b32/ : SUCCESS in 1h 38m 57s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/cdcee9b/ : SUCCESS in 1h 19m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"b69069360c8c3e4a316232579775852334a97ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 11:11:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6fcf9666558e076a02787b6f8f181b9fb7d52a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 12:59:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4c53803e62bd26d1ceb11f844a1eb53cc0648e74","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:00:36.000000000","message":"Topic s-share-group-type removed","accounts_in_message":[],"_revision_number":9},{"id":"7e536327e80f3fc799587afdec53a96ecdec4f70","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:00:42.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"96ea219f65785c0c42976ea571ffece924bc6fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:25:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"91f0a89bf693ca1858aa2ef674e81f786f04f367","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:51:19.000000000","message":"Topic s-share-group removed","accounts_in_message":[],"_revision_number":10},{"id":"ca6f97146279b5dceb7bb9a48acbbcdab8a4323f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:51:26.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":10},{"id":"4efb9378a56c97132e59a92f022280e716e6caa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 15:40:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6503d8b03ecd4f1ea8912a1206e5b3da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c4975c9b6245ec9ef144c49f449590 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0207111c56af4559b467697778e48b6d : SUCCESS in 1h 22m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/417ebd6c48a64be59357c6e55c88d2a6 : FAILURE in 2h 09m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/24021c685d9a4084aaafca9da4b54af0 : SUCCESS in 52m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5813c30e78444bf1b1f44aebc1e7d60c : SUCCESS in 1h 23m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ec706839e11d41f6814159c0425dfe85 : SUCCESS in 1h 43m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f1d4419ef2b64544a6730ea6b73a874a : SUCCESS in 1h 11m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0961f89decf94e22a427268f3eb14230 : SUCCESS in 23m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cc10f799ac8b46c9b2ff69ff753f6295 : SUCCESS in 45m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/71c17ff8da9f46cdadc74805cb357311 : SUCCESS in 54m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5dc0afe89237457d98cd9eaaea434c98 : SUCCESS in 1h 37m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1b48a95bb390496884cede48a30bf277 : SUCCESS in 52m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/287cf6e927ae407aa114571cbb828ac0 : SUCCESS in 32m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/022a27f72a6545a79d4052a6bf7c5740 : SUCCESS in 37m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/637536322cc24db39dad48fc9fdf87bc : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3dd7a5f5f35c06f47d23f47e405f6758d01b66e3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-29 17:29:04.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/10/upstream-check/manila-tempest-plugin-ontap-dhss/b3cc761/ : SUCCESS in 1h 47m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/10/upstream-check/manila-tempest-plugin-ontap-no-dhss/53ac325/ : FAILURE in 1h 17m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"74edeb493eec85988cbeb53014ecebb63002d533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 06:50:23.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"76cd593231d60efa7a8ef16823679aa16e1f784c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:29:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf6effdc0c04f60bb24f1aa7496c943\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72877f3bafc643d8baba07f811d6c3b2 : SUCCESS in 2m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/35a31bdd610345c58a8e64ea29766d6c : SUCCESS in 1h 33m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/886a1694b7014f3f91e1f3026197394a : SUCCESS in 1h 38m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e987d9dabf9d414ba49c1442cba55a66 : SUCCESS in 1h 28m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7672be20c6ba4b68b42253de59025e2d : SUCCESS in 1h 25m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/390953ca467b4164bb875b1d1afd7f57 : FAILURE in 1h 36m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/50a8964a27384b6cb5713248c930ba34 : FAILURE in 52m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/01bb290e369c4597825861072f4dbf09 : SUCCESS in 38m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9bbfd969626f4f16aca97bdeee99308d : SUCCESS in 38m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/964c4c8cf71a4997aa1fbbd32d08416e : SUCCESS in 1h 06m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/28cb267acda74277ba13138e138c0b3c : SUCCESS in 1h 17m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2785706b816c40c1853066db8f160c4e : SUCCESS in 34m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/413233479f67428fa777d409e571a6e9 : SUCCESS in 40m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9972d558ce044880b8a637e57c616993 : SUCCESS in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"266dad97a6226130cdeba2225434a10574498e4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-30 11:06:35.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/11/check/DellEMC_Unity_Manila_dhss_true/5d4c24c/DellEMC_Unity_Manila_dhss_true/1449 : FAILURE in 1h 05m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/11/check/DellEMC_Unity_Manila_dhss_false/765b9f9/DellEMC_Unity_Manila_dhss_false/1550 : SUCCESS in 56m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"09bfe7cfa461b2c1163095184fb5c08fbaaaa222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-18 12:18:39.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"530d6b549d99226f151c4354e1e9c56ef79d387f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-18 12:19:31.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e740b19505162030dcf4137cba8fc1061e8a9dfc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 12:48:49.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_PowerFlex_Manila/c6ab936/DellEMC_PowerFlex_Manila/45 : FAILURE in 28m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"507789d651dfd67aa6667c7cc5ecc31089b639d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 13:47:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/707c786e65084c689a3ad027c5b155e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a839ed6e1ca54e0aa07f0c53ebda0490 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e9babc314c8c425c9b89f65ca094a352 : SUCCESS in 1h 19m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b30f8b02bfe84142924698f403ea6584 : FAILURE in 59m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/181a86a242484b9aa528823058ecdff3 : FAILURE in 36m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/267d3409418d48699e63befb924129d8 : SUCCESS in 1h 17m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/06613d03d1bb4e05a25ce08715195103 : SUCCESS in 1h 20m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/42d50a8c5cb34d7991c17d5aa481c762 : FAILURE in 30m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5f33640e1b054d519e5b5d31107456a9 : SUCCESS in 50m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cb69d04f7fd0495aa1577fd5d06f58a4 : FAILURE in 11m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c8cbf95f178e44fbb1d15270246c4eef : SUCCESS in 1h 12m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/18c2c05b4e274649af40b2212e2b26d0 : SUCCESS in 1h 06m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a8eedb6f19d34062a53d75680e8b0526 : SUCCESS in 25m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1ddd30e2bd8a4aa4821821979c27a24d : FAILURE in 32m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/06336498ade844d3a97a245010b17aac : SUCCESS in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d04a6242bf1d73e1c6f3e0d125eeee5a2e3d5793","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-18 14:56:03.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_Unity_Manila_dhss_true/14eb48c/DellEMC_Unity_Manila_dhss_true/244 : SUCCESS in 2h 32m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_Unity_Manila_dhss_false/bf11340/DellEMC_Unity_Manila_dhss_false/262 : SUCCESS in 57m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"bfe8d9190f4583ff9115da4ee5be869f8f71aaff","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-19 07:37:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"46d7ec5b6d638ed4f3f01cc275769d939b3dc6e2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-19 07:44:47.000000000","message":"Topic share-groups removed","accounts_in_message":[],"_revision_number":13},{"id":"c548c605a1c982eecad1a908c08c1917e8716de3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-19 07:44:56.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":13},{"id":"8ccb98912aee424c8866c7b926447c50e4163e49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 09:22:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c3c304ba05476ea4a4e6d7457f32ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb6b2bc57f74751aecd9aae62e95970 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c0294ee0625047eab15f4858a20d6225 : SUCCESS in 1h 29m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d1071743a12542b28084ee25f4947cc7 : FAILURE in 35m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c5ffdeaf46b3469dbe97b2e758ada38f : FAILURE in 49m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0babdd7c4be84ca7bb9817ed4b124124 : SUCCESS in 1h 15m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1f1b8c175d1e4288a6e7f86f2dd21a3f : SUCCESS in 1h 23m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a0d78799dcaa46e9a32a9856b6e48033 : FAILURE in 14m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/872a6b535441407e9fc19ccae49fb646 : SUCCESS in 38m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3d32d2bb83e641b58d5714a5ccfe7e75 : SUCCESS in 44m 09s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/17fdeac55a3645beb978a26c8411375c : SUCCESS in 1h 33m 02s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/86443a3367e4471699e0113d319efbec : SUCCESS in 56m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/88fd6a28f2dc43bab1e368af70950784 : SUCCESS in 36m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/282c2aef1cb14c7ea5ccb23b3536e6b9 : FAILURE in 17m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2f5a9e29051c4bcfafd931f8638f4774 : SUCCESS in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"661f4111b889394cd50b30144a369f7e76cc95a6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-19 09:47:12.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_PowerFlex_Manila_dhss_false/98a38da/DellEMC_PowerFlex_Manila_dhss_false/57 : FAILURE in 24m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"301e265f59e4102d15ab4545de562f38ca2fe6ee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-19 10:09:14.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_Unity_Manila_dhss_true/2a252c4/DellEMC_Unity_Manila_dhss_true/252 : FAILURE in 46m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/865535/13/check/DellEMC_Unity_Manila_dhss_false/cbceccc/DellEMC_Unity_Manila_dhss_false/270 : FAILURE in 33m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"c49c1cdfaeee03bd3ad07e6ebef94cb6a2b6a1ed","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-19 10:34:44.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/13/upstream-check/manila-tempest-plugin-ontap-dhss/1ed0225/ : FAILURE in 1h 11m 42s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/865535/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/7ec5947/ : FAILURE in 1h 10m 22s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"aa38e91aa694d1ea0f3d1083212de8014f4e9169","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-07-19 13:48:54.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"09c11630511e345b1e2eb6eae6dacd5da326234b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-10 23:50:27.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"41154f23418fa067b218807fa54646bbd7f3f3f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 23:50:57.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"c801da11b9b81935b1c5708237dc944edfb86a40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 01:09:15.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639f65a6e22b42af829be954b087533f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b452acb645841ae9aa9c549d0537bfa : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0805d98512cc40b497cfc5ac4daacf59 : SUCCESS in 1h 13m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fb302c0ee37e4d2a89cfda8627f37ef9 : SUCCESS in 23m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78cc03e751c34f00ad136bf238225dfa : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":13},{"id":"6c61cc706cc182c8615b4f55b1b17b165b1c12e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 01:09:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04","revisions":{"b4cdc2ed37a56c4cc0cfaa3e0aa239fc5ff17248":{"kind":"REWORK","_number":1,"created":"2022-11-24 12:06:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/1"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4cdc2ed37a56c4cc0cfaa3e0aa239fc5ff17248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4cdc2ed37a56c4cc0cfaa3e0aa239fc5ff17248"}]},"branch":"refs/heads/master"},"dc1a71d44743db9da4f039b3c5542bc9f664a33c":{"kind":"REWORK","_number":2,"created":"2022-11-24 12:55:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/2"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:55:23.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc1a71d44743db9da4f039b3c5542bc9f664a33c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc1a71d44743db9da4f039b3c5542bc9f664a33c"}]},"branch":"refs/heads/master"},"d17cbf11f65b3fce09d655241b8f51f100be6294":{"kind":"REWORK","_number":3,"created":"2022-11-27 11:35:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/3"}}},"commit":{"parents":[{"commit":"5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d","subject":"[RBAC] Add share type extra specs tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 11:35:25.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d17cbf11f65b3fce09d655241b8f51f100be6294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d17cbf11f65b3fce09d655241b8f51f100be6294"}]},"branch":"refs/heads/master"},"44f03344574386d5c7873668a08664573f8c461f":{"kind":"REWORK","_number":4,"created":"2022-11-27 13:15:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/4"}}},"commit":{"parents":[{"commit":"5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d","subject":"[RBAC] Add share type extra specs tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 13:14:54.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44f03344574386d5c7873668a08664573f8c461f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44f03344574386d5c7873668a08664573f8c461f"}]},"branch":"refs/heads/master"},"287e44b85f70b652406ee1a6a98d229ee74adf6f":{"kind":"REWORK","_number":5,"created":"2022-11-27 17:29:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/5"}}},"commit":{"parents":[{"commit":"5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d","subject":"[RBAC] Add share type extra specs tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a1c0b06c732fbd00a10ebb3cac21dfb70ff405d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-27 17:27:33.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/287e44b85f70b652406ee1a6a98d229ee74adf6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/287e44b85f70b652406ee1a6a98d229ee74adf6f"}]},"branch":"refs/heads/master"},"c1e2199f027a5884a094bf2711d4b42cddff6a07":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-15 09:10:43.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/6"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-01-15 09:10:43.000000000","tz":0},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1e2199f027a5884a094bf2711d4b42cddff6a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1e2199f027a5884a094bf2711d4b42cddff6a07"}]},"branch":"refs/heads/master","description":"Rebase"},"9c92df53f7b66e0cd4d844a3f6bc1459c8c3d2b2":{"kind":"REWORK","_number":7,"created":"2023-01-15 12:24:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/7"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-15 12:21:03.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c92df53f7b66e0cd4d844a3f6bc1459c8c3d2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c92df53f7b66e0cd4d844a3f6bc1459c8c3d2b2"}]},"branch":"refs/heads/master"},"88d91966ee47f7d2a96a4ee90c0a6283b58607eb":{"kind":"REWORK","_number":8,"created":"2023-01-29 11:11:06.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/8"}}},"commit":{"parents":[{"commit":"314ad730ca9e8eddc190212713db09dc741ee86b","subject":"Fix incorrect url request for share_group_reset_state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/314ad730ca9e8eddc190212713db09dc741ee86b"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-29 11:09:32.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88d91966ee47f7d2a96a4ee90c0a6283b58607eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88d91966ee47f7d2a96a4ee90c0a6283b58607eb"}]},"branch":"refs/heads/master"},"97f41fc51660b57af6e92fbfbdcb0579258843fe":{"kind":"REWORK","_number":9,"created":"2023-01-29 12:59:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/9"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-29 12:59:13.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97f41fc51660b57af6e92fbfbdcb0579258843fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97f41fc51660b57af6e92fbfbdcb0579258843fe"}]},"branch":"refs/heads/master"},"50eebc0ed0b5176253ce7288755fdf1f365a5709":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-01-29 13:25:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/10"}}},"commit":{"parents":[{"commit":"314ad730ca9e8eddc190212713db09dc741ee86b","subject":"Fix incorrect url request for share_group_reset_state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/314ad730ca9e8eddc190212713db09dc741ee86b"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-29 13:20:35.000000000","tz":120},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50eebc0ed0b5176253ce7288755fdf1f365a5709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50eebc0ed0b5176253ce7288755fdf1f365a5709"}]},"branch":"refs/heads/master"},"8c9cfd7123f2d5a77ed093ce39792023ce244bcc":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-03-30 06:50:23.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/11"}}},"commit":{"parents":[{"commit":"3d6c65f7e87915cf59b8650bfef5b6a45b5d8f0d","subject":"Fix incorrect extend syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d6c65f7e87915cf59b8650bfef5b6a45b5d8f0d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-03-30 06:50:23.000000000","tz":0},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c9cfd7123f2d5a77ed093ce39792023ce244bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c9cfd7123f2d5a77ed093ce39792023ce244bcc"}]},"branch":"refs/heads/master","description":"Rebase"},"195fac5f62351662aebce4540595f2f8a1fba70b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-07-18 12:18:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/12"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-07-18 12:18:28.000000000","tz":180},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/195fac5f62351662aebce4540595f2f8a1fba70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/195fac5f62351662aebce4540595f2f8a1fba70b"}]},"branch":"refs/heads/master"},"796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04":{"kind":"NO_CHANGE","_number":13,"created":"2023-07-18 12:19:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/35/865535/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/865535/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/865535/13"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:05:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-07-18 12:15:43.000000000","tz":180},"subject":"[RBAC] Add share group type tests","message":"[RBAC] Add share group type tests\n\nThe tests validate policy of share group type actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: I759fe2b5048aea466cab9107cca68d70548bd347\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
