)]}'
{"id":"openstack%2Fmanila-tempest-plugin~865536","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I783194a10b0f39823def07e5af93f36168c5a217","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_group_type_id","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-11-29 19:08:50.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-11-28 21:59:16.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-11-30 11:10:09.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-11-30 08:23:10.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I783194a10b0f39823def07e5af93f36168c5a217","subject":"Fix argument name in \"delete_share_group_type_spec\"","status":"MERGED","created":"2022-11-24 12:12:53.000000000","updated":"2022-11-30 12:33:38.000000000","submitted":"2022-11-30 11:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865536-share_group_type_id","meta_rev_id":"4f4d6ccf9caec3d750c80575ecc600e59f9f3826","_number":865536,"virtual_id_number":865536,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 11:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-11-29 19:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-11-28 21:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-11-29 19:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 13:50:40.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-11-24 13:50:40.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-11-24 13:50:40.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-24 14:12:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 14:59:14.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-30 12:33:38.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"a6aabff2514e77778a718a8e13c85951ba5197ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-24 12:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61cbaf696442bf81138269bdf362c6319f4f4291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 14:12:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27787c8d85b24f2a9898553b0f49850f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7b192c9c8a442bb90e3cba4d5f602c : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f8d05182bb049ed91e2aa8d13496d99 : SUCCESS in 1h 02m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b60f69df7dbf42ea892222f20bf0ec44 : SUCCESS in 1h 58m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3e4a6be48cf4429f9f56fa79b946d909 : SUCCESS in 59m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/896acb3bcd8f42b8a239c28f664b4608 : SUCCESS in 1h 07m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/764863eddc764c329ba703fdbd82db39 : FAILURE in 1h 36m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aff3bbda2e38465bb60d47901197a96c : SUCCESS in 1h 02m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/30c96c412be94f7a828d5639c86d7bc9 : SUCCESS in 26m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/40fd79b155bf4bd1b25329dda94f56f6 : SUCCESS in 34m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/87d8d0587c394059b6439a98e4eb396f : SUCCESS in 1h 10m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/47e8b973673d4c0fbfe6091713e17fe9 : SUCCESS in 50m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8938a1be01c34f05bf74a5cc9785a3e3 : SUCCESS in 1h 02m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4e9569b65128413facb32d82ea09df39 : SUCCESS in 19m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/882000d5ad6442fc9b7e05dfc255fc5f : SUCCESS in 36m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/90d0afbb2d37435f84235925945b76e2 : SUCCESS in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fc48f1f882d68c258ddb3e4476268c94c49ced1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-24 14:59:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/865536/1/upstream-check/manila-tempest-plugin-ontap-dhss/1d3dbcc/ : FAILURE in 45m 53s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/865536/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/86e0a31/ : RETRY_LIMIT in 10m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b70fa91e90aed706187b1213f9df68b79894ae9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-28 21:59:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"246c0303fe3fe0a4baaab090f6027752ffea9b15","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-29 19:08:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3b2ae97673d71851f6643418d94957c10376c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:09:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"527bf6d21a9c72652841c0f3326a99e547df853e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 20:34:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc35fb7cf192464e9e80e3159a1f64ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0cffb85358e4ac3944f7e44e2d29bf9 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a719beef56ad4e37805af39091a35d7a : SUCCESS in 1h 24m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1384bdbb64fe4ffb9596f0263674756d : SUCCESS in 24m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c79885d4571b461fb8a281ef71903097 : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3163a7cf37a7fb5a3ec5244e6dc1bb928fe2c384","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-30 05:37:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11bb02d81024583170403ebeae13aec0767ac4e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 07:35:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf2ffcaa9eca455fb2cb5f69eccaee94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a71bb01945748a39e5a368f73e483f2 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/caa7454a94774dc980451530afc6124d : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/50ca510f3b004f0fb5a5b39b47179b3b : FAILURE in 1h 55m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/52be309517ed41af964d336a54a10a62 : SUCCESS in 1h 05m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8eaba2096b5943c09cf7473fc6df12df : SUCCESS in 1h 31m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/61eb91689c4b4cb488596ace71b3ad47 : FAILURE in 1h 47m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/96fb79ca6ea948be89128c53b40d303c : FAILURE in 1h 05m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8bfc8c2981f84603aecff358261114e9 : SUCCESS in 25m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/05328bbd721f45a9bd11967a94f2e54a : SUCCESS in 27m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/faa5eba04cd04b37a91a6b891ede7c60 : SUCCESS in 1h 13m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d3ca8bfff09a4fafa364f023528fec2c : SUCCESS in 53m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d2003cead61b49888e075bff42f154b9 : FAILURE in 57m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a66472176b064ba6bc180793e985f5cd : SUCCESS in 21m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2b8a659c5d854b31b3985c7d21b2d449 : SUCCESS in 42m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4969c12340a8407888fded0e45be9522 : SUCCESS in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca0eb771dfa8abb8428c1d8e7ef3623791214396","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-30 08:23:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e855701090f481f19484c744dd3f567163da49a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 10:07:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa44275e88964df98c6ea450e96d4bc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1891f8b9b1a848c09191e0dd617e453d : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ab63c5d7675e4fc480354be86bd12427 : SUCCESS in 1h 05m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a0d0ec0fb274fd3895e5881ef565c8b : SUCCESS in 1h 43m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/04fcba28a6de4cda8842b0446d83202d : SUCCESS in 58m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/748f49970e6d4a8e919832ef79f4f6d5 : SUCCESS in 1h 11m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d97107b00c6849ba9cbb531dc7cec376 : FAILURE in 1h 31m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b4d067b7847411e9cd49d8c6b0bfd49 : SUCCESS in 1h 02m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/272a11929a07488b8047aef0fd71b92d : SUCCESS in 21m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f86b2a8cf334c0ab6e1c0085674826c : SUCCESS in 26m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/dc3473ec47214ab9a33aed4b7fbb183e : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c4432d15388c468ea3c117dd00af429e : SUCCESS in 55m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/be3719a962624649bfd8bd4da5d3ea22 : SUCCESS in 53m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6fb54918d2644f3d9e09312f8bd185bb : SUCCESS in 20m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7c3a6058243f4f579d1748a0a2568e41 : SUCCESS in 39m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b2af2dafa0244dc0b2c224ef9aae9b20 : SUCCESS in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4e62749dd37e29389475bcd75f51dc87b263e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 10:08:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f848fb6fb87f4d30c808c4d46143587e7474a02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:10:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07e33a5e996942b3b163710b47c168c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdfae989642494dab10cf28f6c50162 : SUCCESS in 3m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ca2476712c8541c0903af52d79de08b0 : SUCCESS in 1h 00m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7b4f8bb65e648bb957bce28a6cbf905 : SUCCESS in 27m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/65f12d7888304ae5859e64ea9baffd55 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d727d0041a9056bdff7796f469ad43fb8cbb8f2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:10:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f4d6ccf9caec3d750c80575ecc600e59f9f3826","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-30 12:33:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/36/865536/1/check/DellEMC_Unity_Manila_dhss_true/84ab48e/DellEMC_Unity_Manila_dhss_true/1058 : FAILURE in 2h 25m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/865536/1/check/DellEMC_Unity_Manila_dhss_false/445a90a/DellEMC_Unity_Manila_dhss_false/1130 : SUCCESS in 1h 09m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b729094ac1b9e1d5f664a163052e8d5e4a62fdbe","revisions":{"b729094ac1b9e1d5f664a163052e8d5e4a62fdbe":{"kind":"REWORK","_number":1,"created":"2022-11-24 12:12:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/36/865536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/36/865536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/865536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/865536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/865536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/865536/1"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:08:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-24 12:08:37.000000000","tz":120},"subject":"Fix argument name in \"delete_share_group_type_spec\"","message":"Fix argument name in \"delete_share_group_type_spec\"\n\nThe method is in the scope of group so it should be\nnamed as \"share_group_type_id\".\n\nChange-Id: I783194a10b0f39823def07e5af93f36168c5a217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b729094ac1b9e1d5f664a163052e8d5e4a62fdbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b729094ac1b9e1d5f664a163052e8d5e4a62fdbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
