)]}'
{"id":"openstack%2Fmanila-tempest-plugin~865811","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I027460222b7d34bb7ad00408739ab00207be0386","project":"openstack/manila-tempest-plugin","branch":"master","topic":"sgt-extra-specs","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-11-30 19:28:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-11-29 19:57:50.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-11-30 20:40:46.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-11-30 20:40:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I027460222b7d34bb7ad00408739ab00207be0386","subject":"Fix incorrect url request for share group type extra specs","status":"MERGED","created":"2022-11-28 09:41:55.000000000","updated":"2022-11-30 20:40:46.000000000","submitted":"2022-11-30 20:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"865811-sgt-extra-specs","meta_rev_id":"a77152838ff2ac90ebdd9b390d1c0be2856289f9","_number":865811,"virtual_id_number":865811,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 20:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-11-30 19:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-11-29 19:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-11-30 19:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 11:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 11:43:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-11-28 11:43:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-28 11:43:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-11-30 03:51:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"dcf6eb9a2e21126de03488ba50bb55a62e8e6ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-28 09:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15014dc90817e7b765f956272e99571fef03de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdbaaad51b424d3ba511398d9d11a50a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7a6119ce7e499fbc7ebc72f83d45ea : SUCCESS in 2m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/78390ad4782b40b3b60790a8c9b67cbe : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/057a02d0c20d4ba59e41805e050c91a4 : SUCCESS in 1h 40m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8b4e160721f74a10866094da1bae0d84 : SUCCESS in 58m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e841cc8604fd4c038a08d41c9cf8ddfb : SUCCESS in 1h 03m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/94e8b3dc434f4c33a6c9b7619966a755 : FAILURE in 1h 28m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1fe88e336aec4c97a87761c0432d4f39 : SUCCESS in 1h 01m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7cf597a84ac48c58c598325ded8f143 : SUCCESS in 29m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/76ae40714a714b799ef42b72019b6b2e : SUCCESS in 27m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e85359ef01fb4d3591a8869c199d423e : SUCCESS in 1h 15m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cf55c265528440749358eb4dbe7db0aa : SUCCESS in 56m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7ef9de41277b4afb871b3e2693debcdd : SUCCESS in 1h 02m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cab391344c08421cb45ba78daa8969fb : SUCCESS in 19m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/96b6f677c3e846389f7116838b997e36 : SUCCESS in 42m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d3e7680c1c3347efa0c2d75301bbd10a : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f2402dbe0479eda7532a61c41a77f5d7026fee9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-28 21:56:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9ce0d3f43c50f824655d9afea3bffcafaaa8c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-29 09:44:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"74a4c9912ffc8dad186d91b3f81200b27b0d5b57","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-29 09:45:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"143cec6a541401bc80f571e4de44fada9e1564a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 11:43:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2820aaac3bc4daba7ef4eecd2791adf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e34d757d3f94d0cbaa78a822e30a732 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5e04a0fc4f1c4d6d83689297d6f9e56c : SUCCESS in 1h 21m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/250e1e8a31f54f17ad4164ff815cd0ac : FAILURE in 1h 49m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/275771eff1aa4a9eba609dcbac412a0d : SUCCESS in 1h 27m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f104570b7466440b836307def8cc5552 : SUCCESS in 1h 23m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/64b4aaa309db495497ebc63d79443f64 : FAILURE in 1h 56m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/33a4c788c8944523a67e49e04262e429 : FAILURE in 1h 14m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0189b08ec77b4d98b0b0807339489c6c : SUCCESS in 26m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3e01bb1953d4da0a976c9504a310920 : SUCCESS in 39m 06s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3ae764f47dc64b73879bb3e73824df6b : SUCCESS in 1h 02m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/402e2f91db3e4bfd88ac292f26ce5e38 : SUCCESS in 54m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/36e31947375847dca76969edac601944 : SUCCESS in 58m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8a06036d73e1431db26312601f9b459f : SUCCESS in 21m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a9bebf3c89cb40dc95c290fca3965df0 : SUCCESS in 47m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/297536ea45d549f3bb71c048d48ac5b7 : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c0b8df3efdf3c82b293551ea5f9f7ecefae8381","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-29 19:57:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c0dcfb498eb3b33b3ddbc65aeddd3f9bfc5c9f2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-30 03:51:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/11/865811/2/check/DellEMC_Unity_Manila_dhss_true/c144915/DellEMC_Unity_Manila_dhss_true/1056 : FAILURE in 2h 56m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/11/865811/2/check/DellEMC_Unity_Manila_dhss_false/c2d5802/DellEMC_Unity_Manila_dhss_false/1128 : SUCCESS in 54m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"32987d1d6751a8d63b5dc9028b46533dbe6f9202","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-30 19:28:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f44283d7ae8fa9cdcd72419577c502c3662f7fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 19:29:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9484785a44b118208e01134ef22dd2fb5f5aa4a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:40:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c16ab1e34f4b9f91ebdb9d692fbe81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c122b6d0961f41c98efc0dd8255c6a36 : SUCCESS in 3m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f26bf2a5ca56411dace5b37484a72192 : SUCCESS in 57m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/64389bd6065342da8fdba9e6b5b0527e : SUCCESS in 23m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b7eb4f3697004cba95ddf564d39cde0b : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a77152838ff2ac90ebdd9b390d1c0be2856289f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:40:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a22ce375bda5eb3874973068898723c1a7a40a6","revisions":{"60ff74ace8bd8c22ab641ab43610cc720992dcfe":{"kind":"REWORK","_number":1,"created":"2022-11-28 09:41:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/11/865811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/865811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/1"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-28 09:18:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-28 09:41:42.000000000","tz":120},"subject":"Fix incorrect url request for share group type extra specs","message":"Fix incorrect url request for share group type extra specs\n\nThe url request should be change as below:\n\nfor single extra spec\n---------------------\n\"group-types/%s/group_specs/%s\" \u003d\u003e \"share-group-types/%s/group-specs/%s\"\n\nfor all extra specs\n-------------------\n\"share-group-types/%s/group_specs\" \u003d\u003e \"share-group-types/%s/group-specs\"\n\nRelated-Bug: #1998088\nChange-Id: I027460222b7d34bb7ad00408739ab00207be0386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60ff74ace8bd8c22ab641ab43610cc720992dcfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60ff74ace8bd8c22ab641ab43610cc720992dcfe"}]},"branch":"refs/heads/master"},"1a22ce375bda5eb3874973068898723c1a7a40a6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-29 09:44:20.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/11/865811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/865811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/865811/2"}}},"commit":{"parents":[{"commit":"72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0","subject":"Merge \"[RBAC] Add share export locations tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72042cdf2472aaa2d4ce5b24c7a84aea68fc1fe0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-11-28 09:18:01.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2022-11-29 09:44:19.000000000","tz":0},"subject":"Fix incorrect url request for share group type extra specs","message":"Fix incorrect url request for share group type extra specs\n\nThe url request should be change as below:\n\nfor single extra spec\n---------------------\n\"group-types/%s/group_specs/%s\" \u003d\u003e \"share-group-types/%s/group-specs/%s\"\n\nfor all extra specs\n-------------------\n\"share-group-types/%s/group_specs\" \u003d\u003e \"share-group-types/%s/group-specs\"\n\nCloses-Bug: #1998088\nChange-Id: I027460222b7d34bb7ad00408739ab00207be0386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a22ce375bda5eb3874973068898723c1a7a40a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a22ce375bda5eb3874973068898723c1a7a40a6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
