)]}'
{"id":"openstack%2Fmanila-tempest-plugin~866399","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iecc7b825913dfebe3b5a383bbe56ffc8665ae3eb","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-12-18 11:06:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iecc7b825913dfebe3b5a383bbe56ffc8665ae3eb","subject":"[RBAC] Add availability zones tests","status":"MERGED","created":"2022-12-01 20:56:59.000000000","updated":"2022-12-18 11:06:05.000000000","submitted":"2022-12-18 11:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866399-secure-rbac","meta_rev_id":"e68921c1353fde7050f9cfb72471c647c1cf704c","_number":866399,"virtual_id_number":866399,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-18 11:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-12-13 19:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-06 12:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-18 09:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 23:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-02 01:34:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-06 12:46:49.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-12-13 19:44:33.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"15ad10d49f88c2f882c66d8a4101888439b8cf27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 20:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c54c285c2bcba7f7a4347f64dbf29d8711ea0d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 21:00:45.000000000","message":"Topic rbac-vhari-az removed","accounts_in_message":[],"_revision_number":1},{"id":"81c7e97b6f2d06a8919740a2586fdc871fa4a910","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 21:00:54.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"322d66d5c6456d225c5a760ede2b303613400f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 22:13:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0657706301a0f18910763f501830040077a4ddad","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 22:28:33.000000000","message":"Topic rbac-vhari-az removed","accounts_in_message":[],"_revision_number":2},{"id":"934e3c190ea2b9a27197a4ba6f7fb6e2d8e11224","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 22:28:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"4a4132e5afc282560853eda4685bff822403000a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 23:56:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f08d478bad4f7eab267fefe188065e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb9dc2e7c6345ff8dc76f0a3e6474b7 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/823ae18d4a674f9a8614c4364904e39d : SUCCESS in 1h 21m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7987a639ba7f4edcb571fb7b3231ffc5 : SUCCESS in 1h 31m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5b16e656c3034a76941b4a94084527a6 : SUCCESS in 1h 06m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c579390adbdf407f9636d6c751d2cc3a : SUCCESS in 1h 10m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/08f68044bbd4417c89cb650cfff33c53 : FAILURE in 1h 41m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d9fab5bf5444f4ca2801de6d9492054 : SUCCESS in 1h 05m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47d7e683ae144b4cb8513866c165dce0 : SUCCESS in 26m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5fa15ff7aa0542cf8f568888f2fb12e0 : SUCCESS in 27m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/39955d92c00545368eebbb63a6c87f9b : SUCCESS in 1h 22m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b2dbd1d4a2f24c3cb328b08ddbbb5d5b : SUCCESS in 57m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/278b89ec190d4d39aef680363530b11d : SUCCESS in 55m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1092f3350181413ea15758aec5b9238b : SUCCESS in 21m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5363ef8ffcee4815a14162e8e8af05a2 : SUCCESS in 44m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cd929d5289f844608f6853d841751b01 : SUCCESS in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"687b97c544a50d30488fa56146aed0ba3db7da8b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-02 01:34:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/99/866399/2/check/DellEMC_Unity_Manila_dhss_true/ee463ae/DellEMC_Unity_Manila_dhss_true/1065 : FAILURE in 3h 20m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/99/866399/2/check/DellEMC_Unity_Manila_dhss_false/a68dadf/DellEMC_Unity_Manila_dhss_false/1137 : SUCCESS in 1h 02m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"18839e2d1c56e973e4d4a6d9c7be246e82192cdb","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-06 12:46:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f783d4da4ab3d40e69417530238d80af34367cb3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 19:44:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d1f5cc9917a3139262e06d84aa7fc5e98cad22c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-18 09:50:17.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2603453ee0302b0cc04de38d8fb5c693a214901a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-18 09:50:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1200df48cd4ebe3ff09f2a2bf923e7a5539a1707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-18 11:06:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d0e34a27fe4a8ab57343e069b1e618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564059f5d1f14bd98c8b643f5c648e76 : SUCCESS in 4m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46c068332c714923a919d296ffb58526 : SUCCESS in 1h 05m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9f019128a18046af96f047328cccf9f3 : SUCCESS in 29m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff36f7c1e4c843c58651e52fbf65a61c : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e68921c1353fde7050f9cfb72471c647c1cf704c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-18 11:06:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"937338e055c3f83979941aa7510f9ab6407731e2","revisions":{"24fe4042150d6cc75d217ce090b5d3480d6a9a44":{"kind":"REWORK","_number":1,"created":"2022-12-01 20:56:59.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/99/866399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/99/866399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-01 20:56:18.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-01 20:56:18.000000000","tz":-300},"subject":"[RBAC] Add Availability zones tests","message":"[RBAC] Add Availability zones tests\n\nChange-Id: Iecc7b825913dfebe3b5a383bbe56ffc8665ae3eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24fe4042150d6cc75d217ce090b5d3480d6a9a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24fe4042150d6cc75d217ce090b5d3480d6a9a44"}]},"branch":"refs/heads/master"},"937338e055c3f83979941aa7510f9ab6407731e2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-01 22:13:16.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/99/866399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/99/866399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/866399/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-01 20:56:18.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-01 22:12:54.000000000","tz":-300},"subject":"[RBAC] Add availability zones tests","message":"[RBAC] Add availability zones tests\n\nValidate policy of availability zones actions\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iecc7b825913dfebe3b5a383bbe56ffc8665ae3eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/937338e055c3f83979941aa7510f9ab6407731e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/937338e055c3f83979941aa7510f9ab6407731e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[]}
