)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867061","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia951c68f08e76dfad35c65cb49bc94e9614798f3","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-12-20 14:49:05.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-01-19 12:35:58.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia951c68f08e76dfad35c65cb49bc94e9614798f3","subject":"RBAC tests for share network security service","status":"NEW","created":"2022-12-08 22:36:30.000000000","updated":"2023-07-18 22:47:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4b7f23a7aa719ed87313c029bd3bcd18b83abf6","_number":867061,"virtual_id_number":867061,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-18 14:37:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 00:01:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 03:05:03.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-18 10:37:56.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-01-19 12:35:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-18 14:51:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e7b108c3945b429a39cd71d3a665db99191873c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-08 22:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"900d36986754e1b8e5c266d9de4d143808308725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 00:01:29.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0acd8ca5c6044e5591953fb798470835\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a25e4c2762d4b42ac2c4b9ce490cf87 : FAILURE in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/94ec1ace49864aa68d48b16129ab35c4 : FAILURE in 26m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/52c9a2d9aad24e2e8711db07c9200d09 : SUCCESS in 1h 23m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e825f0e0ef247deb7a6cd3d71539f6b : SUCCESS in 1h 10m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/639faf051a674fb5988b1a4cfd83494d : FAILURE in 28m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ae99656d65b74ff994860c1f9f45b580 : FAILURE in 39m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5ef96399f77a4dd2976f1c9146ccbd9d : SUCCESS in 1h 00m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e91a01b4fafd4ee8b739cff13745d5a3 : SUCCESS in 27m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c889acd338e5465cafebc9b61225b4cc : SUCCESS in 33m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8949143249784f02b5a451b28261c0eb : FAILURE in 27m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/223f93f6de8947eb9c09762a3cddc81c : SUCCESS in 1h 01m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/867be4a664744c10878c9439eae0d849 : SUCCESS in 1h 22m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e349d179c1944edd9342d737ae8631be : FAILURE in 24m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07ca0e7ebf1b415dbffd75721d1557af : SUCCESS in 40m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f4bbda3decc94a47918bbf7f4efb8c9f : FAILURE in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6162c7d67aa04f86b0b4053064ab92145fd2fb65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-09 03:05:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/1/check/DellEMC_Unity_Manila_dhss_true/136147b/DellEMC_Unity_Manila_dhss_true/1070 : FAILURE in 4h 28m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/1/check/DellEMC_Unity_Manila_dhss_false/f128567/DellEMC_Unity_Manila_dhss_false/1142 : SUCCESS in 1h 09m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"49c55f52dd667acb93696e3f52494a578af1fbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-09 14:12:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d17dfa3436a50a5a64746ddb8a3d10a89b41c003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 15:49:38.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dfac08b77214459b365f49c5942aa48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036542b6dc654e6ba18f7099505de8b6 : FAILURE in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0195a7a1137f468a90c48401e996be37 : FAILURE in 27m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ec22fe6d86f042928a5d36bc6711434d : SUCCESS in 1h 34m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0d4949360f9c4d4ab75b488761bd728e : SUCCESS in 1h 04m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7b65687f266b435b990c00fcbee52ae5 : FAILURE in 28m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c39b436222a41c088694a2aee765f80 : FAILURE in 23m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a0648571e9d49219ec021e7945972c7 : FAILURE in 1h 02m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/854859eb2250444da4b00fe69c49ab96 : SUCCESS in 24m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7da2ed6ef0c7444d90902b51fa6b368f : FAILURE in 39m 32s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5afe5ae796214d13becb957bc8d04a65 : SUCCESS in 1h 02m 37s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a4ac69ac56a84740ba97601ad7e272cf : SUCCESS in 56m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/53ad79e64a5e427884f460044968396c : SUCCESS in 1h 15m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/71045d49d75b44f88c27195a68064d86 : FAILURE in 21m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4645cb45975744378f57a6b536037ddf : FAILURE in 47m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a802f76699ce40dea6d0109564b65e9b : FAILURE in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94b73c10d6a28bc557f836793c29e5a6cd2ada7e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-09 19:22:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/2/check/DellEMC_Unity_Manila_dhss_true/e01e4f2/DellEMC_Unity_Manila_dhss_true/1072 : FAILURE in 4h 07m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/2/check/DellEMC_Unity_Manila_dhss_false/0b177df/DellEMC_Unity_Manila_dhss_false/1144 : SUCCESS in 57m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e7faf81456bad7a79f7dda7aafc93bd702f245e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-12 19:17:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29ac7aa446fb6e494d97a0b9a28411b6f57e177d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-12 19:36:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"43c04a20992935ff2faf420b854c73c0e9346cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 21:33:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37dff3d5c1ba4a07892d11241f1179b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0f62b9eb104f0e817829b9b4190a91 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f4ab60102ff94271bd6bb0592fe0166a : SUCCESS in 1h 11m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5207d8400da047ad8b1bfa9d91037d6f : FAILURE in 1h 55m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3887fb64eb354416926a47d16c6cc120 : SUCCESS in 1h 06m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/11a4fb3afb614ec28d1e4d30db2a671b : SUCCESS in 1h 14m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a9d0d97df6cb468e98137c0a3b65190a : FAILURE in 1h 33m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a5f74f0c54d49dfb30fef1d07250d70 : SUCCESS in 59m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7e580c553fff4243a6c92b26e9f5e066 : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b00a0061ea6a46a2b6d0f149a408a32c : FAILURE in 30m 19s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d517f7890e1d4442b2816259687e05e5 : SUCCESS in 1h 13m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/787f8867589c44618646a37486f0b4c7 : SUCCESS in 1h 08m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/604a7aa14f854038ace5168bf253b4fa : SUCCESS in 1h 29m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/84f682ce1c6c41b3801c398e53a14017 : SUCCESS in 26m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/70126ad1c05d47188e68a2d405ea6366 : FAILURE in 47m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/31ac30c7c7d447b0bee0306ce910503a : SUCCESS in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b2a3431dc2562e6a1854d03b107b034fea5c600","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 17:59:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"3a6395db02b76239a09a1b47b041c83e385a8969","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-18 10:37:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e9704f42c83559642a7cee04d72fa148d022f4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-20 14:11:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"92f88f79dfb869c49405f5a38b378f549ba2a030","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-20 14:21:13.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cfe98e0504908b002e1ceea76503efbdb0db478a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-20 14:21:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"253d0e3ba9688e23917ff9fcc96847b24f5d5162","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-20 14:49:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"93fd9e18a55212aa4e2fc9e8e7e6c067c09e2f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 15:07:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f8aac04d70646d4b283cf5ed57ea213\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874ad3283d544826b5ac5e44a15fe9fe : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3906ca549f434f5784d3a371f7b0e465 : FAILURE in 32m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d050d39b684a42af8fc3d15e40af8a54 : FAILURE in 38m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7e145503a8f14ef8a34a11a7f14f570c : FAILURE in 39m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8dd890a9c5c140889fb6fff551c44eab : FAILURE in 25m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bf2b48cd8b1e4fe5a66945283696ad52 : FAILURE in 34m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4fc2c725e39542728cea2600227a6560 : FAILURE in 38m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e88be1030ddf4850b5ab9943e12fabde : FAILURE in 21m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1470da0767c842708fb87362bdce39b7 : FAILURE in 11m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ba1391d1181b40fab2f09a7d1a14480f : FAILURE in 21m 27s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f5644121cde843b5bd0f803517822451 : FAILURE in 30m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/40db47f5912f4039b34746d45f938815 : FAILURE in 38m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5e13bf3adeab4a41bec088b33e23be78 : FAILURE in 14m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a0ce469589f04725b86ed48ef000910c : FAILURE in 35m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4115093959654159ae7be72115b5b8af : FAILURE in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"17ffdd60dab520ce24975330f350c2ebd7c1ce6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-20 19:49:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b150e67fb4b0575b1ac34f27245f23803d51bf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 23:00:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/063bad2e71764aed94ba450a9f351c73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db9957b51d0b411095b4819863831c56 : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ed9b56a5f0454705adcc736986ad4e4a : SUCCESS in 1h 18m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8008888d8eae49d2badc8dbf27cc773b : SUCCESS in 1h 52m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6447900bb92a4052b8dddb988e17bcca : TIMED_OUT in 3h 02m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0ad910606cce4f7c9f50aafc946fb15a : SUCCESS in 1h 33m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/70d681f176e44d4db390ad1044c48c72 : FAILURE in 1h 25m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/132c3940df0040e8bedcdc3813ef4acf : SUCCESS in 1h 01m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e85e736e0b264711a6efead56d2f9733 : SUCCESS in 40m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e76d5f16a8084fe6989afbeee29c370a : FAILURE in 32m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2767d11f25b541958290b8d18a33c752 : SUCCESS in 1h 33m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c17ef95cba8b4060a4f7234dd6476245 : SUCCESS in 58m 25s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6153dfeade7643a9a5a62675cd9ee6cf : SUCCESS in 56m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c6cb32064fb949cfacce01612dcef24f : FAILURE in 37m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a0f01ea7456d402eb13211e08617b5f2 : FAILURE in 39m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1ecf382e357a41dba7592b8253bf9cbd : SUCCESS in 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c5af04ee05653af252e013f0a9435a92218216eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 08:26:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a9f0db1774e8a840f6bddd3d521d8bfb48f0071d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 08:26:51.000000000","message":"Topic carlos removed","accounts_in_message":[],"_revision_number":8},{"id":"c02b10a199a0189dbb9f53fc57bbb2d44f171f88","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 08:27:36.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":8},{"id":"8baf300145877171c06dd7fd75cb27f59c9e1173","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 09:45:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6f66aef72970f8d7a86ff70f825ef1038016dd5b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 09:45:16.000000000","message":"Topic carlos removed","accounts_in_message":[],"_revision_number":9},{"id":"829cd02d2d646a61678af82e69d4dc97eb1df788","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 09:45:28.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"94478b0e2545a7f3bc021c05faf0510d58b73862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 09:45:46.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 775034,18","accounts_in_message":[],"_revision_number":9},{"id":"ab6afd77dc3602ad24de7db3a9f6bc0d928d1c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:35:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"43334dfb1b4fbc18f015bafb5eb3abb754023bfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:48:03.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"e7d69b8064f453c30edc3f0c2732e23160a24f59","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:49:20.000000000","message":"Topic s-rbac-security-serivce removed","accounts_in_message":[],"_revision_number":11},{"id":"e0ce29b1db37fdbda26ab22def34ceb676ba3f41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:50:18.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":11},{"id":"f8bf0ff7f3c965358a46ab15d6931a54e0ca5b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:33:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a61af0880d24eeb898e7d3aee5dd439\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1412dea57642d987d921162f95069b : SUCCESS in 3m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b338a9a6f8084350ac46e0611bc98dfa : SUCCESS in 1h 07m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6b18eb12a59d4add9aa41cdd279b92a5 : FAILURE in 2h 39m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b06ea73b157e48c19d9b235be4abf8df : SUCCESS in 1h 15m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/56d21bafaad9482d91c0e78a1e677b9b : SUCCESS in 1h 04m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/48820fe883524dccbae224badffa8719 : FAILURE in 2h 10m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6287314625d446e998a0d3d912a3ff32 : SUCCESS in 1h 38m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9db84a3f2c40443aac62d681c76a37ef : SUCCESS in 43m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f3ecc45913524a259cedd40f3954dd84 : SUCCESS in 50m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2b1bfdc6cf8644a1a9bf4fe287215c7a : SUCCESS in 1h 05m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/478a601baa934f1c97740895b034d89f : SUCCESS in 1h 14m 59s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7f813ab35bc54fe28cc631dccc1ab743 : FAILURE in 1h 29m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d2aae564d9ac4644b8f3584bd53d0452 : SUCCESS in 35m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4447701ebf8640d6a309b769f4d74176 : SUCCESS in 40m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ae7a6c4a1c5a421684dc4c291bfb369b : SUCCESS in 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"42cea8ed337ee4086cdc7f475a4acf73a96c696c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-19 08:19:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5a6dbaa4e11fa204c9ae4f1d2eee6bfe1125cda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:48:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f6faf32f854b4eb48cd0d18edfef0d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4da372613b447c59c78a02f5011f040 : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b27fde2fa4ec40df9e8861b3aa6a455b : SUCCESS in 1h 29m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ebb7908e3fdd486298ddf12590755f19 : SUCCESS in 1h 46m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/48f74c089cc5437897528ebf75f7de5e : SUCCESS in 1h 21m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eca96b45b8c04157a1d16a841e59b75f : SUCCESS in 1h 12m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/50d86cb10c4645f0a232477a81cc83ab : FAILURE in 2h 27m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe9f58ccce584e80b4fb9b978d0e1eb1 : SUCCESS in 1h 12m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8475dac0785f4d12b55c2640337f8f19 : SUCCESS in 29m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d2f4567d306744cfb1317b6106f9d4ed : SUCCESS in 32m 15s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d8362ac715d24625a42cddbacb706894 : SUCCESS in 1h 44m 18s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/67725eec183f4b02a37702d449218f02 : SUCCESS in 1h 04m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a7f2e9af939b415a896a7edd943b939f : SUCCESS in 1h 24m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e3e00c0c6beb4382be964281e1bbc507 : SUCCESS in 20m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/099e863c61c44458aa13159f19cd0cd2 : SUCCESS in 53m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5c688b5b111d4071aae579c8abf156e1 : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6eede34c7a3cd403232981f6afb7a117b6c6fbb0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-19 12:35:58.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/867061/11/upstream-check/manila-tempest-plugin-ontap-dhss/25a96ad/ : SUCCESS in 1h 46m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/867061/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/839d855/ : FAILURE in 1h 14m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"268963d0ffe83b7e4f7c452bb307e4668c04c74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 09:37:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"04c94bb891cee4d971cef6b66cbd9f0d2c8158f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 09:38:28.000000000","message":"Topic carlos removed","accounts_in_message":[],"_revision_number":12},{"id":"00b7253cbfc4562a13e852a27a23e319f17d0b07","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 09:38:55.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":12},{"id":"290651083903343dd3058038f1bd91a306f2f781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 11:29:31.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"23945c20517c4441a83733d63d95b274fc9a56f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-15 12:33:56.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/13/check/DellEMC_Unity_Manila_dhss_true/446b9a3/DellEMC_Unity_Manila_dhss_true/1373 : FAILURE in 50m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/13/check/DellEMC_Unity_Manila_dhss_false/a7d8ed3/DellEMC_Unity_Manila_dhss_false/1474 : SUCCESS in 1h 04m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"ca33194bad0d7156feddf890336d7744ee3c9c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 13:57:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d73bc16548b4c4aa37adbc7504c921e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2aec3e409b4feda0b9a3627c32c465 : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f1b6b2230bb541c1a5cceafcc66edf1d : SUCCESS in 1h 14m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1eecb1e3b8ff43c0a59d5b8b5bd55b73 : SUCCESS in 2h 25m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f6146137190f4b3286b7fb3f310bdc5b : SUCCESS in 1h 06m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b880f095808c427eb4e25bad2120a714 : SUCCESS in 1h 23m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/69c463db4a51441da22b52ca3a78dd7e : SUCCESS in 1h 19m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/557f050b10474e25899364f10f43ad34 : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bb35dbbaa44943da8fb8c28e2b71a6a7 : SUCCESS in 38m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/786d9d9c20f24270b116d4941fcb54d3 : SUCCESS in 39m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7029def681e2488b814a1e3febd1b655 : SUCCESS in 57m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d517f689765c416893f43c2c317d2318 : SUCCESS in 46m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/59fe81b8c0b84dc3853462402deb5c87 : FAILURE in 21m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/467bbed2f0da44d4824de71d28bb53e1 : SUCCESS in 51m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9940728248c74bd495fb29c4c0d4325e : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cd5cf8a023036cfc573fbbc8479fc9a68325a709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-29 12:22:06.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"e3802467a6e6646995b543b8ebea29b14ea5dbd6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 13:25:35.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/14/check/DellEMC_Unity_Manila_dhss_true/7086df6/DellEMC_Unity_Manila_dhss_true/1442 : FAILURE in 51m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/14/check/DellEMC_Unity_Manila_dhss_false/b56c5c6/DellEMC_Unity_Manila_dhss_false/1543 : SUCCESS in 1h 03m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"d69e4414ca0446964208b0f4c52eb075614e348c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-29 14:27:12.000000000","message":"Topic rbac_security_service removed","accounts_in_message":[],"_revision_number":14},{"id":"ca8ce61c16461cae0d3272acd0cc8dcd6a4e78fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-29 14:27:24.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":14},{"id":"0696e546bc2d04447d3b1dd6af81e49bfee842c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 14:37:53.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a82f8f38714570a0fd6696af24fd68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d48b65a9115416ab5d355d125120d93 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/592f9485f6e344f6a6eca74144f5e98b : SUCCESS in 57m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ca6b916bfe264a5ba6859aa895cb38dd : SUCCESS in 2h 10m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f77686a0cdb146e8a3f0ada60c75f2ff : SUCCESS in 58m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/91abd514a8f246428c587280fac02d94 : SUCCESS in 1h 15m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/62fac0bde46b474b886486c790eb8288 : FAILURE in 1h 59m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8c5e1eaf55254ab48ea98b829393c4a5 : SUCCESS in 1h 07m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b54a2c9069a4705884b54c3768a2452 : SUCCESS in 26m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a00b92cea8314dedbb3246f46775dcd3 : SUCCESS in 34m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/68b9cdaead6a4f818afe4a9a6784eb04 : SUCCESS in 1h 20m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e83805f9ef654d24921e070960f28c00 : SUCCESS in 1h 08m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e1bf1e4ea6ee488baa74c406370a6daa : SUCCESS in 33m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bde269e480564b008f70d799707ff8d4 : SUCCESS in 33m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/51bea6a0e8604d7ab909b62b683ab85d : SUCCESS in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f757bbb116afb6d44e7a466d41eb8bb768ba8291","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 15:45:33.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/14/check/DellEMC_Unity_Manila_dhss_true/0fe7369/DellEMC_Unity_Manila_dhss_true/1443 : FAILURE in 52m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/14/check/DellEMC_Unity_Manila_dhss_false/4f05cdf/DellEMC_Unity_Manila_dhss_false/1544 : SUCCESS in 1h 07m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"f0c1b6568bda39ce99a4c66dfff1d2bd2ee53dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-18 12:31:04.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"0251f89b2693f8130dd0a0a612c016674013cd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:37:43.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2377ffacf394ab9916d071e1049e98f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a870aa74d584f7ca45dd9a3aab95048 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d822782f393545de9ddb48ffc9432da0 : SUCCESS in 1h 22m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c5ff81fdf0fe46aaab0b1ec6deec9d5f : FAILURE in 51m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0975c518230247b79f0b60636f25bff0 : FAILURE in 47m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e919616c83d64e06b7f66265c4202848 : SUCCESS in 1h 11m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c0951c116bae4d14b89bdec1037698d5 : SUCCESS in 1h 42m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7eefd5db9204424fa3a7078cc8fdf9f2 : FAILURE in 42m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8932adbdba424e52bc13e68512dedb5f : FAILURE in 28m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9789c30cb47a4475aad87b230cdb339c : SUCCESS in 41m 28s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c136eab58cc541e4b2d7de2c3d284f24 : SUCCESS in 1h 37m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/69386cfbd8d14587812d88f505309675 : SUCCESS in 1h 52m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8d1bf8cac00b499987062c3b5133534a : SUCCESS in 33m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8197d5590db4df1b2f08e053655a640 : FAILURE in 35m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c7f970b7fea94b868455a3fa1d7db266 : SUCCESS in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c4be597fc3e1cd497bf33ec40bf4f04655742b6b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 14:51:27.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/61/867061/15/check/DellEMC_PowerFlex_Manila/0be0136/DellEMC_PowerFlex_Manila/49 : FAILURE in 30m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"f4b7f23a7aa719ed87313c029bd3bcd18b83abf6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-18 22:47:58.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/867061/15/check/DellEMC_Unity_Manila_dhss_true/ca98a90/DellEMC_Unity_Manila_dhss_true/248 : FAILURE in 46m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/867061/15/check/DellEMC_Unity_Manila_dhss_false/4000175/DellEMC_Unity_Manila_dhss_false/266 : SUCCESS in 1h 07m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"8110bd4bfd6f995ffdee26018cbdd4312f0bdcad","revisions":{"1b7e8bc2f48fdf941b6676c30808567239fda2ee":{"kind":"REWORK","_number":1,"created":"2022-12-08 22:36:30.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:36:18.000000000","tz":-180},"subject":"[WIP] RBAC tests for share network security service","message":"[WIP] RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b7e8bc2f48fdf941b6676c30808567239fda2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b7e8bc2f48fdf941b6676c30808567239fda2ee"}]},"branch":"refs/heads/master"},"6e630ab44267ac99df991e5071ad1e871b790d74":{"kind":"REWORK","_number":2,"created":"2022-12-09 14:12:16.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/2"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-09 14:11:50.000000000","tz":-180},"subject":"[WIP] RBAC tests for share network security service","message":"[WIP] RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e630ab44267ac99df991e5071ad1e871b790d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e630ab44267ac99df991e5071ad1e871b790d74"}]},"branch":"refs/heads/master"},"08d2eb56b967d1f4d4807ad4708040b62351e843":{"kind":"REWORK","_number":3,"created":"2022-12-12 19:17:08.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/3"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-12 19:13:13.000000000","tz":-180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08d2eb56b967d1f4d4807ad4708040b62351e843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08d2eb56b967d1f4d4807ad4708040b62351e843"}]},"branch":"refs/heads/master"},"2187c9eee61b201ecb408f47eb73b242abf03950":{"kind":"REWORK","_number":4,"created":"2022-12-12 19:36:40.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/4"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-12 19:36:23.000000000","tz":-180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2187c9eee61b201ecb408f47eb73b242abf03950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2187c9eee61b201ecb408f47eb73b242abf03950"}]},"branch":"refs/heads/master"},"1f2c32daf56b8f9305d8f78ac50c7ce61b221978":{"kind":"REWORK","_number":5,"created":"2022-12-20 14:11:44.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/5"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-20 14:11:27.000000000","tz":-180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1f2c32daf56b8f9305d8f78ac50c7ce61b221978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1f2c32daf56b8f9305d8f78ac50c7ce61b221978"}]},"branch":"refs/heads/master"},"2ff1d98036c37d8e1d6efb9d8f580a025a18f397":{"kind":"REWORK","_number":6,"created":"2022-12-20 14:21:13.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/6"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-20 14:20:38.000000000","tz":-180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ff1d98036c37d8e1d6efb9d8f580a025a18f397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ff1d98036c37d8e1d6efb9d8f580a025a18f397"}]},"branch":"refs/heads/master"},"a4f566f8efe9c67ff1551f3dcad6d7e427995d9f":{"kind":"REWORK","_number":7,"created":"2022-12-20 19:49:47.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/867061/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/7"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-20 19:49:33.000000000","tz":-180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4f566f8efe9c67ff1551f3dcad6d7e427995d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4f566f8efe9c67ff1551f3dcad6d7e427995d9f"}]},"branch":"refs/heads/master"},"dce949ba6cc777bac657f10adf54f3e8d6f9622c":{"kind":"REWORK","_number":8,"created":"2022-12-22 08:26:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/8"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-22 08:24:44.000000000","tz":120},"subject":"[RBAC] Add tests for share network security service","message":"[RBAC] Add tests for share network security service\n\nThe tests validate policy of share network security service\nactions for admin, member and reader users in a project scope.\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dce949ba6cc777bac657f10adf54f3e8d6f9622c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dce949ba6cc777bac657f10adf54f3e8d6f9622c"}]},"branch":"refs/heads/master"},"e369f295ff2e6bd644b13f911d0bbb39677338db":{"kind":"REWORK","_number":9,"created":"2022-12-22 09:45:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/9"}}},"commit":{"parents":[{"commit":"3820a359fefe5a79b8522e6f3d2c4d614208111c","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-22 09:44:57.000000000","tz":120},"subject":"[RBAC] Add tests for share network security service","message":"[RBAC] Add tests for share network security service\n\nThe tests validate policy of share network security service\nactions for admin, member and reader users in a project scope.\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e369f295ff2e6bd644b13f911d0bbb39677338db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e369f295ff2e6bd644b13f911d0bbb39677338db"}]},"branch":"refs/heads/master"},"33b109d53d5cdaba33d86b6609c5bed02701ee6e":{"kind":"REWORK","_number":10,"created":"2023-01-18 20:35:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/10"}}},"commit":{"parents":[{"commit":"60dafff1a2c381a96e01dccbdb142eee9a9bcdd4","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60dafff1a2c381a96e01dccbdb142eee9a9bcdd4"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 20:34:58.000000000","tz":120},"subject":"[RBAC] Add tests for share network security service","message":"[RBAC] Add tests for share network security service\n\nThe tests validate policy of share network security service\nactions for admin, member and reader users in a project scope.\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33b109d53d5cdaba33d86b6609c5bed02701ee6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33b109d53d5cdaba33d86b6609c5bed02701ee6e"}]},"branch":"refs/heads/master"},"76e2b786d39b4c7da0846ac39bd176d32992cec7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-01-18 20:48:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/11"}}},"commit":{"parents":[{"commit":"4b35a13e2618c81efb104fa1dc5fd8f2dc648e1d","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b35a13e2618c81efb104fa1dc5fd8f2dc648e1d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-01-18 20:48:03.000000000","tz":0},"subject":"[RBAC] Add tests for share network security service","message":"[RBAC] Add tests for share network security service\n\nThe tests validate policy of share network security service\nactions for admin, member and reader users in a project scope.\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/76e2b786d39b4c7da0846ac39bd176d32992cec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/76e2b786d39b4c7da0846ac39bd176d32992cec7"}]},"branch":"refs/heads/master","description":"Rebase"},"7c8259780c4c4715541c955d6fd6b8c7c2bbcc09":{"kind":"REWORK","_number":12,"created":"2023-03-15 09:37:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/12"}}},"commit":{"parents":[{"commit":"191ae7f02b9e5905a1549669dbd7615e6dd29dc0","subject":"Add functional tests for add/update security service feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/191ae7f02b9e5905a1549669dbd7615e6dd29dc0"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-15 09:37:13.000000000","tz":120},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7c8259780c4c4715541c955d6fd6b8c7c2bbcc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7c8259780c4c4715541c955d6fd6b8c7c2bbcc09"}]},"branch":"refs/heads/master"},"1bde4562c7f9d55027b98489add8f10ea8dad9d6":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-03-15 11:29:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/13"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-03-15 11:29:31.000000000","tz":0},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bde4562c7f9d55027b98489add8f10ea8dad9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bde4562c7f9d55027b98489add8f10ea8dad9d6"}]},"branch":"refs/heads/master","description":"Rebase"},"1884c70817b1fca6bef1cd0c8cf2d4a8ce9e69ca":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-03-29 12:22:06.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/14"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-29 12:20:43.000000000","tz":180},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nDepends-On: Id8381f3ec11c5f745ef95c7a936863a08ffa962d\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1884c70817b1fca6bef1cd0c8cf2d4a8ce9e69ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1884c70817b1fca6bef1cd0c8cf2d4a8ce9e69ca"}]},"branch":"refs/heads/master"},"8110bd4bfd6f995ffdee26018cbdd4312f0bdcad":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-07-18 12:31:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/867061/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/867061/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/867061/15"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-08 22:33:25.000000000","tz":-180},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-07-18 12:31:04.000000000","tz":0},"subject":"RBAC tests for share network security service","message":"RBAC tests for share network security service\n\nDepends-On: Id8381f3ec11c5f745ef95c7a936863a08ffa962d\n\nChange-Id: Ia951c68f08e76dfad35c65cb49bc94e9614798f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8110bd4bfd6f995ffdee26018cbdd4312f0bdcad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8110bd4bfd6f995ffdee26018cbdd4312f0bdcad"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
