)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867119","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-05-16 13:39:13.000000000","reason":"Change was submitted"},"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-05-16 13:39:13.000000000","reason":"Change was submitted"},"35586":{"account":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"last_update":"2023-05-16 13:39:13.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-16 13:39:13.000000000","reason":"Change was submitted"},"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-05-16 13:39:13.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-05-16 12:12:45.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-12 03:39:32.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2","subject":"[RBAC] Tests for Share Instance Export Locations","status":"MERGED","created":"2022-12-09 14:56:01.000000000","updated":"2023-05-16 13:39:13.000000000","submitted":"2023-05-16 13:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":69,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867119-secure-rbac","meta_rev_id":"5dde78c8fcfd011958f1e74d95101e3b907f501e","_number":867119,"virtual_id_number":867119,"owner":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-16 13:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":2,"date":"2023-05-16 12:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-12 03:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":1,"date":"2023-05-16 12:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 17:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 23:53:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-12 23:53:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-12-13 21:32:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-12-13 21:32:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-12-13 21:32:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-05-08 16:12:24.000000000","updated_by":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"reviewer":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"state":"CC"},{"updated":"2023-05-10 01:00:10.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-05-10 13:49:15.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"}],"messages":[{"id":"ff522cfff6ba21c9f61b8f70a11dd5508e1e2bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-12-09 14:56:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5749fd961afcf9489d72a95df11023a9a6f47599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 17:10:52.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d24b1a551ff45108c9fa5874b978e9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281c30501e664259be1ef64bab9cc750 : FAILURE in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ad99b0be9d1b4833a96b8fd23eafdfbd : SUCCESS in 1h 21m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aa4011b50ff042b696d8f29a3b657fbc : SUCCESS in 2h 08m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d6c3c835229f4e139d11ffd6dfefdb57 : SUCCESS in 1h 28m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/380711af3f8d41338c2136eedb30b300 : SUCCESS in 1h 29m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb7d65c632f24605a5f5e35ec08c64a5 : FAILURE in 1h 50m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/01963e99ffc9416599d84d82c7531bdb : FAILURE in 1h 01m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c6010de8d9ed41a19d83946660079f5a : SUCCESS in 36m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b41b181c63824e9693d870b2f2e4387b : SUCCESS in 38m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3e0cb33cde994d068ceb577f4f62a184 : SUCCESS in 58m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f6d8971676e540e1ba044683291c900a : SUCCESS in 1h 13m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/dd61ae3b5cec44658ac1170feb9694bf : SUCCESS in 1h 25m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2777a8d049ca42df841df0d999538a28 : FAILURE in 30m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6b47dcbdcfd9445a8c3b3a4e34315334 : SUCCESS in 36m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/69b7ef41f7864ac2a98331cd3e16a032 : SUCCESS in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da97c5e99570458ec3ec8982d285010b3f5d1406","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-09 23:53:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/1/check/DellEMC_Unity_Manila_dhss_true/dcccdcf/DellEMC_Unity_Manila_dhss_true/1073 : FAILURE in 4h 30m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/1/check/DellEMC_Unity_Manila_dhss_false/d0ae57b/DellEMC_Unity_Manila_dhss_false/1145 : FAILURE in 1h 10m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0c42672b3a3c4e0be515f599f590fdee2350f15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-12-12 19:47:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8987d7d850a591abd80811f3c89839cc6cf64a93","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-12-12 19:50:13.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"87e363291387f6cb5fe37e4ca478147160da023f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-12-12 20:10:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"616bc9ff7fd229bd9a8ff5568da9c189fb2955a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-12-12 20:12:18.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c53ca131d86a0161629a4697992f8fcb8b48b6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:10:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13397456aa21407b96bf7e0e53cba31b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b2ed87e184496eb1266b029c4f44f2 : SUCCESS in 3m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/76390fbbc60f417bba87bb0c8c512a76 : SUCCESS in 1h 27m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8febbba73b554e95959965cad93428aa : SUCCESS in 1h 51m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c87d7c635e0e4845a7e6dadbf0dfef6c : SUCCESS in 56m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1062254cc50646ae9cee9bdec923a787 : SUCCESS in 1h 18m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c33f0d0dd0f40beb98e143256d6623d : FAILURE in 1h 51m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5492dcfc013d41ed8be74753f74e6fc8 : SUCCESS in 53m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c3b8d331124543b3a7822fc7f7549b93 : SUCCESS in 25m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d70fedaf679d4bf5a0f105322382b60c : SUCCESS in 26m 42s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f853e434b94249c0aa9421d60acdb506 : SUCCESS in 1h 10m 56s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/46e965b951ee4a8182e94339671b16a7 : SUCCESS in 1h 04m 36s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9be5fcaff5564a2098ac3329150364e4 : SUCCESS in 1h 05m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fabf862101d44532b83ba57ac756a199 : SUCCESS in 21m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0b1739f7d38f48ed8f4019de38718ad7 : SUCCESS in 43m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3ac46e0a015143008f81f5b7fa194fd3 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef539cac2fe61c167a28435e5d22fec71b395e94","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-12 23:53:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/4/upstream-check/manila-tempest-plugin-ontap-dhss/dca63b4/ : SUCCESS in 1h 42m 22s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/5c7d2c0/ : RETRY_LIMIT in 7m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"421347ec054e9552437dda1628a35ae5dbb50a25","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 21:32:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"455fde37f13a148272d591ac57b2c06b949e59ca","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-14 12:51:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9b0a170b2b51e6dfecd298e51bc2a41f6813fa6e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-18 08:38:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f50615df66fcdded3f6f63ad2e8a45096399ffdb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-04 13:28:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc91dd8d6d454e7c32e9d238b1b0c20dde5d23cf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-04 13:28:42.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"21fb66d3edfb2d7cbb6cda53c81b20a6408f7a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-08 16:03:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d45bd64bd6db419f72bb652ac5008ff3e0c9b0f2","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-08 16:12:24.000000000","message":"Patch Set 5:\n\n(13 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1a497a7af0d2297e70073a754b4679e5565b6182","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-08 16:23:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"03103581b625f54d342066e1524a01376e683c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 17:49:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba5c407bbd64c289764fa0597a3abef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b407509f89624047a7dda2ca6fd2b5d5 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3ee8425cd5f44e6a853438d8fa3daf6 : SUCCESS in 1h 04m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2dd30869e3ab43fab465eef847e07ee6 : SUCCESS in 1h 39m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6585e7dce9e942c8bd9bdc5d4ea00890 : SUCCESS in 1h 03m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/61b680c5de7f40a7b24145d9a45d2010 : FAILURE in 4m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f736c95180cc40f0833ddc07c912a56c : FAILURE in 15m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/97c5d2ea472a48ab84ab99a88e80b82f : FAILURE in 59m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/72ed700752bc4923a42501e5f3feae5e : SUCCESS in 26m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ee8231fc13a4870a3a0c79c1cf4a7ab : SUCCESS in 43m 36s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/01949f332dde4dd9ad25b44c3afbc039 : SUCCESS in 1h 30m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/230144329ded429a816b26b41e36ae7f : SUCCESS in 1h 07m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7eb2066eb23a420bb42807d04a582c25 : SUCCESS in 24m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4c357e2feff54462a94c01e18bf3635b : FAILURE in 52m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53c0a9d81dd54ab3aa915bf04e96d448 : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f8955d5453bb506f238060ec470d8e4bd16bd755","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-08 20:04:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/5/upstream-check/manila-tempest-plugin-ontap-dhss/57f8cfe/ : FAILURE in 1h 19m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/37cf64f/ : SUCCESS in 2h 13m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7613fd7f93b0f695aa863a293557e0ffbeae0635","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-08 20:16:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ceb6e752ebed968f86984f2ac47e3dae640d9299","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-08 20:17:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d106de5fe12ce16bc9c81db4427da1ae59785c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 21:44:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c94b9f144854dee9b0c64d9de0e3ef2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d317ddb82e4cfba60f6d56e09e0679 : SUCCESS in 2m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bbb33a1c6b6c4dc194214ca9bd7efe86 : FAILURE in 1h 19m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0d752ae169b94f248aa6376bbb8cb094 : SUCCESS in 1h 26m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63a86d717ded4b4788ec3a9e1ae5cb07 : SUCCESS in 50m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/146ab16231724ce1b7078509e561a2e9 : FAILURE in 7m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6a0f1092fda34f8686ec81c480493b7b : FAILURE in 7m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7e9aded5b9734d43bb92f360826c238d : FAILURE in 1h 00m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/25be138b4dba4317b05e603c17b047d4 : SUCCESS in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/03e840ffed52435e8163c32b8c9f891e : SUCCESS in 43m 40s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/53d8898d75ed431cad3795ef51d1d050 : SUCCESS in 57m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/33802f4f96bb4872af86856d73dec3a7 : SUCCESS in 1h 01m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/65734876a8e2429db043cdc6f553c273 : SUCCESS in 32m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7cac62bfde044ca08e1a28601cdb117e : FAILURE in 49m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/885b90d3b5ec4354bed216867926d299 : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"26b4a800a51c638c9e467cca6a27a698bc8ea578","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 01:18:57.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/5/check/DellEMC_Unity_Manila_dhss_true/aaf6ae5/DellEMC_Unity_Manila_dhss_true/88 : SUCCESS in 2h 51m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/5/check/DellEMC_Unity_Manila_dhss_false/113d02f/DellEMC_Unity_Manila_dhss_false/87 : SUCCESS in 1h 12m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"5816c9dd946ff0ab250d31e7e1124155a9ea629b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-09 06:47:06.000000000","message":"Patch Set 5:\n\n(12 comments)","accounts_in_message":[],"_revision_number":5},{"id":"18541435f807a8f20d5467c92b471b54e0c76df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-09 15:15:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"50bdc113b35b670fac4893506a3717de8cb83dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 17:44:20.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ef0a01a14834fa1988dba2e2441920e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f068f28d2fc4750870b28954b389563 : FAILURE in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d0ebbd16004c4cf5b936cdbb6c1a51c3 : SUCCESS in 1h 19m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/51c21c3e09654afbbfa656b883933bb8 : SUCCESS in 2h 20m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d4e0a7de47c84fa783eda045f3d5f91e : SUCCESS in 2h 18m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/431d42aaad2d44588e2d18d01832121d : SUCCESS in 1h 12m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5413a66546114667bfd7a1c0798846e9 : FAILURE in 16m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d4dcafdb9a3e4992aea0bf9c246911d5 : FAILURE in 1h 50m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c952c38730e1428ea4d56680fa0538d4 : SUCCESS in 41m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2e19d60bfe0e47359a39bb075750de9c : SUCCESS in 44m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f7fbe6da0bcc4946be7685b95ab3dbaa : SUCCESS in 1h 16m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/527d20093b18433abd825c63db3e9a2a : SUCCESS in 1h 16m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ab4e99457e0f4ef9a44623ab2141a539 : FAILURE in 34m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7718ea264698477a936012d510466536 : FAILURE in 57m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e29eb7fcd53844a98d7f16718f5eec8b : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3fb03e7a0fef84cbeb27268dfe5312b44ce3420a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 17:55:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/6/check/DellEMC_Unity_Manila_dhss_true/42a2458/DellEMC_Unity_Manila_dhss_true/94 : SUCCESS in 2h 39m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/6/check/DellEMC_Unity_Manila_dhss_false/a021dd5/DellEMC_Unity_Manila_dhss_false/94 : SUCCESS in 1h 03m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"5938bf1951ed96aa5003e5f8837b4fabbcff5c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-09 20:37:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"48c3d85fbd20aea2bf7d5be4171a5aa6a3673c75","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-09 20:40:09.000000000","message":"Patch Set 7:\n\n(13 comments)","accounts_in_message":[],"_revision_number":7},{"id":"95df6df8419bc5c6df9e4ac9f4db2361a0634cfe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 23:15:49.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/7/check/DellEMC_Unity_Manila_dhss_true/c66ae81/DellEMC_Unity_Manila_dhss_true/95 : SUCCESS in 2h 38m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/7/check/DellEMC_Unity_Manila_dhss_false/8a6ecf4/DellEMC_Unity_Manila_dhss_false/95 : SUCCESS in 1h 02m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"280041eb3eab893f9e3d1fea3385d7153d42998f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 23:47:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465b42097c2c4de2b7941731a55d4f31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1188dd4878654544809c1f86dccd12b1 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da9c89e2e523432cb2aaf39086507725 : SUCCESS in 1h 06m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/173c238765c84e289b6c2c3096829530 : TIMED_OUT in 3h 02m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9fcc4eecb1f04b66b9209b56bd9aa763 : SUCCESS in 1h 19m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ec19663ad9984e5c93b56410c12eefa8 : SUCCESS in 1h 04m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5206977937ca498781d9380a8f4e4837 : FAILURE in 14m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ca7922af14584753867e66271a91c140 : FAILURE in 55m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4239c280fd4340f58d9592c479e8ea5b : SUCCESS in 26m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e1745ae33f05416f84f4a91e60460d11 : SUCCESS in 27m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/14195c1c3d2348a682ae35d0839c223c : SUCCESS in 53m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/33090c12282f40ad9fb550241fdee37d : SUCCESS in 1h 07m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f579389c88a14780a7b198b8a7e5c1f6 : SUCCESS in 37m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c105c333d1a4c66aab3f3570e25cdc4 : SUCCESS in 40m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4a57d0a8ab2d471082acbeff9f0ccd84 : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f58a63cbd3693cf40ce3bc1a746cf2fc33c4cadc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 01:00:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"88a46661d83dd63afde8cbd3ac24876108182de8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 01:04:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bdb6e19d13b4bd0443be6bd4995c78d3855992b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 02:19:30.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/7/check/DellEMC_Unity_Manila_dhss_true/2c6291f/DellEMC_Unity_Manila_dhss_true/96 : SUCCESS in 2h 32m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/7/check/DellEMC_Unity_Manila_dhss_false/40bc07a/DellEMC_Unity_Manila_dhss_false/96 : SUCCESS in 57m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"c2aceb1d888442378d119eb6c63b2f412cbdff0c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-10 03:22:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/7/upstream-check/manila-tempest-plugin-ontap-dhss/9457e30/ : TIMED_OUT in 3h 20m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/0bc3b12/ : FAILURE in 2h 24m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6adc95e492b4c2b859a76a0cedb21e57a5c775e8","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-10 05:42:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"375fceb4c2930e3b933ead1786c98fb4f50dc2a9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-10 05:46:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07ed630234c5c84a0bfdae2497b6b453416b4fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-10 12:50:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5754d72607ebb15a37695213b790d410ccbc8a21","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-10 12:51:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f07f30745f8a95fb84ef4f6906be7d6ec94b025d","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-05-10 13:49:15.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d2c596cff28baf3520212db1c77e462301094b66","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 15:16:29.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/8/check/DellEMC_Unity_Manila_dhss_true/04e4ad6/DellEMC_Unity_Manila_dhss_true/99 : SUCCESS in 2h 25m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/8/check/DellEMC_Unity_Manila_dhss_false/8b8e2b6/DellEMC_Unity_Manila_dhss_false/99 : SUCCESS in 1h 06m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"202e27348b7de9327ddc42b7ca8f050cee629eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 16:01:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5542bed705d47f28b04237dcaffee9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cea1b3640e440499326a035de1cd427 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0be361c51cce4491a9c38894f5c79bdb : SUCCESS in 58m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1052e93a9b614d578d1234ea9806dce6 : TIMED_OUT in 3h 02m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/634058bfcde84c0d9af2817f5956c3aa : FAILURE in 2h 55m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/21c911bd9c6a420491a6642e1375e037 : SUCCESS in 1h 18m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1fb3b69c54d84074990fb24c0ff3eefe : FAILURE in 16m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7f22fcf1c2d345c0bb491022941c93e8 : SUCCESS in 1h 06m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e25f9ee87ab47fea46d4ab73945381d : SUCCESS in 31m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9420584a2377452a8993e09e8493494b : SUCCESS in 32m 05s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9d8fd26e3a07476ba4e1b86e41203b18 : SUCCESS in 1h 07m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/42fabfaaab7d4851a67ec72d08c72a16 : FAILURE in 1h 05m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/41df0d2b226c4cf7bb2d37a3d85981fa : SUCCESS in 22m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d04572dea00043b7a4f3baae9cd11b16 : FAILURE in 52m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0b737c3153d43a5926780ada5151117 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c72a83fb87853a7a61dcb1d7a7bc8b676f9f56d3","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-11 06:07:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"59026f27f86849e611369e3b7db297d881075c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:04:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b891fbef1e436a9a40108675d261b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47bb49055714422b9098868c8d7e8a2 : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b4509373f7954703acc957d335139515 : SUCCESS in 1h 22m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5fbfda25610b497da1f69ff72447abcc : SUCCESS in 1h 55m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb21bdb2b6254a7aa131f8a772f96c96 : SUCCESS in 1h 17m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4be73ca3fdfb445c972633e0a6726ac9 : SUCCESS in 1h 05m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/06ccbc66c26449ab8d914c49ae2d8ab2 : FAILURE in 17m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5d51738f82434f05845b5ba3870793d0 : FAILURE in 1h 00m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe129817219a4adfbbc3d104c4f33980 : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a44b6bdf501e4dd68400dd7a1a313411 : SUCCESS in 38m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0404097c776a4706b479c0508f5bde3f : SUCCESS in 1h 21m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d7c73e37eb7b446997457f6e2f154ebc : SUCCESS in 1h 12m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2def564582e441f283c0bbfbcfe2805e : SUCCESS in 23m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/494659034f2844a69fb19c920ae94f96 : FAILURE in 43m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/16c0e0ccbf774744a2da202a9c117a98 : SUCCESS in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"08bb8edf28d93fa846ab3acc939cd502b738fa60","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-11 08:34:51.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"66ddbd3b170d78301d6d676a68c8d3c88d745ee6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-11 11:15:58.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/8/upstream-check/manila-tempest-plugin-ontap-dhss/a2867ba/ : RETRY_LIMIT in 8m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/867119/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/54c7b87/ : TIMED_OUT in 3h 10m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"bdf839e814eb60ac1ece65d741832c83ec5b6c64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-11 11:43:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/8/check/DellEMC_Unity_Manila_dhss_true/b9e92cf/DellEMC_Unity_Manila_dhss_true/104 : SUCCESS in 3h 21m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/8/check/DellEMC_Unity_Manila_dhss_false/4609b7e/DellEMC_Unity_Manila_dhss_false/107 : SUCCESS in 1h 14m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"5cadd1c49cca0c99607efce06f30ae3de62594d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-11 13:02:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"1ba0b1daaf072dab26ffe8e6360ae9bc944203f8","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-11 13:02:39.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9fa4afe30d1b911f095b8a32abd8ad781d4982ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 16:11:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12721865ebfa4654b335fee4d8179c07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26dfd63635ca4677ac4419b8353e9f1d : SUCCESS in 3m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2c1687169c9d4481a4f86549e1236085 : SUCCESS in 55m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5802b82b83e6427a855e054a2f14d08a : TIMED_OUT in 3h 02m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c7206e148e3a4c17beefb17860328819 : SUCCESS in 1h 23m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7a4f3a9cf1174e08a05f62f556a3db22 : SUCCESS in 1h 07m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/45a5d7a9f1084f0cb1292c468d75e62c : FAILURE in 4m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/99c43cdd572045f18a37bd5e6e198f8c : SUCCESS in 1h 23m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6fbcb0eab6c54905bd50b6292dadecd1 : SUCCESS in 35m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc3d3bfef2724daab6fe83c8f73d7154 : SUCCESS in 38m 53s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b4f06443868647fdbb0b3307be36267e : SUCCESS in 1h 20m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a3112568c2224f04813d0a24529edf9b : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/abeb0a3e65874badb0309a43d8dc4718 : SUCCESS in 33m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/37afd1681b3547c9ada80708f74bde70 : SUCCESS in 54m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c95196c49ae3443c9c350c36fb3ddcd2 : SUCCESS in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d34588413ba4fc7ce5f32abe6a9263004989863d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-11 18:05:11.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/867119/9/check/DellEMC_Unity_Manila_dhss_true/7653e45/DellEMC_Unity_Manila_dhss_true/106 : SUCCESS in 3h 18m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/867119/9/check/DellEMC_Unity_Manila_dhss_false/d452f32/DellEMC_Unity_Manila_dhss_false/109 : SUCCESS in 1h 17m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"fec9148b9270d9fd156af553ec3e804cfe1c7d0b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-12 03:39:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"354c5c9f7296b680d4992caaf17d796fdb2e501d","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-16 12:12:45.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6092d94f9d37228197685427876ad8288fa7aae0","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-16 12:13:31.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"31342e87abc27879a5e48df7f89613049f976610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 12:14:02.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"726371c475c7cf21baec8492eb035f6b8b9790d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 13:39:12.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff944c6a2634b7c9c00ec37c86cab5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39ce2f23b5c4278955f2d08c0801173 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8b6af212f6414c6a95c8fe31bf007d65 : SUCCESS in 1h 20m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9d00221fad1e4d15ac0dc66e5ddd7444 : SUCCESS in 31m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ee0eaddb7b424bdd977c189f6138f691 : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":9},{"id":"5dde78c8fcfd011958f1e74d95101e3b907f501e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 13:39:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"4246d0b9fe084bace3858912f20bea65b7421979","revisions":{"43e87093c65111f2db9bbdb927917d2c97d7a7af":{"kind":"REWORK","_number":1,"created":"2022-12-09 14:56:01.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/19/867119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"subject":"(WIP) RBAC tests for Share Instance Export Locations","message":"(WIP) RBAC tests for Share Instance Export Locations\n\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43e87093c65111f2db9bbdb927917d2c97d7a7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43e87093c65111f2db9bbdb927917d2c97d7a7af"}]},"branch":"refs/heads/master"},"f1cf2021b4918b6240d198c369e752d44f3da168":{"kind":"REWORK","_number":2,"created":"2022-12-12 19:47:17.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/19/867119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-12 19:41:41.000000000","tz":0},"subject":"RBAC tests for Share Instance Export Locations","message":"RBAC tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1cf2021b4918b6240d198c369e752d44f3da168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1cf2021b4918b6240d198c369e752d44f3da168"}]},"branch":"refs/heads/master"},"42647c4a327f00533157b11d11ed545aeaf83f65":{"kind":"REWORK","_number":3,"created":"2022-12-12 20:10:53.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/19/867119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/3"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-12 20:10:24.000000000","tz":0},"subject":"[RBAC] tests for Share Instance Export Locations","message":"[RBAC] tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42647c4a327f00533157b11d11ed545aeaf83f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42647c4a327f00533157b11d11ed545aeaf83f65"}]},"branch":"refs/heads/master"},"c12ffcbfdffcffa05fedb482d3d9243afa9327cf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-12-12 20:12:18.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/19/867119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/4"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-12 20:12:02.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c12ffcbfdffcffa05fedb482d3d9243afa9327cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c12ffcbfdffcffa05fedb482d3d9243afa9327cf"}]},"branch":"refs/heads/master"},"a052374d50f33b0723f80d594d5e11d4141a58a3":{"kind":"REWORK","_number":5,"created":"2023-05-08 16:03:52.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/19/867119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/5"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Renan Vitor Da Silva","email":"renanpiranguinho@gmail.com","date":"2023-05-08 16:03:41.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a052374d50f33b0723f80d594d5e11d4141a58a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a052374d50f33b0723f80d594d5e11d4141a58a3"}]},"branch":"refs/heads/master"},"ac146353fafc7d0f0e05b3794c03d7812159329f":{"kind":"REWORK","_number":6,"created":"2023-05-09 15:15:11.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/19/867119/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/6"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Renan Vitor Da Silva","email":"renanpiranguinho@gmail.com","date":"2023-05-09 15:14:29.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac146353fafc7d0f0e05b3794c03d7812159329f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac146353fafc7d0f0e05b3794c03d7812159329f"}]},"branch":"refs/heads/master"},"81f941e3ced4ed697c32389c22ed71255b6df8e9":{"kind":"REWORK","_number":7,"created":"2023-05-09 20:37:01.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/19/867119/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/7"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Renan Vitor Da Silva","email":"renanpiranguinho@gmail.com","date":"2023-05-09 20:36:34.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81f941e3ced4ed697c32389c22ed71255b6df8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81f941e3ced4ed697c32389c22ed71255b6df8e9"}]},"branch":"refs/heads/master"},"181d8d659e446216f63594c16df54aca24e9e85a":{"kind":"REWORK","_number":8,"created":"2023-05-10 12:50:48.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/19/867119/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/8"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Renan Vitor Da Silva","email":"renanpiranguinho@gmail.com","date":"2023-05-10 12:49:56.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/181d8d659e446216f63594c16df54aca24e9e85a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/181d8d659e446216f63594c16df54aca24e9e85a"}]},"branch":"refs/heads/master"},"4246d0b9fe084bace3858912f20bea65b7421979":{"kind":"REWORK","_number":9,"created":"2023-05-11 13:02:10.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/19/867119/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/867119/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/867119/9"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-12-09 14:55:02.000000000","tz":0},"committer":{"name":"Renan Vitor Da Silva","email":"renanpiranguinho@gmail.com","date":"2023-05-11 13:01:45.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance Export Locations","message":"[RBAC] Tests for Share Instance Export Locations\n\nThe tests validate policy of share instance export locations\nactions for admin, member and reader users in a project scope.\n\nCo-Authored-By: Renan Vitor \u003crenanpiranguinho@gmail.com\u003e\nChange-Id: I5996a1fb5365e80acc3d1cb5418a62fd60eae4b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4246d0b9fe084bace3858912f20bea65b7421979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4246d0b9fe084bace3858912f20bea65b7421979"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
