)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867134","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iafa8377368e7f37feb3ff756733bd5930fd1040d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-01-24 18:19:43.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-01-24 18:19:43.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-01-24 08:14:17.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"Iafa8377368e7f37feb3ff756733bd5930fd1040d","subject":"[RBAC] Add Tests for list share services","status":"MERGED","created":"2022-12-09 18:38:27.000000000","updated":"2023-01-24 18:19:43.000000000","submitted":"2023-01-24 18:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867134-secure-rbac","meta_rev_id":"ef3b17fa607546dda36859469da6ff3ce7728563","_number":867134,"virtual_id_number":867134,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 18:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-01-24 16:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-24 15:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-01-24 16:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 21:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 22:38:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-12-10 02:38:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-13 12:47:58.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2022-12-13 12:57:12.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2022-12-13 21:20:02.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-12-18 13:02:00.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"CC"},{"updated":"2023-01-24 15:28:20.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"91a41685e38ccada8597117cdfd0d8be62b6e6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-09 18:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10511e0574b1cc100442274d6d21bd4f49ef0c84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-09 18:47:52.000000000","message":"Topic srbac-services removed","accounts_in_message":[],"_revision_number":1},{"id":"981fde618ff4fac132d8d014cdece107b4cb237a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-09 18:48:07.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"498772404fe9ef4fdc3f857127bc1ea426d17933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 21:05:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69cefe0e1dac478594358718225dee37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb1f9b8d00f4495bdf95cf62c342b13 : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd9b2c03e2de41c68670b0da801ad440 : SUCCESS in 1h 21m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9dc8e3967be74e93a00840ca03e4e1e5 : FAILURE in 2h 20m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/85943d25997147438a9488a5665d5ebe : SUCCESS in 1h 16m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1b085689ed4743b7b739a07772d5f49b : SUCCESS in 1h 21m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fe0b2e6ebff54c47a0bb371250b4a541 : FAILURE in 1h 51m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c5d2b90b45fb40e9b64c8e80fc617b59 : SUCCESS in 1h 27m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/338ed83707c8485caced046bac946ae4 : SUCCESS in 41m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6fb616926c384d85aaa099b4de41ec9e : SUCCESS in 40m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c600d14909cf47c8881ea48fb04657f3 : SUCCESS in 1h 26m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5089abd54b9a4a81aaffa7a3557d2abf : SUCCESS in 1h 14m 09s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5dfc251eb8434822a6887b9723d1b8f9 : SUCCESS in 1h 05m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1a58d799a0c243b0a682eb1f51925c10 : SUCCESS in 32m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9581a6a4ec8c4f63b88959f29c78cac8 : SUCCESS in 41m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/09d1e82d397f451aa95037ef2e947187 : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e23ca5ae7bb3333a58d31892dbad923010ec3a0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-09 22:38:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/1/upstream-check/manila-tempest-plugin-ontap-dhss/f1fe620/ : SUCCESS in 1h 31m 34s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/2d4e357/ : POST_FAILURE in 1h 27m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"17e47c79f090d721e2746e20040a0f05eb19693c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-10 02:38:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/867134/1/check/DellEMC_Unity_Manila_dhss_true/29cf9ff/DellEMC_Unity_Manila_dhss_true/1074 : FAILURE in 2h 45m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/867134/1/check/DellEMC_Unity_Manila_dhss_false/4f4aa5d/DellEMC_Unity_Manila_dhss_false/1146 : SUCCESS in 1h 09m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"de59ab3d015bb4621ad470e36a80f28d2ddc94de","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-12-13 12:47:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2eb4bd23f7156dc86d6fbeaefac025d0c3d43a21","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-13 21:20:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"72dfde2eeaa113e93f9273083a61b761e23ec352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-17 23:44:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a226baaa726ba514c08e5e519e13bd364d618191","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-17 23:45:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"93d51c53020b3ec78ee95c7113fb3f1cee58c747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-17 23:46:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"800b39443d40a5bd7a415886436ac8591d581cb3","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-17 23:47:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fac312b2e82a548c64b601f19b0a431f69ed4c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-18 02:18:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a6376cccad749ab9d661a9dfa008e19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a1adfbf91b4438a5f86d94e8118e60 : SUCCESS in 5m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5988b016333f49bc8183e871f0d86e40 : SUCCESS in 55m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/608a58484c8c4738b3e75001dd65ea9d : SUCCESS in 2h 25m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/039019a575e3400b834b8956f9461f6d : SUCCESS in 1h 18m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6c448ba51e754623bda69f60ddc257a3 : SUCCESS in 1h 28m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7577f77bac3441d782a4969307adfe2e : FAILURE in 1h 51m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/61d95330a27a4acd9c222804e292a2a5 : FAILURE in 1h 21m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22c90db64ce147728ee5da7ba68441b3 : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8f6fca40e2e74a50b8de083e955eb43f : SUCCESS in 32m 00s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2c4aadf3f1da43aebf8ad61ab67ca579 : SUCCESS in 1h 06m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/421e1700f287438080edc5f848aa8ef8 : SUCCESS in 55m 29s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c844ab252ac441ceab6c1b4ff228ef71 : SUCCESS in 59m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a3f3dc8abf004e899b8eb8d6df1758fc : SUCCESS in 21m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a504976f45a44d79996bc604300bf4f5 : SUCCESS in 37m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/84f49ecb7eaa4133a4215faa4f6702cb : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56884198690560cc29c98efa3642728c7e0dc443","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-18 03:48:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/3/upstream-check/manila-tempest-plugin-ontap-dhss/3596a09/ : SUCCESS in 1h 29m 14s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/127aff0/ : RETRY_LIMIT in 8m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9038221a77a1bca59c050b931f6b8e3f57d4216d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-18 13:02:00.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7a3c4f4b4de1f1dfe22b27859c99de979cf14ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-01-23 14:12:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebdaa29ad7d287b51ebdd263d591dde46493e4ba","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-01-23 14:14:01.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0865fbce5b614a9aeb881d71fc40af257ff73630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-01-23 15:23:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12efa2306f59d3b44233c46075262816cb36678f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:20:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b74f858ae164c0da9df9acea345d1d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c284b8eb2941d6b64119c2626b5ab4 : SUCCESS in 3m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd07f0af7f7b4dbc8ce901cd4b687fc7 : SUCCESS in 1h 11m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1b80cf1333e2420794213ecddd90ca09 : SUCCESS in 1h 54m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f2dc9a0945a43c992af76c008ba0d4a : SUCCESS in 1h 19m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9bf77164166448cda3c1248e5ddfdee3 : SUCCESS in 1h 36m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c6ade4c2e2924cc1b1ee929ac59fe916 : FAILURE in 39m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/013a4e19fd744b0c8b8a071e2776de04 : SUCCESS in 1h 08m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cfcf7098adae410ca157573a68ab1847 : SUCCESS in 36m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d748f218c2c24bfdb1e304fd70fa6785 : FAILURE in 41m 56s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a17954dea7554c41a588d35da76c6094 : SUCCESS in 1h 17m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/25673ea6ece44fb7b86836e7304b243f : SUCCESS in 1h 00m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b21636be1796454c8362ea8c67e36f69 : SUCCESS in 1h 11m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/87b98f6f25744aeea3f28598b4dd2548 : SUCCESS in 28m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2009e37f653c40e2b05c71ce372bf22d : SUCCESS in 1h 06m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b00ee76637ef4b83955d805b59e3895e : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"01154539255503e9728b86afbde8b0a4c46f844a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-24 08:14:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e68922d67f14269caf3547de601b8af76011b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:27:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173364a9ad094a469f319892eb8d4c76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc44ccf23864f27a8204375af58c6c7 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a5c98bc6e7384b22bdb3f7ab13ae03f9 : SUCCESS in 1h 19m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0fdb259d8080402d99861164bf6e550f : SUCCESS in 2h 04m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/53c7b0f9bb2a48329671c2eb4da92266 : SUCCESS in 1h 18m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/14011d60a2b74c0c8b4b33ebb34e54c5 : FAILURE in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2426ecd7f9d44cca9d71f7b12ae8cbec : FAILURE in 2h 11m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5b63b7bc915146acad7e8f8840fd8c58 : FAILURE in 1h 09m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c87e96bad8bd4bc3a90706e73aa1e1ac : SUCCESS in 22m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e52a032603e84966ab69c3cc46493531 : SUCCESS in 37m 59s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0d470f30837b4a908527d5349bbbc88a : SUCCESS in 1h 30m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0d91a7e7fae8431f994106b74ddb4ca6 : SUCCESS in 1h 09m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/179e0f46898b4ebb9fe2952d7209ace0 : SUCCESS in 57m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cb435615c1d24c10a42cd2dc65ad78f3 : SUCCESS in 29m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6a22c0df0e33412ba1e5186d8d641e0b : SUCCESS in 58m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/75c50c548778413e82ed9d3087c797d3 : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"309ab71979d5cfd60e7ebaf825e0b4521817c4a0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-24 10:31:22.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4cf60589963760d4527fb0f763dc5d78f02dccb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-01-24 11:00:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dea4573a504b5fbd41d4bc315bd1d96bb128e819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 13:13:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b9c9c6f2d9147f2b28d19db46cdb9d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871c5da3b8444dc4b88a2226d97bc093 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c5d037993e5e45f1ab05afd00e5aa5b0 : SUCCESS in 1h 14m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/139a29b20bb3454983b2e0c3ccc3755c : SUCCESS in 1h 56m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4b3a0d6cb844a64917e7a7f041b67aa : SUCCESS in 1h 31m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6d04090865f344a6ad63466edc0c6ab7 : SUCCESS in 1h 48m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59263e2a313a42d9b0d5fa709df9829b : FAILURE in 1h 50m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b159115de6b44b6782753b4c95795073 : SUCCESS in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/01c4c2dee8304bbe92b25838dc8423b9 : SUCCESS in 32m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b29f030cd0fc48309faefe6acf929d67 : SUCCESS in 40m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/568c6faae05d4a45889975c75765111c : SUCCESS in 1h 10m 07s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a717da2a76ce4fe78ccdf6df567e06ae : POST_FAILURE in 57m 27s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8608815b382a49b19c4ea3fe20a1c2c8 : SUCCESS in 1h 02m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4860661c17da42e88e67d1769a02baad : SUCCESS in 23m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/03fd2756cb1c4820a3314a87813647a4 : SUCCESS in 44m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3d4b2dd51cda49998814bb01975ab1a4 : SUCCESS in 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b42d96099f2fed0af558aaa04ced6d147116de64","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-24 13:23:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f1abfa0e8c899569c73f8d9f52bbb9a0596457ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 15:24:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f62a3a3d54b4e318829b30f6bc5109a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4a9cb15a904225a7c5f81726348193 : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/add4c0ee7ebe45d0b196d8afed27cc61 : SUCCESS in 1h 32m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/79ddc19adfd74338956d7595b06cf2f7 : SUCCESS in 1h 57m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bd79df8ae6e6429692fd800d46ed91de : SUCCESS in 1h 08m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c2f2b2c422eb4f3aa9090e82ab43c087 : SUCCESS in 1h 19m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba0648588cd14d039b3ac1daad02837a : FAILURE in 1h 30m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/20c2c3f974f747938e34a74ec1b824c6 : SUCCESS in 1h 01m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9f6cd67d5f4f411482630916531efef9 : SUCCESS in 39m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/20dfb9a267f749b793c4592036864512 : SUCCESS in 43m 09s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c2e36419b54b415388bcaf2f2dedab0c : SUCCESS in 1h 05m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b7a12ef8d7614f7580f3054b2dcfbf20 : SUCCESS in 55m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2f9112a3153740399ef5a2f90175308c : SUCCESS in 1h 08m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d07cadb9839d497bae1344faf21fd7cc : SUCCESS in 22m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9209fe3c3fad4c6a892eaf9577c956f5 : SUCCESS in 46m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7cd80b08623c4cf2abeddb920fd97da2 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca127f14c2d644fa99ae48d9451286be606a8576","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-24 15:28:20.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5b32b8f623174404903a57187cc4e54429ea1f3e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-24 16:50:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3c658cfcfdbd32d6a63f127b64a6266aa08a73b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:51:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b618a0bc1f675940af81728d3237f0df8da958b8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-24 17:22:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/6/upstream-check/manila-tempest-plugin-ontap-dhss/e362746/ : SUCCESS in 1h 56m 54s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/867134/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/bcb55a8/ : SUCCESS in 1h 29m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"31f45400d68be65782fd9a582caa96c39cba3ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:19:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97cd33aebbff4c8cab9725e4b89175cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfbfab7fa10491c92da7c0f121006e9 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1416d67abc804870885d4e479e425b97 : SUCCESS in 1h 10m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ba712f398414834b295520fc3f1fbcc : SUCCESS in 25m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7bee4fbb6e6f420da25e1fbecd795b41 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":6},{"id":"ef3b17fa607546dda36859469da6ff3ce7728563","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:19:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"85bc9c4c1664022f65e37d6c7279b990e7c55267","revisions":{"c9580ff1cdfb633f2211ab819a3135056f736aef":{"kind":"REWORK","_number":1,"created":"2022-12-09 18:38:27.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c9580ff1cdfb633f2211ab819a3135056f736aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c9580ff1cdfb633f2211ab819a3135056f736aef"}]},"branch":"refs/heads/master"},"457c2bc782752ef72446a29e3fcc01d2811f665e":{"kind":"REWORK","_number":2,"created":"2022-12-17 23:44:10.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-17 23:43:58.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/457c2bc782752ef72446a29e3fcc01d2811f665e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/457c2bc782752ef72446a29e3fcc01d2811f665e"}]},"branch":"refs/heads/master"},"3b34cb3830a24f89a9e09dae1daf23d4a64a8f4a":{"kind":"REWORK","_number":3,"created":"2022-12-17 23:46:36.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/3"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-17 23:46:30.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b34cb3830a24f89a9e09dae1daf23d4a64a8f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b34cb3830a24f89a9e09dae1daf23d4a64a8f4a"}]},"branch":"refs/heads/master"},"9df08791da5a301c5192584aef54cfcca73aa9d0":{"kind":"REWORK","_number":4,"created":"2023-01-23 14:12:45.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/4"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-01-23 14:12:44.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9df08791da5a301c5192584aef54cfcca73aa9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9df08791da5a301c5192584aef54cfcca73aa9d0"}]},"branch":"refs/heads/master"},"846d3e244c179d450ab0a6fab53b716f1537833a":{"kind":"REWORK","_number":5,"created":"2023-01-23 15:23:26.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/5"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-01-23 14:42:43.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/846d3e244c179d450ab0a6fab53b716f1537833a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/846d3e244c179d450ab0a6fab53b716f1537833a"}]},"branch":"refs/heads/master"},"85bc9c4c1664022f65e37d6c7279b990e7c55267":{"kind":"REWORK","_number":6,"created":"2023-01-24 11:00:36.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/34/867134/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/867134/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/867134/6"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2022-12-09 18:29:40.000000000","tz":-300},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-01-24 10:59:51.000000000","tz":-300},"subject":"[RBAC] Add Tests for list share services","message":"[RBAC] Add Tests for list share services\n\nThe tests validate policy of share services list action\nfor admin, member and reader users in a project scope.\n\nChange-Id: Iafa8377368e7f37feb3ff756733bd5930fd1040d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85bc9c4c1664022f65e37d6c7279b990e7c55267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85bc9c4c1664022f65e37d6c7279b990e7c55267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
