)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867185","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I31322352dcd7982443fa1f590a643b341c059ff0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"refactor","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-01-12 08:05:46.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I31322352dcd7982443fa1f590a643b341c059ff0","subject":"Move general methods under utils file","status":"MERGED","created":"2022-12-11 10:00:16.000000000","updated":"2023-01-12 12:37:55.000000000","submitted":"2023-01-12 11:13:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867185-refactor","meta_rev_id":"fab58fdd22ddffae9e14860087d2fa3c79514f20","_number":867185,"virtual_id_number":867185,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 11:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-01-03 18:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-01-10 18:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-01-10 21:04:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-11 12:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-11 16:37:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-12-11 17:24:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-03 18:20:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-01-10 18:27:26.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"4eb2da90188321dad292b26efb16387024aae96f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-11 10:00:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2d92c428933a6462950a5de3dbba7ac7f4a6450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-11 10:04:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"457207aa0b4eba7927de6327a11199f522593010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 12:07:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1631f2c223314beda19b56c54a1e628d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73aa31664b7d4f18bac11c548378b2c6 : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99283b40b7f445468267b36e9bd40009 : FAILURE in 51m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/884d56ae178f4a7bb19d4b3f96e1061e : SUCCESS in 1h 22m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e6c0f55077141f3b45ad6cd5f88faa4 : SUCCESS in 54m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7a252308b4be47b2ba672c6373e6be16 : SUCCESS in 1h 00m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e049efc1843545078cf1e3fe425a1b1b : FAILURE in 1h 58m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2eef06228976466eb3f63460ce9456cf : SUCCESS in 1h 03m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/866c764b6fd54c02bb912a06c230dca8 : SUCCESS in 23m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d189dee3962c44448bef9260eaae730d : SUCCESS in 29m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b1cfabc2c26445b08722058eabd1dd2a : SUCCESS in 1h 35m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/60690b31beb641dda9735ea6df372e07 : FAILURE in 49m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/06550e2aec794117ac79d884275fe3c6 : SUCCESS in 59m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e25df1e8c27481f8f138ba7feee0458 : SUCCESS in 19m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/763629b01c164c32b0dbeeb13c5d8e7e : SUCCESS in 41m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/34228072571b4c40ba5110c3a4af9e5b : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b5889afefd2f0a0492beeb4816df6f795952645","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-11 12:54:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9352e759c963e01f44afaa50f94f21d545f1f214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 15:04:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad24b5e80cfa4cd3b155bca6d5eedbb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1080c80cb5be46e4b60cee200e81ded0 : SUCCESS in 3m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1ab2a554f0d7475cbe9b3dbf7b67a325 : SUCCESS in 1h 16m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6fc4131dedeb46ad99a069b0443be8a5 : SUCCESS in 1h 59m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77c92f8b25fc4eb19b5e1cedd1e13e3e : SUCCESS in 1h 29m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bc90c226dc3a467da93a4cf7fdf8812e : SUCCESS in 1h 11m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2fdd16e0d3b4c09830c6ff5a4e7d036 : FAILURE in 2h 09m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1f2c4fadb03e41f488c1bdde3dbf7fb3 : SUCCESS in 44m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d3911118df0c47ba8734ef9550272f03 : SUCCESS in 24m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4991da6b2898404796257a576a93b4f5 : SUCCESS in 41m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fc95649bed554abe9265376d1e640240 : SUCCESS in 1h 11m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/30da5ffe1d824c55bc8772ca80a99433 : SUCCESS in 42m 49s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3ca4e4b524114c60b63876df8fcc0045 : SUCCESS in 1h 21m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/148b0469bc444b1cb22b9be99296c865 : SUCCESS in 18m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2abe80b131e043fc811fc83710145d71 : SUCCESS in 40m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25e75ef058ad4cb091c60eba47a2bae4 : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f9be99066b7d21f425ce854a78ac30b024aa416","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-11 16:37:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/867185/2/upstream-check/manila-tempest-plugin-ontap-dhss/7f655ec/ : SUCCESS in 1h 31m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/867185/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/4c4a2a8/ : RETRY_LIMIT in 8m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"01d430d62b9a683790a69ce38b2110255bc70ba4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-11 17:24:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/867185/2/check/DellEMC_Unity_Manila_dhss_true/efe87de/DellEMC_Unity_Manila_dhss_true/None : NOT_REGISTERED\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/867185/2/check/DellEMC_Unity_Manila_dhss_false/e7855bf/DellEMC_Unity_Manila_dhss_false/1150 : SUCCESS in 1h 14m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1665e48d1944dcefeafdbca748f3f37c5fab36d9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-03 18:20:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b811adf445018bef0a60653cf22aafd4e60df664","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-01-10 18:27:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0bfc93130848a71248b710e7b730cf4e1e13830","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-10 21:04:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f02f7ba4539f32868b2821f60ddee4b41039acd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:05:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1bf0ada24f7e15381e6f9b4fb99c1c30c6fc612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 22:09:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/059be128dd6c420e9fd902e6d74ab3ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f03149a47af44befa94084a090481b06 : FAILURE in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/232dce260d384dfeb92e1f2c7b2afd71 : POST_FAILURE in 1h 03m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8967fbf8fe2349c3aa732af6004e69da : SUCCESS in 25m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/473fe8eaa1c9404ba5ebb4bebb6c4446 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":2},{"id":"07275400848a3b58eef5cfd49506f240f7942564","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-12 08:05:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4da4d4e34c03cd9d2dd5d9034694c530e50686ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:21:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17f66b0a8a940a7a91c0385557aaaac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e848ea0d1e144c30ac3f36a3da44bc11 : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b943f643be24aa98f1223c9e1878e8d : SUCCESS in 1h 08m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9007b6ecfbc40258dab60c177eebc78 : SUCCESS in 2h 10m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8a57df5df75d4c3dba645cb442bec226 : SUCCESS in 1h 20m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0c129f4cc74e4e83b0db8a077df1f65f : SUCCESS in 1h 30m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0591b9bbe7c0494c8d09396b7201cc82 : FAILURE in 1h 31m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/572d9cf2a64e47368fc4a92f77d04c4f : SUCCESS in 1h 05m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6ee0efad29064b0b815ca644fe93cb83 : SUCCESS in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/06a9891f2228432f986818c1a0c2813d : SUCCESS in 46m 28s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/33c6f9c1220c43fc9047281e039da2ce : SUCCESS in 1h 21m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ea479b04f6564bfbafd2d712c289faca : SUCCESS in 1h 22m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9234e490386c436db8fd1b0a6455d0c3 : SUCCESS in 1h 07m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4b188df24b934cfdb9145fe5cfb9f06d : SUCCESS in 23m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e4d048089b8d456294601908d6d1c518 : SUCCESS in 1h 00m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/46e7989706c14eac9e254b312bf8beac : SUCCESS in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec51a5c0b55d10b0821706ad8f6ec5701f7a4912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:22:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7c60e1ecede33b4a2c9da218efb2d14b08e8d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:13:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a2000863b34f17a45583134fd07c5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d946bc1de0c84359a63fde9a4884d3e1 : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2bed8a892cd44a369688333ba359ea08 : SUCCESS in 49m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d905b3da194845a69063042efc663931 : SUCCESS in 25m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a765f2e5f2bc42ad80a343b1ba8f7120 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"02b43a1c3ec02fe4c035f6f28b0bdcd38915d5e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:13:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bdeb1d80c5f38763a7e6ad793ae8537913ca9767","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-12 11:28:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/867185/2/check/DellEMC_Unity_Manila_dhss_true/f7ed4d3/DellEMC_Unity_Manila_dhss_true/1134 : FAILURE in 1h 03m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/867185/2/check/DellEMC_Unity_Manila_dhss_false/ac1513d/DellEMC_Unity_Manila_dhss_false/1204 : FAILURE in 1h 06m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fab58fdd22ddffae9e14860087d2fa3c79514f20","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-12 12:37:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/867185/2/upstream-check/manila-tempest-plugin-ontap-dhss/6e19c45/ : FAILURE in 2h 15m 16s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/867185/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/6b37b00/ : FAILURE in 55m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b63ec7845b3a416db7d43b3af9e0fb6779dd834","revisions":{"ab77b42df8dea21ca7c4bd5ef8d4b1052e0219bd":{"kind":"REWORK","_number":1,"created":"2022-12-11 10:00:16.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/867185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/867185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-11 09:57:59.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-11 09:57:59.000000000","tz":120},"subject":"Move general methods under utils file","message":"Move general methods under utils file\n\nAs a follow-up to the patch [1].\nBetter to move these methods under utils file\nso other classes can benefit from it.\n\n[1] Ia0cf15210e697988cd54ae82db8bb76aa17a5996\n\nChange-Id: I31322352dcd7982443fa1f590a643b341c059ff0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab77b42df8dea21ca7c4bd5ef8d4b1052e0219bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab77b42df8dea21ca7c4bd5ef8d4b1052e0219bd"}]},"branch":"refs/heads/master"},"7b63ec7845b3a416db7d43b3af9e0fb6779dd834":{"kind":"REWORK","_number":2,"created":"2022-12-11 10:04:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/867185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/867185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/867185/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-11 09:57:59.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-11 10:03:59.000000000","tz":120},"subject":"Move general methods under utils file","message":"Move general methods under utils file\n\nAs a follow-up to the patch [1].\nBetter to move these methods under utils file\nso other classes can benefit from it.\n\n[1] Ia0cf15210e697988cd54ae82db8bb76aa17a5996\n\nChange-Id: I31322352dcd7982443fa1f590a643b341c059ff0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b63ec7845b3a416db7d43b3af9e0fb6779dd834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b63ec7845b3a416db7d43b3af9e0fb6779dd834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
