)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867213","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iee0f739d9701b2efe9b275ea22859318b58a46be","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-09-12 19:40:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee0f739d9701b2efe9b275ea22859318b58a46be","subject":"[RBAC] Add share groups tests","status":"MERGED","created":"2022-12-12 12:12:58.000000000","updated":"2023-09-12 19:40:25.000000000","submitted":"2023-09-12 19:40:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867213-secure-rbac","meta_rev_id":"fb1fd6b858283dc9a6189f2920b561f8048a9a37","_number":867213,"virtual_id_number":867213,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 19:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-09-12 18:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-09-06 07:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2023-09-06 07:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-09-12 18:54:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 12:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 19:09:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-30 00:22:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-18 13:19:07.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 19:50:29.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-07-28 01:44:17.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-12 07:08:18.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-12 18:54:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"480aac584ee0458d4cff2b3c3600d5e1c678d485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-12 12:12:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fd0a918c17dfa575e76054ab4f6dd7a758effff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 12:13:21.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 865535,5","accounts_in_message":[],"_revision_number":1},{"id":"1b827b81745d1c50d8ce5370df3c08153f2048c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-12 19:09:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/1/check/DellEMC_Unity_Manila_dhss_true/6022a5f/DellEMC_Unity_Manila_dhss_true/1079 : FAILURE in 2h 52m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/1/check/DellEMC_Unity_Manila_dhss_false/8864454/DellEMC_Unity_Manila_dhss_false/1152 : SUCCESS in 1h 01m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"47e0431f9aea70f9cdc62a0453006b01fc995a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 09:18:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6d833a0e8ad4e0b89875f88cbb74eda4da8f6830","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-15 11:32:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/2/check/DellEMC_Unity_Manila_dhss_true/9162d45/DellEMC_Unity_Manila_dhss_true/1146 : FAILURE in 56m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/2/check/DellEMC_Unity_Manila_dhss_false/f3a6a88/DellEMC_Unity_Manila_dhss_false/1216 : SUCCESS in 1h 03m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"77100f1ab56f0ee5d678b4f8ff8ec282bcb278fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 11:41:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/379aecbb417b4d0a8304ae7e8859830e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776b7cf6efc743498e1c6d455bbbb22d : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2e2aac7f7d984b3781bd14caee48861b : SUCCESS in 1h 15m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/17d070cc856e4dec8123bc03fb0085e7 : SUCCESS in 2h 16m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b64c7e96ccce41b9a9a692d4c967ac92 : SUCCESS in 1h 43m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ec30cb8c86e3486fb4af73296dcbdc90 : SUCCESS in 1h 32m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/393868623a3e4d4e911d19f85217a0ac : FAILURE in 2h 17m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6cbbad6771e84aa1a7224b899483e75f : FAILURE in 1h 07m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a93e320fd38642da8283657102b25d92 : SUCCESS in 37m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/251f9e8562294a6fb295a92ccd52b67b : SUCCESS in 36m 41s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6927eb38d30a4e6ea12c740de7fabec3 : SUCCESS in 1h 29m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fb4b855a14074dc393d7889c6ba9e709 : SUCCESS in 1h 11m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fafb7f8aea374f0fa8bfa1f4a2e5418b : SUCCESS in 44m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a2ecedb7e3964f49b8b97640b3709357 : FAILURE in 35m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21af3060c571457ebfeb6aa9a09d20af : SUCCESS in 57m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e670709c31114b7cac8d21695635cc32 : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"672c35fcd2a442f7ed462da7ccf5f89f47d3d8f2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 12:35:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69af35967c84f51ba7051decf5a6fd510f68c55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 12:35:48.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 865535,7","accounts_in_message":[],"_revision_number":2},{"id":"2e3be79ca52e90f0f459980d8d7b5eb892d2e146","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-15 21:01:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88330bfb4e608036f50dd836edc74f69e3519d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 21:01:45.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 867213,2","accounts_in_message":[],"_revision_number":2},{"id":"6802241236dfa605903f89c23bea164340ab49d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-16 06:08:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1b1a9414d59bfab1136865ebada032f6157191a0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-16 07:19:56.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/3/check/DellEMC_Unity_Manila_dhss_true/a18eb79/DellEMC_Unity_Manila_dhss_true/1149 : FAILURE in 1h 05m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/3/check/DellEMC_Unity_Manila_dhss_false/4c5cac6/DellEMC_Unity_Manila_dhss_false/1219 : SUCCESS in 1h 11m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bd49a1a907020cd2e12c7f85c8417676fccf8135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 08:27:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b15fd004655842f1ac47df648293b42f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce8a374d3145408d9bffa7a2c69b774b : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0d2bd5a4dbc548e882d31425af91874f : SUCCESS in 1h 10m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a906c2e922894e46b824375db98caea0 : SUCCESS in 2h 13m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/efc96ba4496148edaa1eacdda236a6aa : SUCCESS in 1h 21m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a2a575e2401c4eedbbc829332d557b19 : SUCCESS in 1h 25m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1f9735f1bf684559aedff8a7348946ac : FAILURE in 1h 59m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/30dfef8fa9c64c858294e94e0add64a5 : SUCCESS in 1h 13m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5b0cfc88e00f4a34aaa016d5e684dc47 : SUCCESS in 31m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0654af2f274743a88c3e646cc1dfb6f9 : SUCCESS in 41m 37s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b839056c69694510b857d67cdb4caddc : SUCCESS in 1h 22m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ca415a4acd28494085109b7498d7a958 : SUCCESS in 1h 06m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7f8372e30017426c9630597cab7593fa : SUCCESS in 1h 18m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c97e657066ee4d3da60ef8f26f9aafa8 : FAILURE in 29m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6902d52384be4a36821cdbf3c1f851c1 : SUCCESS in 51m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab2debe1e2274a2492baf6d6e02dee14 : SUCCESS in 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72a4287997c0beaf686c935b77ec0b18bbddc9a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 11:11:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a6fa11c63f5c746b9c3cd2333584cb96892286ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:25:55.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"95ee99c3f20a5f67fe476d90e99d8021d42e4881","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:51:09.000000000","message":"Topic s-share-group removed","accounts_in_message":[],"_revision_number":5},{"id":"cdcb0c8ad72fcc86f9c2dc0f689e267110c337df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-29 13:51:14.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":5},{"id":"b783b890a97ad424ac9a44ad5b3a4b9de6da9e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 22:40:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93f3d7e83ce647cc813b97a03ffeb493\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a3f28ce6724fd9afca2576e470bb11 : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/952cf64ed3ec4b8f8a4c799894bcbd4b : SUCCESS in 1h 26m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7b6c0f38e7eb4be8bed44a0f6e8edef3 : FAILURE in 1h 26m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c2ddcca8ab341ad863cd3999eb7ed38 : SUCCESS in 1h 11m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f0fb3deb55994abaa3d573788ae3005a : SUCCESS in 1h 03m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8d36a9347c4b471ab05370d595b0c872 : RETRY_LIMIT in 3h 01m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/85ce5d2d7c434241ba74133b81c9922a : FAILURE in 56m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4faea4eaacae4b858b32f6c044ea40e5 : SUCCESS in 33m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/951b5345c6a34a31a3dace5d7e11cea4 : SUCCESS in 28m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/cecad424931d4cb9ae695ef39c38edec : SUCCESS in 1h 13m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fae2a7870e724fa998055d2d56f7076c : SUCCESS in 55m 04s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/513337e9bdce44a28c8c72d3a31ec674 : SUCCESS in 1h 03m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ec05439141b249d6819f569ed94dd393 : SUCCESS in 22m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f46f70ed060b409492d5010ff6b288dc : SUCCESS in 52m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ac395b0acf574358924f383e54f56f8b : SUCCESS in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d4ac03ff3129268f21440c444055ab637d038713","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-30 00:22:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/5/upstream-check/manila-tempest-plugin-ontap-dhss/4500cef/ : FAILURE in 1h 40m 14s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/ef381a0/ : FAILURE in 1h 26m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"cd16ddc0ffcff8710b7e8a4f48eafd26976c0dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 07:01:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"14d8c9144499163b1a2a8680c88bf64c54792427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 09:16:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5efadf8168d94440b9d42b684eff054b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eff8e113c91346e386b6840cde6fc2d9 : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81556381639f4a249ee5810e58f1fc71 : SUCCESS in 1h 45m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/353285ce567546d280a67d1d5dc1008f : SUCCESS in 1h 52m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1c715ed5b9bd4d4283dba0fdd8453d0c : SUCCESS in 1h 46m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7b66881c78854ec3a21c8c528e8747d0 : SUCCESS in 1h 10m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f4b060aad59949f395d810f0f01c0e5c : FAILURE in 2h 13m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dca98f1b9af04d928d76b46a4f8a91a6 : SUCCESS in 1h 16m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/651f97f967244a8fb4a9ad4c05cf1fa0 : SUCCESS in 23m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/34e69e4650264a2d9c197f248856fced : SUCCESS in 43m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3de86d97f9eb44e7a732ec788f6826b8 : SUCCESS in 1h 35m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/af27801efa9e4be0ad4b308cc31ee06f : SUCCESS in 57m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/570828b18d3d48ed96c266e631e2a945 : SUCCESS in 41m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7f6a217b4ddb45a4a9b3485734c500ba : SUCCESS in 56m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a86ed6ff1ffb497ebebc1d6e482392e2 : SUCCESS in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af70e827576599166f950b504bf1ac83bf675cc2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-30 12:21:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/6/check/DellEMC_Unity_Manila_dhss_true/229c396/DellEMC_Unity_Manila_dhss_true/1450 : FAILURE in 1h 06m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/6/check/DellEMC_Unity_Manila_dhss_false/32dcaa8/DellEMC_Unity_Manila_dhss_false/1551 : SUCCESS in 1h 15m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"0fb7e2278b4e13c154fa6a8182faed7082c1a84d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-18 12:19:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"25a271200b478e2ba7354036983640b96fe21c09","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 13:19:07.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/13/867213/7/check/DellEMC_PowerFlex_Manila/ce511a4/DellEMC_PowerFlex_Manila/46 : FAILURE in 30m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"d3e54cf2569165cc82a5ca2b46f7a5918c129f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:09:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7337f265b3734b8e891244ef33e64866\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e28d7c91b864c98902ceaff0bbe6874 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9842f5faf8d4d1e85aefa631e688d1b : SUCCESS in 1h 34m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/998eb3dce3e04f55a2bdd3fd5e9ecc7f : FAILURE in 1h 06m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62663a30b4f140e6b10df1bd9bac42c6 : FAILURE in 34m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cbc9c8e06d7b408eafdce39ebe472f32 : SUCCESS in 59m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e5780c2aa4264bfb802907479e9eb9a3 : SUCCESS in 1h 24m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4e68a8a19bfb44b89ded03bdd8946feb : FAILURE in 21m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/81abef88e9a949bcb1963ab4c6bc14c8 : SUCCESS in 40m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21cbc95ec4604741b672e9e02ff655ec : SUCCESS in 39m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f9f19725b76d4875ab079444bba01c10 : SUCCESS in 1h 04m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6e9f17c9f2c84d7faa0ae74c4e96f768 : SUCCESS in 1h 09m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7793ab752f624a98ad6547878cb79641 : SUCCESS in 29m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21887a21a3264238882007613d2b0d35 : FAILURE in 22m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8d4a3576252e4dfd830f13c9357b7401 : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1e3c5b8d73b92e62e6adc80c01303e75cab0d8ca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-18 15:49:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/7/upstream-check/manila-tempest-plugin-ontap-dhss/688203c/ : RETRY_LIMIT in 7m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/dd6b909/ : FAILURE in 1h 39m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"0c5fa1b2a24ad2e0d0029b51874a114095938470","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 16:23:07.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/13/867213/7/check/DellEMC_PowerFlex_Manila/27c769f/DellEMC_PowerFlex_Manila/52 : FAILURE in 30m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"469f23f58f0003a36967cf2ad0a421e9b511d513","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-18 17:29:10.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/7/check/DellEMC_Unity_Manila_dhss_true/ea28258/DellEMC_Unity_Manila_dhss_true/245 : SUCCESS in 2h 33m 05s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/7/check/DellEMC_Unity_Manila_dhss_false/3451acd/DellEMC_Unity_Manila_dhss_false/263 : SUCCESS in 58m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"5135b5730bd4c553bbe8f6807d1d47fe1e0710f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-19 07:45:04.000000000","message":"Topic share-groups removed","accounts_in_message":[],"_revision_number":7},{"id":"c6ea980280e2d55ec7bbbe499f90d2baea18fd2f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-07-19 07:45:12.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":7},{"id":"c4ae1962f7043f4629bb29839e3466319a28549e","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-07-24 19:50:29.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7936f7c218bf1811445c3d5b583ff7f9bfb55463","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:44:17.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"bf6deeaa09e8e7178b40860d697e31b299ad7643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-06 07:01:22.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b1a446a2ed9ee600cf3eff091b3c3ae415612b18","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-06 07:37:45.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_PowerFlex_Manila_dhss_false/3ed362b/DellEMC_PowerFlex_Manila_dhss_false/319 : SUCCESS in 36m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"bc1db3f8117cb70e88db2b33357d6acf6790145e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-06 07:46:15.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_Unity_Manila_dhss_true/bb21014/DellEMC_Unity_Manila_dhss_true/539 : FAILURE in 28m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_Unity_Manila_dhss_false/e6b1427/DellEMC_Unity_Manila_dhss_false/567 : FAILURE in 39m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"1217132c2fb3c64f5a07a5d72d472bd45ca3b8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:30:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4361d3cf5b99474682bf43540f405cfe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e56bd06d1ed45f29fed0bd194590680 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4961197e09641dab4695fa399e63157 : FAILURE in 15m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8afc2c0d0e3442dd88fe7625d2a2c569 : FAILURE in 33m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/986818c9b6d4483fa130049d56ac33cc : FAILURE in 44m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/054ac378edc846eab2c56f071e38b19a : SUCCESS in 58m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c740c1b1ef1e4584b8f2942dd8ae1ed2 : SUCCESS in 1h 24m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cbcc7a8ba0fa4378bee31ee277643dce : FAILURE in 59m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f2dc7aed114e4c03a48eaaaaaa48a49f : SUCCESS in 34m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dc462c49f0184ee1a7061a01d87b0de6 : SUCCESS in 34m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6436c443d58742e28ba53443b72d7be6 : SUCCESS in 1h 05m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f15ffdd8406547238366344196e190d9 : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a277be0a152747749097e85a11330d4d : SUCCESS in 24m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d92901f3f353415ca98dd1c5b16e3cd7 : SUCCESS in 43m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3c479f68a8b04a38940a50568bd4e41e : FAILURE in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"59b0c2d381bac5cb505af97073951176d98f5524","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-12 05:20:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5e4e57a1720c0f8b5b19db6efa9a828f30264171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 06:31:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f620b0af2072489e8e968ebec101ed48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1e042026224631aef1eeedaa61bf51 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3cc550e073114bbf983635919991295f : SUCCESS in 43m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dbf7702f13c64c18b2fab6d19ef28ff2 : FAILURE in 50m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/186dcd02fce44bbc99cc99a03e941757 : FAILURE in 25m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cbd722f6410c4215ae76e25a4bffe5b8 : SUCCESS in 52m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c6c2c4425b5843b481d6e0c5a00e96cd : SUCCESS in 1h 06m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8fbe5e93c1742e5bcb425c3c66e881d : FAILURE in 52m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/472e90ba8f124d5586cc466b82308a82 : SUCCESS in 23m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ac7b51c6e04e4684b686112d5994a30e : SUCCESS in 38m 27s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/17c23d7cbc994519b7447eb9c879132c : SUCCESS in 49m 20s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c194d54573e14028987333c4e0e38f9f : SUCCESS in 39m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f0efaa8ada8240259fe1cfa4e75b42ed : SUCCESS in 24m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e9bef5018de544c69f06fac0bf74927f : SUCCESS in 34m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/098ce88d506141ebb58f92a5cfa49ec6 : FAILURE in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"88117946659a62b721e99cb6b8c0f78d96466ad3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-12 07:04:59.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_PowerFlex_Manila_dhss_false/c471fd8/DellEMC_PowerFlex_Manila_dhss_false/355 : SUCCESS in 32m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"2740879b3b16be4f909b7f3748569bdf8efefc6b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-12 07:08:18.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_PowerStore_Manila_dhss_false/15935d6/DellEMC_PowerStore_Manila_dhss_false/151 : SUCCESS in 36m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"91385bb47e4cdae35501834697127b6d0b3890ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-12 09:22:39.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_Unity_Manila_dhss_true/7c27da0/DellEMC_Unity_Manila_dhss_true/573 : FAILURE in 1h 00m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/867213/8/check/DellEMC_Unity_Manila_dhss_false/c26da8c/DellEMC_Unity_Manila_dhss_false/595 : FAILURE in 2h 50m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"ca8c05e8d2f3d9828a12aa8d4444bdbaf52f770f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-12 09:43:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/8/upstream-check/manila-tempest-plugin-ontap-dhss/9b2db0f/ : RETRY_LIMIT in 7m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/867213/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/da1b85b/ : TIMED_OUT in 3h 10m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"10ecf62066328fb5ede26018ba18bda3c7c9cd97","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 18:54:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d8f627cd3e1ca59b340a9725d255fcf052914f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:55:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"65f582bc5c90a3f56afcdfc4c2caa7543a49b72b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 19:40:25.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a4472ba6784c3a96f25a3b8511a802\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3006b8a1df5746f6a262a736bd263e6d : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/28473e7632534e79be47991967cef3b0 : SUCCESS in 35m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c40f9cc9f9b547d2b44dd946e7ba3916 : SUCCESS in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9a1fb560a50c4fa5bd5c2bb0501d25fc : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":8},{"id":"fb1fd6b858283dc9a6189f2920b561f8048a9a37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 19:40:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0f7345a4ddc3b777c6f0eb880df26a8673a5361b","revisions":{"6f4712e913ffb01d27fe39e0d3b3bcc3daa4dc6c":{"kind":"REWORK","_number":1,"created":"2022-12-12 12:12:58.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/1"}}},"commit":{"parents":[{"commit":"287e44b85f70b652406ee1a6a98d229ee74adf6f","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/287e44b85f70b652406ee1a6a98d229ee74adf6f"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:12:42.000000000","tz":120},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f4712e913ffb01d27fe39e0d3b3bcc3daa4dc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f4712e913ffb01d27fe39e0d3b3bcc3daa4dc6c"}]},"branch":"refs/heads/master"},"a34a9e0c51f60cefeb421f7f87e239df8c55ce2a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-15 09:18:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/2"}}},"commit":{"parents":[{"commit":"c1e2199f027a5884a094bf2711d4b42cddff6a07","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1e2199f027a5884a094bf2711d4b42cddff6a07"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-01-15 09:18:17.000000000","tz":0},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a34a9e0c51f60cefeb421f7f87e239df8c55ce2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a34a9e0c51f60cefeb421f7f87e239df8c55ce2a"}]},"branch":"refs/heads/master","description":"Rebase"},"04316fab9a9db023d1960dce79fc2418872396c6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-16 06:08:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/3"}}},"commit":{"parents":[{"commit":"9c92df53f7b66e0cd4d844a3f6bc1459c8c3d2b2","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c92df53f7b66e0cd4d844a3f6bc1459c8c3d2b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-01-16 06:08:26.000000000","tz":0},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04316fab9a9db023d1960dce79fc2418872396c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04316fab9a9db023d1960dce79fc2418872396c6"}]},"branch":"refs/heads/master","description":"Rebase"},"cf30e766cbb06c0c089c8dcc30081e9c0fcc31b5":{"kind":"REWORK","_number":4,"created":"2023-01-29 11:11:06.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/4"}}},"commit":{"parents":[{"commit":"88d91966ee47f7d2a96a4ee90c0a6283b58607eb","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88d91966ee47f7d2a96a4ee90c0a6283b58607eb"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-29 11:10:40.000000000","tz":120},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf30e766cbb06c0c089c8dcc30081e9c0fcc31b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf30e766cbb06c0c089c8dcc30081e9c0fcc31b5"}]},"branch":"refs/heads/master"},"222a5d2f69368e8448c0130638d467bb54f98da7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-29 13:25:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/5"}}},"commit":{"parents":[{"commit":"50eebc0ed0b5176253ce7288755fdf1f365a5709","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50eebc0ed0b5176253ce7288755fdf1f365a5709"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-29 13:25:40.000000000","tz":120},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/222a5d2f69368e8448c0130638d467bb54f98da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/222a5d2f69368e8448c0130638d467bb54f98da7"}]},"branch":"refs/heads/master"},"ef493e1686be5ea90f6f00f6e10a984d5fd8d7e6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-30 07:01:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/6"}}},"commit":{"parents":[{"commit":"8c9cfd7123f2d5a77ed093ce39792023ce244bcc","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c9cfd7123f2d5a77ed093ce39792023ce244bcc"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-03-30 07:01:22.000000000","tz":0},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef493e1686be5ea90f6f00f6e10a984d5fd8d7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef493e1686be5ea90f6f00f6e10a984d5fd8d7e6"}]},"branch":"refs/heads/master","description":"Rebase"},"92a5ee48e59133758266d6a39e29764c7d509b57":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-07-18 12:19:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/7"}}},"commit":{"parents":[{"commit":"796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04","subject":"[RBAC] Add share group type tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/796f3ebff8cd3c4a08a205dcf5d0b87d550d3b04"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-07-18 12:19:19.000000000","tz":180},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92a5ee48e59133758266d6a39e29764c7d509b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92a5ee48e59133758266d6a39e29764c7d509b57"}]},"branch":"refs/heads/master"},"0f7345a4ddc3b777c6f0eb880df26a8673a5361b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-09-06 07:01:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/13/867213/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/867213/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/867213/8"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-12 12:11:07.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-09-06 07:01:22.000000000","tz":0},"subject":"[RBAC] Add share groups tests","message":"[RBAC] Add share groups tests\n\nThe tests validate policy of share groups actions for\nadmin, member and reader users in a project scope.\n\nChange-Id: Iee0f739d9701b2efe9b275ea22859318b58a46be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f7345a4ddc3b777c6f0eb880df26a8673a5361b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f7345a4ddc3b777c6f0eb880df26a8673a5361b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
