)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867231","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I97001030fef2b6c6b90f0b42670c0acf7659d47a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{"33986":{"account":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"last_update":"2022-12-12 17:28:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I97001030fef2b6c6b90f0b42670c0acf7659d47a","subject":"(WIP) RBAC tests for Share Servers","status":"NEW","created":"2022-12-12 14:08:06.000000000","updated":"2022-12-15 15:31:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f8db7122f67d2f465826749cef4c67d39940196","_number":867231,"virtual_id_number":867231,"owner":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-13 18:48:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 17:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 23:07:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"d736ccc326a88558aff498c58b40312a14a4b8f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2022-12-12 14:08:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a29d0af0207d5c0225678384b049f5291e1c53c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2022-12-12 15:07:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f0fdd4ae8172fc6526f189a801dacd2a60f1e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 17:28:02.000000000","message":"Patch Set 2: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea3d48205ce3467e8e570ede0b998eab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4af36fe94604b0892c98282a254918a : FAILURE in 3m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/60ad21cd7b044aa5b66a264df2f324cd : SUCCESS in 1h 03m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f3f7d9e5cdcd4ebd8e30b75f6154d863 : SUCCESS in 1h 42m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2dfb779b508349cba9da3d12055176f3 : SUCCESS in 1h 02m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48fee749ec4440a8acbcce30ecc983f7 : SUCCESS in 1h 09m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e5e4510fd39c405180078080c7ec2972 : FAILURE in 2h 14m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/caca23d467ac4d609ed04bfe68171a7c : FAILURE in 55m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/401ec030ea8d45f2b45556ad1378e2d5 : SUCCESS in 24m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/30efcbf1959a44d7840f3843b2c273d0 : SUCCESS in 38m 08s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c5cf61a8ffc34db9a9ac52d5f5a7bfbc : SUCCESS in 1h 12m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c47662dcff4041029e5dea19251299d6 : SUCCESS in 46m 23s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5d868b3efd0f454da030a2fec7342783 : SUCCESS in 1h 13m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ac2b3884f6cc4f8c9c2732b2d1fe098b : FAILURE in 23m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8f5e68cc1a2a4f0ea2135601ccac3358 : SUCCESS in 35m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d1a2ca5fe65849fd9da38cb2c9f7e269 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"702885da4c7415a152c4c946ed6422cabfaa53d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-12 23:07:20.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/867231/2/check/DellEMC_Unity_Manila_dhss_true/b6da89e/DellEMC_Unity_Manila_dhss_true/1080 : FAILURE in 3h 57m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/867231/2/check/DellEMC_Unity_Manila_dhss_false/c225e87/DellEMC_Unity_Manila_dhss_false/1153 : SUCCESS in 1h 01m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"831a362fe17f456a520560e61bdb2f385d5901b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2022-12-13 16:56:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ca234c219718008b66cc29d47b2436534ab9202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 18:48:47.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b77106a7199442d9a7b246eee416ec9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0a2c0731984d4984d6dacd55dd644f : FAILURE in 2m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99da3f958df44608a57d8111d25523e9 : SUCCESS in 1h 08m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/16fb773a468945c9b4b84ff9dcf5d482 : SUCCESS in 1h 50m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/974cb2d5e7bb42b28b41119bcdc2c4a4 : SUCCESS in 1h 12m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d35c1c8a057b4cc3998d36388535af68 : SUCCESS in 1h 04m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ed6cc35c3987432f9e2622e8fe80f0d7 : FAILURE in 1h 31m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d93e47ebb4d74f24b6e923fa235c0837 : FAILURE in 1h 08m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e7a923288434c128b159953ef0ff0ac : SUCCESS in 29m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ca5a609aaf74f51bbc2ae39d1fead49 : SUCCESS in 29m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/526054278bfe470884ec5059e600ff73 : SUCCESS in 1h 24m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/35efbbc1d6b04af6893cf3836511e9cf : SUCCESS in 58m 48s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/23a155e6d4d241728d97991a6321775e : SUCCESS in 1h 00m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/79b12ae26f444acc86e451d02bbaa6ac : FAILURE in 22m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/da04e38066de4467a4587b377ebb0218 : SUCCESS in 38m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b4e3910eb7b64868a4a2d291e3286d77 : SUCCESS in 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7f8db7122f67d2f465826749cef4c67d39940196","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2022-12-15 15:31:37.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"96eb2b71711b6ded0eaf970401649a7d114b0578","revisions":{"89008f0f2dfd2d8febafe88cdd76821f29014a4e":{"kind":"REWORK","_number":1,"created":"2022-12-12 14:08:06.000000000","uploader":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"ref":"refs/changes/31/867231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/867231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lucasmoliveira059","email":"lucasmoliveira059@gmail.com","date":"2022-12-12 14:05:55.000000000","tz":0},"committer":{"name":"lucasmoliveira059","email":"lucasmoliveira059@gmail.com","date":"2022-12-12 14:06:56.000000000","tz":0},"subject":"(WIP) RBAC tests for Share Servers","message":"(WIP) RBAC tests for Share Servers\n\nChange-Id: I97001030fef2b6c6b90f0b42670c0acf7659d47a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/89008f0f2dfd2d8febafe88cdd76821f29014a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/89008f0f2dfd2d8febafe88cdd76821f29014a4e"}]},"branch":"refs/heads/master"},"90f4bfb4623b534821c5ca0660c2d0661e14ce95":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-12 15:07:25.000000000","uploader":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"ref":"refs/changes/31/867231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/867231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lucasmoliveira059","email":"lucasmoliveira059@gmail.com","date":"2022-12-12 14:05:55.000000000","tz":0},"committer":{"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","date":"2022-12-12 15:07:25.000000000","tz":0},"subject":"(WIP) RBAC tests for Share Servers","message":"(WIP) RBAC tests for Share Servers\n\nCo-Authored-By: Helena Dantas \u003chelenamylena@gmail.com\u003e\nChange-Id: I97001030fef2b6c6b90f0b42670c0acf7659d47a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90f4bfb4623b534821c5ca0660c2d0661e14ce95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90f4bfb4623b534821c5ca0660c2d0661e14ce95"}]},"branch":"refs/heads/master","description":"Edit commit message"},"96eb2b71711b6ded0eaf970401649a7d114b0578":{"kind":"REWORK","_number":3,"created":"2022-12-13 16:56:41.000000000","uploader":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"ref":"refs/changes/31/867231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/867231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/867231/3"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"lucasmoliveira059","email":"lucasmoliveira059@gmail.com","date":"2022-12-12 14:05:55.000000000","tz":0},"committer":{"name":"lucasmoliveira059","email":"lucasmoliveira059@gmail.com","date":"2022-12-13 16:51:20.000000000","tz":0},"subject":"(WIP) RBAC tests for Share Servers","message":"(WIP) RBAC tests for Share Servers\n\nCo-Authored-By: Helena Dantas \u003chelenamylena@gmail.com\u003e\nChange-Id: I97001030fef2b6c6b90f0b42670c0acf7659d47a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96eb2b71711b6ded0eaf970401649a7d114b0578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96eb2b71711b6ded0eaf970401649a7d114b0578"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
