)]}'
{"id":"openstack%2Fmanila-tempest-plugin~867358","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0f1b2d775d62d2dda888eb8db41e784a778857eb","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-05-10 13:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_33301\u003e replied on the change","reason_account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-10 13:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_33301\u003e replied on the change","reason_account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}},"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-05-10 16:02:56.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-11 09:15:17.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I0f1b2d775d62d2dda888eb8db41e784a778857eb","subject":"[RBAC] Tests for Share Instance","status":"NEW","created":"2022-12-13 12:33:27.000000000","updated":"2023-05-11 09:15:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":19,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"d811c1da6d44d2a292c28aabd888ba2690f803c1","_number":867358,"virtual_id_number":867358,"owner":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-10 16:02:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-05-11 09:15:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-13 14:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 16:15:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-17 00:33:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-03-30 06:48:54.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"CC"},{"updated":"2023-03-30 06:49:05.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-05-09 15:22:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-05-10 17:54:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"39b70f25a191b23cac88e9f168ba371853a1d7b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-12-13 12:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d10399f2345cafedd4df7e1a39dc2b453116939f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435332fc06614648af6a7972aa506e48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1fb6e6982b4a2e89fa75b040f593f9 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e4d1fb7849b541a18de155f36023facb : SUCCESS in 1h 00m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/41d5ecc765bb456a8ee4266ac7f45b8e : FAILURE in 1h 59m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a40c0138464a4a7e966daa0df77a8427 : SUCCESS in 1h 00m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f7f77196d5cb4caebe8129fadb30b09c : SUCCESS in 1h 11m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1549f3482c074313be086e580b4bfc13 : FAILURE in 1h 27m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e5d36af3eaa6408ca38f5573a75d099d : FAILURE in 56m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab698a0da0c044eaa82a8e25ee3241bb : SUCCESS in 19m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/82d99d3df4ee42f7a0d9503e3f2c67be : SUCCESS in 29m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2b555d3ccce54c5a89bf9d9945d2fcb9 : SUCCESS in 1h 11m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ea3ca157fb514a32838e685605aa3eba : SUCCESS in 1h 20m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e3129b6022644c07a2de243896233522 : SUCCESS in 1h 20m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/727f87b937534185a8e94bfe30eb6216 : SUCCESS in 21m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7e879a104af449da9daaf2e3ef935001 : SUCCESS in 41m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25abc0aafdfd41bcb18b70b1ff39e5c8 : SUCCESS in 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"326d04e9374179b493cbad1c5bd9989c3375fce6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-13 16:15:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/867358/1/upstream-check/manila-tempest-plugin-ontap-dhss/231c871/ : SUCCESS in 1h 40m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/867358/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/3a38896/ : SUCCESS in 1h 21m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fa0cf1c40a9a8535204b355c32cdc991b74b3e63","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-17 00:33:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3525add291b6cd27ff7aa6e2c1a2211e95aa9559","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-20 15:11:00.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"970e1a644111362e2907a4d7a9be3ec97e0131f8","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 06:48:54.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ad321db279a411294a4822a89ceab425c6ae9517","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 06:49:05.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"4aa8b92aa1f10138d949d1a7e2585d158cc51ba7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-09 15:22:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a7f62022ec4263bced7787a6646a7119b36797e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 17:34:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff1ab8a65b204dc1987fa5248bdc46e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57592f2b1c794367be8d4afad3639b9b : FAILURE in 3m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/64951e99ba734b43bf9b5490682fd57b : SUCCESS in 1h 36m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2ba22d4aab9d4153b9df80a9d6e9b0eb : FAILURE in 2h 02m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77aea7366bb741228237d4a4a939d612 : SUCCESS in 1h 11m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cf3f9005fb704cdfaf3970e8134992e6 : SUCCESS in 1h 08m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7b769214fb6b4a3db0093a56d004a444 : FAILURE in 13m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9f62ae13570482aaee249ac9ed5404b : SUCCESS in 1h 13m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b1cc1373728e4aecb7a1781d647700b4 : SUCCESS in 39m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9991830eb01645f889e27a255fd6f1c0 : SUCCESS in 43m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/63e59bfcfd574f99b1e6dc41fc0bfd0d : SUCCESS in 1h 25m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/79ffb59f928242edafcd9b7490150718 : SUCCESS in 1h 28m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b0ced284fbb54ce09f2820a44071c482 : SUCCESS in 42m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d67b267256f54fb68323828faf07b20a : SUCCESS in 49m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb26efaf8b97460bbf3abad59dcd10e3 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34c0ea5adb786acdff52eb937ea5ad5af1971fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-05-10 13:45:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4391c8a870cc9dcc15aaa4306e7c781b85beeec6","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-05-10 13:48:09.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"44df0d23c0ff19c513ffde42b35a899ed360b2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 16:02:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42bd8e1cd22b4b3184a7e2ae906f492e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccdde58b2b974c459e7f4f77a2e06d6a : FAILURE in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f889a22f553b44e9aafab31d918a46f6 : FAILURE in 1h 33m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/01d91220741040248484c2e9c534eb87 : SUCCESS in 2h 12m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/81c7e2ec43e048818d960725bce5eedb : SUCCESS in 1h 34m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0df4b5b693ae487f96b16f2ac6ae3b21 : FAILURE in 1h 21m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5906b60b170f4e2ca3b32aa8217c7d5b : FAILURE in 14m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3acca74e578d4c71ba4f34cd6cbb7370 : SUCCESS in 1h 17m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c1f636e139e544c088dbe63a347e0ee7 : SUCCESS in 38m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cced3e3b2d1c472b992d706f5c81b501 : SUCCESS in 38m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e97b1fc7c2dc4400981d03c6e180852b : SUCCESS in 1h 28m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/86722c335d854eec82335794f0eb38cb : SUCCESS in 1h 20m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/099c4cce01fd45b9a73fb9a792496cfc : SUCCESS in 37m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/acc16d20fd8343ee88a4d385a9818722 : SUCCESS in 51m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a1a9f3c5d6f04436baacd15222236526 : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cdc0de6ba784b674149911678921bd6a647cda0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 17:54:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/58/867358/2/check/DellEMC_Unity_Manila_dhss_true/6fd9358/DellEMC_Unity_Manila_dhss_true/100 : SUCCESS in 2h 38m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/58/867358/2/check/DellEMC_Unity_Manila_dhss_false/c61c9e3/DellEMC_Unity_Manila_dhss_false/100 : SUCCESS in 1h 03m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d811c1da6d44d2a292c28aabd888ba2690f803c1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-11 09:15:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7a4d963b0ee168254ea6d91c66c2db34c2f65a5","revisions":{"d31ecc0eb6bf62d839416589fdf209904222d100":{"kind":"REWORK","_number":1,"created":"2022-12-13 12:33:27.000000000","uploader":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"ref":"refs/changes/58/867358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/867358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/1"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"CaiqueMello","email":"caiquemellosbo@gmail.com","date":"2022-12-13 12:31:47.000000000","tz":0},"committer":{"name":"CaiqueMello","email":"caiquemellosbo@gmail.com","date":"2022-12-13 12:31:47.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance","message":"[RBAC] Tests for Share Instance\n\nThe tests validate policy of share instance actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I0f1b2d775d62d2dda888eb8db41e784a778857eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d31ecc0eb6bf62d839416589fdf209904222d100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d31ecc0eb6bf62d839416589fdf209904222d100"}]},"branch":"refs/heads/master"},"e7a4d963b0ee168254ea6d91c66c2db34c2f65a5":{"kind":"REWORK","_number":2,"created":"2023-05-10 13:45:18.000000000","uploader":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"ref":"refs/changes/58/867358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/867358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/867358/2"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"CaiqueMello","email":"caiquemellosbo@gmail.com","date":"2022-12-13 12:31:47.000000000","tz":0},"committer":{"name":"MelloCaique","email":"caique_mellosbo@hotmail.com","date":"2023-05-10 13:44:47.000000000","tz":0},"subject":"[RBAC] Tests for Share Instance","message":"[RBAC] Tests for Share Instance\n\nThe tests validate policy of share instance actions for admin,\nmember and reader users in a project scope.\n\nChange-Id: I0f1b2d775d62d2dda888eb8db41e784a778857eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7a4d963b0ee168254ea6d91c66c2db34c2f65a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7a4d963b0ee168254ea6d91c66c2db34c2f65a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
