)]}'
{"id":"openstack%2Fmanila-tempest-plugin~871775","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia679ab28d9fd9d79294e990c39cd4cd9631f1742","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-01-26 11:51:33.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-01-26 15:24:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia679ab28d9fd9d79294e990c39cd4cd9631f1742","subject":"Skip replication test if microversion is not supported","status":"MERGED","created":"2023-01-25 19:05:02.000000000","updated":"2023-01-26 16:58:13.000000000","submitted":"2023-01-26 15:24:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871775","meta_rev_id":"87ba1c6b891d3f3d33bf22a420ac42c31bff9d6a","_number":871775,"virtual_id_number":871775,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 15:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-01-26 14:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-01-25 20:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-01-26 14:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 19:44:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-01-25 20:16:22.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-01-25 22:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 16:58:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"16d41cbaf7d5b87c97c0d11ba75c0d00fb1e49f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-25 19:05:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4feda38609c8d0ddf27632ab63bd94afc14e627","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-01-25 20:16:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2170485eea3227092d4ecb49c0ad97ea2e2c48c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 22:21:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/074cef10d28e401a9070caa0d93efd9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4e4ac371ed4524995a1f65cda0be03 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/072bf3fc2b264542affa41b31eee4e30 : SUCCESS in 1h 26m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/19e823ae9740476e89b860a816573ddd : TIMED_OUT in 3h 04m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/17cebf72a0144029b20a027adfe0d247 : SUCCESS in 1h 09m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/29c3cfe58d404fe3bdf57a90a1ea8448 : SUCCESS in 1h 35m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3904254b184844b8be811c4f650fe5e9 : FAILURE in 1h 51m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/05918ca3e7f441a29bb3606f72ec03f4 : SUCCESS in 1h 31m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9d51b496d9b342b3a5c9b1baff9c6a2c : FAILURE in 9m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/08e96efd320148ffbcd5014287d4fe74 : SUCCESS in 33m 32s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ed705371d8b543d39dc48c7fa3aedc13 : SUCCESS in 1h 23m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/447c75b52bc1466e9c5f301bb16f8729 : FAILURE in 1h 07m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f48ca04ee3dc4da5b6edbcef946fa83f : SUCCESS in 1h 17m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fc6095bd58234d1095d8c58be989a0be : SUCCESS in 19m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0bd827a20aa84a3b9710603ddfbb0e39 : SUCCESS in 48m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/36cd03a414ce4de7972bb36803a35a4c : SUCCESS in 28m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cd3f4960da04e766662e724d7e4a12675a2091d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-26 11:51:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"439f9e38b444a9410a0cb04fdad95b54c94d3ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:48:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ea33409ca54826a791cd40f32b21ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5383f705c7cd4803b34d1de700884602 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a77e5f91acd448498a8167ade19c644d : SUCCESS in 1h 27m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/835636442a4948f29758ca5fb1bb3415 : FAILURE in 1h 53m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/412265e642754ae4b0f13f0f2e3bd4d9 : SUCCESS in 1h 02m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/37c1ea9b499043ff873426431d7e782d : SUCCESS in 1h 04m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/739fb95d8e584c938a54d3601d9ddb6e : SUCCESS in 1h 42m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6418ba3d26cf471587c994af3f25bc9e : SUCCESS in 1h 14m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/45d088c80db24deb809befa84cee492b : SUCCESS in 27m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7e1ced85061f4975bb4049d42843abc2 : SUCCESS in 33m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0169da0ab2e94bfda738a4c8af55cbd9 : SUCCESS in 1h 04m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5b6b5a67c57340888748b8afddaad31a : SUCCESS in 52m 31s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/79c52ac4f0344ae7bb0d95c8ce354c3a : SUCCESS in 1h 27m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4ddfda221f3042d094cd4cea161201e0 : SUCCESS in 25m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/895c7177a7454cb09864b73f015d8685 : SUCCESS in 48m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9855045d2b8f4c91ad7a20fb310f3304 : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76fe2e80aba889db1ab0ed2ff1c53726c47690fe","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-26 14:09:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85e96fa344f94e917efedb9d933ed90c61a39464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:10:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f635f84ba0e553fdd26aecfa816c9b75fdcf3a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:24:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650284f8c1054694a3ba083cf1cab552\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b0a2837e03b42abb07d4bef858cbe7c : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6a2acebc08db44deabec8b37e392701e : SUCCESS in 1h 14m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b59b81b30c8540beab97b8eea2088e24 : SUCCESS in 45m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c92d6019c4bb46ba99da285bb72e0009 : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3aa2c63093869a17e083691b99bf86e5e6522920","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:24:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87ba1c6b891d3f3d33bf22a420ac42c31bff9d6a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-26 16:58:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/871775/1/upstream-check/manila-tempest-plugin-ontap-dhss/cf6cfd7/ : TIMED_OUT in 3h 08m 51s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/871775/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/e7da87f/ : SUCCESS in 1h 23m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","revisions":{"54e31e8f5b517d2f1b350c091b29aa518f2c8739":{"kind":"REWORK","_number":1,"created":"2023-01-25 19:05:02.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/75/871775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/871775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/871775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/871775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/871775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/871775/1"}}},"commit":{"parents":[{"commit":"e4f923feb7f76d63dfbf50366314cc1209437d45","subject":"Merge \"[RBAC] Add Tests for list share services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4f923feb7f76d63dfbf50366314cc1209437d45"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:04:54.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:04:54.000000000","tz":-480},"subject":"Skip replication test if microversion is not supported","message":"Skip replication test if microversion is not supported\n\nWithout the skip, we break testing against old/stable branches.\n\nChange-Id: Ia679ab28d9fd9d79294e990c39cd4cd9631f1742\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
