)]}'
{"id":"openstack%2Fmanila-tempest-plugin~871778","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Icf221a3169d352e1bc4d09b9139e4e1dac75d774","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-jobs","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-08 18:42:16.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"Icf221a3169d352e1bc4d09b9139e4e1dac75d774","subject":"Swap wallaby with stable/zed LVM job","status":"ABANDONED","created":"2023-01-25 19:16:09.000000000","updated":"2024-10-24 21:13:54.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f93686fd3f5d92b0ed4fb73b1e8db92f91968d54","_number":871778,"virtual_id_number":871778,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-27 02:48:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2023-01-27 01:04:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2023-02-08 18:42:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 21:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 01:04:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-01-27 04:38:39.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"eb104655e9322636bedcdbf051f58469b0fcff2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-25 19:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5bb7ca115433ae3a3f2e4cb273bd4746ea75702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-25 19:40:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c7c5e135fbe44bf1f69e791c58e8da96fe91c9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:22:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03d3808a560b4a3ead6d23bd8c9a7c33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a0629d3723433ba14cfaf3ee16cd94 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/380bb644f732439d8879748aa051ce60 : SUCCESS in 59m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d0a87840e8b442fdbba4a3eae1174ec1 : FAILURE in 1h 27m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99d00901ad754d838abbb62e480f1c1b : SUCCESS in 1h 00m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3a259399ab3549ad9d2a39813d7e9b0f : SUCCESS in 1h 26m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4247de6f325a4913a9f768505038435c : SUCCESS in 1h 39m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7bef13aa19874bbabd582141d1be4602 : FAILURE in 1h 04m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3f6afbf98c0b48c9b964341a05ce58d6 : SUCCESS in 38m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1d2a0ee4942a4925a3ece313e309cc41 : FAILURE in 48m 37s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/1aaf273013774c6286b6fd88b74f5ac0 : SUCCESS in 1h 00m 15s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2be5621f14024f0cb55dd4373fa2b326 : SUCCESS in 1h 29m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f875c3c1605941fc82b0e54e6103907d : SUCCESS in 1h 20m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4dcae7ecbdb7485f9bd517ae92e79d86 : SUCCESS in 32m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/277ba2a9749e41e3b5beab0444bb9c93 : SUCCESS in 38m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/113407b0df074dc691d7369d9b1ff248 : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"794214ff20b2fd3c0a9f662bb44e66292d9fbc60","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 01:04:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ff12641d51f687dbd519a859473505107a94d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 02:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd4a00a29e9423095b6d4c3c96f10ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3deb8ecdb644cfda6a687735ac12be6 : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8a53874df0834c97b864979ea2ce0d01 : SUCCESS in 1h 02m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fdd97e98a7074c1eb822a5bd6bc7496b : SUCCESS in 1h 41m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f13d823d4594e028093e317d38c2520 : SUCCESS in 56m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/38afd41b43ae479ab71be72aa23b983a : SUCCESS in 1h 03m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/51f549c4cbfa41aab071678f003b9bbf : SUCCESS in 1h 39m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa5e4bab64c848d8a4adf0f97ccccdf8 : SUCCESS in 1h 33m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a2a71cf7efa6413c95cd73bb96a282b6 : SUCCESS in 25m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bb4a11d48cca4f178068a7f00d73f7bd : SUCCESS in 31m 47s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/68b7c8909ed649d38b54a4ad3b3dd29d : SUCCESS in 1h 01m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/14c8242ed38d4d459c8f9964af948868 : SUCCESS in 55m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2586a891e4334bca91e943408b071125 : SUCCESS in 1h 08m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/30d4ef91fa3041c8a301bdaddd9d9cd0 : SUCCESS in 33m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/18105bf9f9c3410b9dd43b53ea9be9b3 : SUCCESS in 1h 03m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/15b469503ead439f9aba6f60bf28ed35 : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58030a8d5af4d60b4fe6f65951eac0807228ba0e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-27 04:38:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/871778/2/upstream-check/manila-tempest-plugin-ontap-dhss/fc64c58/ : SUCCESS in 1h 49m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/871778/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/5968568/ : FAILURE in 1h 19m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5726bf244183da459840acde3de8c8af4da1c2ae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-08 18:42:16.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f93686fd3f5d92b0ed4fb73b1e8db92f91968d54","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-24 21:13:54.000000000","message":"Abandoned\n\nWe\u0027ve transitioned stable/zed to unmaintained/zed.. the jobs were adjusted elsewhere","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f781dede452f413b5f283a282e3f7b7528738f78","revisions":{"7aa292681ceaa86608c1c8c5b99694b0557dcf0c":{"kind":"REWORK","_number":1,"created":"2023-01-25 19:16:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/78/871778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/871778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/1"}}},"commit":{"parents":[{"commit":"e4f923feb7f76d63dfbf50366314cc1209437d45","subject":"Merge \"[RBAC] Add Tests for list share services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4f923feb7f76d63dfbf50366314cc1209437d45"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:15:42.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:15:58.000000000","tz":-480},"subject":"Swap wallaby with stable/zed LVM job","message":"Swap wallaby with stable/zed LVM job\n\nthe wallaby release of openstack is now\nin extended maintenance and we have tagged\na \"wallaby-last\" on manila-tempest-plugin.\n\nWe\u0027ll save CI resources by switching the\njob here to test stable/zed instead - the\nlast stable release as of this patch.\n\nThere\u0027s a risk of breaking CI on wallaby\nbranch with a tempest plugin change - it\u0027ll\nhappen eventually. When it does, we should\nuse the \"wallaby-last\" tag as a way to get\nthe CI unblocked on that branch.\n\nChange-Id: Icf221a3169d352e1bc4d09b9139e4e1dac75d774\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7aa292681ceaa86608c1c8c5b99694b0557dcf0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7aa292681ceaa86608c1c8c5b99694b0557dcf0c"}]},"branch":"refs/heads/master"},"f781dede452f413b5f283a282e3f7b7528738f78":{"kind":"REWORK","_number":2,"created":"2023-01-25 19:40:58.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/78/871778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/871778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/871778/2"}}},"commit":{"parents":[{"commit":"e4f923feb7f76d63dfbf50366314cc1209437d45","subject":"Merge \"[RBAC] Add Tests for list share services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4f923feb7f76d63dfbf50366314cc1209437d45"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:15:42.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-01-25 19:40:57.000000000","tz":0},"subject":"Swap wallaby with stable/zed LVM job","message":"Swap wallaby with stable/zed LVM job\n\nthe wallaby release of openstack is now\nin extended maintenance and we have tagged\na \"wallaby-last\" on manila-tempest-plugin.\n\nWe\u0027ll save CI resources by switching the\njob here to test stable/zed instead - the\nlast stable release as of this patch.\n\nThere\u0027s a risk of breaking CI on wallaby\nbranch with a tempest plugin change - it\u0027ll\nhappen eventually. When it does, we should\nuse the \"wallaby-last\" tag as a way to get\nthe CI unblocked on that branch.\n\nChange-Id: Icf221a3169d352e1bc4d09b9139e4e1dac75d774\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f781dede452f413b5f283a282e3f7b7528738f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f781dede452f413b5f283a282e3f7b7528738f78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
