)]}'
{"id":"openstack%2Fmanila-tempest-plugin~871779","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I79ef199c5fe8b3aa37a3932d5dd48b54dd9b7a36","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-11-04 12:22:38.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-01-26 15:28:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I79ef199c5fe8b3aa37a3932d5dd48b54dd9b7a36","subject":"Add dummy dhss job to stable/zed","status":"ABANDONED","created":"2023-01-25 19:31:54.000000000","updated":"2024-11-04 12:22:38.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6fb8e30823c7fb4ca9247e07a1f89a8442a930a4","_number":871779,"virtual_id_number":871779,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-24 23:42:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-01-26 15:28:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-01-26 15:28:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 19:33:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-01-25 21:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 15:28:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-01-26 18:55:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"166950ed41826f502cf0d6c3786584ba99cfbbe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-25 19:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc104e8e906bfc216d695db74a0b1c347d9948aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-25 19:33:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3624b0db5164516802650fc16d7149ec89fae48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be44433018814995a46a76c6ee30b69b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a7cd0ac5b944c7be17a87a0718bea3 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fdf59f9a39bf4c909b8abc5ff91e2fe4 : SUCCESS in 1h 15m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3eb2e2bd747d4c4bb528080221eb614a : SUCCESS in 1h 53m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ddb52eaba468496aad1b3e32232d7ae6 : SUCCESS in 1h 06m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0b37028da88a468e928be7f178c4e783 : SUCCESS in 1h 23m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/db5cc054f8554e26b123875fc8fa65e3 : SUCCESS in 1h 36m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb80305ed27240f19e94b0de086052f5 : SUCCESS in 51m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f3bd9b1f64154a0892994980d83068c9 : SUCCESS in 31m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a97c78c082ba4a989704bdf1255adfbb : SUCCESS in 25m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8537a1e0c88c43b897260cfdc321c8c7 : SUCCESS in 1h 16m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/77c70f699cdb4f51b8620932a4e4fa0f : SUCCESS in 1h 06m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d132cf90715c4201b76d11be7139fae6 : SUCCESS in 1h 14m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e28f6f4c455845c29a698524a5285159 : SUCCESS in 21m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5dfb3f7d6b4c43e789d626908642436e : SUCCESS in 51m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/61ac32c045494e23af3e4089920e8a3b : SUCCESS in 21m 14s (non-voting)\n- manila-tempest-plugin-dummy-dhss-zed https://zuul.opendev.org/t/openstack/build/6c446fe1730c441689f974a650cd51ed : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"850387c3e459580a507204741042be62318aa527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-26 12:06:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aff6bfdd1691ffac4de50dbaf02aca4324a98a55","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-26 12:07:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d9da4a3b59b3cbf4a5de32315e1dfb22aa97bc0b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-26 15:28:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aca3b48dce3132020790553bf8f556b5567f77fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:14:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8002e4459a9a4326a408f6a658141704\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5676f506894027ad059992906cbabc : SUCCESS in 3m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/37e0b100dc0048b4b9154db6c571ebb3 : SUCCESS in 1h 23m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/451ce0ab9fc94c15b73d9486ca03136e : FAILURE in 1h 50m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6c9a28853d44ed69dbdf92b7bf41968 : SUCCESS in 1h 03m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/849235496c56432aaeaed8acef95872d : SUCCESS in 1h 23m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/631abc0a3d504f21aca9c2bde01bdade : SUCCESS in 2h 05m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43003fd08fc44f0d96cf74ab3afde185 : SUCCESS in 1h 09m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2e432dc7c794a6c9b2a17f858402d1a : SUCCESS in 37m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4c282b899fce4bce98fa161e55a2c5db : SUCCESS in 35m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/78ddfaa47b164c3eb6070eb0c64578ef : SUCCESS in 1h 19m 57s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/64c5b5c9dac04900b8b4ed0847cf24a1 : SUCCESS in 1h 05m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1ac8d5d4c6194cdb886301c89b7db83b : SUCCESS in 1h 13m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/66a2cff6c196453f8b4de17ae8377fb1 : SUCCESS in 28m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0e53c40dbe184ec884344eeee4b7c454 : SUCCESS in 50m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e17f6793781c46b1b88977acf5935277 : SUCCESS in 29m 43s (non-voting)\n- manila-tempest-plugin-dummy-dhss-zed https://zuul.opendev.org/t/openstack/build/dca29d0618c44cc2ae34a7007f7c74c1 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6470bcb65c0b2852682d1f1ae01647de2eb733fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:15:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a25f48c57cec39dc9a92d7926d7fe0db98e90af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 18:47:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d46228907f4c4749b6e954f85ce095b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14bc588e6ee435dab340ff793c680cb : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4d445f01606d47919d7f7058121e8b84 : SUCCESS in 1h 30m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46751f415e8b482faf466af94f8d601a : SUCCESS in 28m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/526fe477d6cf493bb46391718a6b0614 : FAILURE in 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"751705ea9fdb50fa5c3c292ae0dc409781c5b5f6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-26 18:55:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/79/871779/2/upstream-check/manila-tempest-plugin-ontap-dhss/6955cdb/ : SUCCESS in 1h 40m 35s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/79/871779/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/3199302/ : SUCCESS in 1h 18m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c0b4ffb4f2bc038039c06b61604cb0a913907c73","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-24 21:14:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac7208d5acc060be87bbc44335e67b761a3264bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:42:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/188d9f79f62541f1a988db097af8bf5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd1e79b58924d05bec1aad0b8bf71b8 : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/27ab65f75e354c35a91ff2735d87818e : SUCCESS in 1h 20m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f82de48143e144439eb163e0d7e7ecfe : SUCCESS in 2h 21m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a78a6077132144078300ec30a2cb2d5f : FAILURE in 41m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f37efcf724a944faae8b9d44bebe37be : SUCCESS in 41m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/35c2190d54e24f1e8306db72a04746a7 : SUCCESS in 1h 07m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa0dc4d734284af89cb6325aa01052c1 : FAILURE in 22m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9efb681022454beb9fda612c248651b7 : SUCCESS in 34m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b179cfe5d3a4b1ab52fc1ad4477bacc : SUCCESS in 39m 54s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c8a9ce64f13745ffbf7338338a7bbc77 : SUCCESS in 41m 15s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/843a271731994cd4ac781d3b111e411c : SUCCESS in 39m 43s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/62c10ec8f0d1485895cb7c139a851d46 : SUCCESS in 51m 14s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/ffbd68db6caa403eb0e05772fd93c347 : SUCCESS in 1h 19m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/496fb03eca644d3caff50377bce745e0 : SUCCESS in 32m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f0f6cbf046d478ca94d73d623d0ce5f : FAILURE in 24m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/db457027609147d9984710f9d209f9ef : FAILURE in 9m 15s (non-voting)\n- manila-tempest-plugin-dummy-dhss-zed https://zuul.opendev.org/t/openstack/build/ab6932d23ccc42cca3820e84bcb15c1d : FAILURE in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6fb8e30823c7fb4ca9247e07a1f89a8442a930a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-11-04 12:22:38.000000000","message":"Abandoned\n\nZed is now unmaintained, so this change doesn\u0027t make sense anymore.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"071b2b22f7dcb24bb3b7853e519f456eb9b877eb","revisions":{"3b4909700a779a97b2365f1c6c97174760ae3674":{"kind":"REWORK","_number":1,"created":"2023-01-25 19:31:54.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/79/871779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/871779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/1"}}},"commit":{"parents":[{"commit":"e4f923feb7f76d63dfbf50366314cc1209437d45","subject":"Merge \"[RBAC] Add Tests for list share services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4f923feb7f76d63dfbf50366314cc1209437d45"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-01-25 19:24:31.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-01-25 19:24:31.000000000","tz":-180},"subject":"Add dummy dhss job to stable/zed","message":"Add dummy dhss job to stable/zed\n\nWe currently have tests being run on stable branches for every\nchange we propose on manila-tempest-plugin. That has helped us to\nprevent bugs and to see how changes are behaving with the older\nbranches.\n\nWe have jobs using the LVM driver, which is great, and provides us\na good API coverage but more importantly good scenario tests\ncoverage. However, the LVM driver has limitations while testing and\ncan not run a lot of the most recent features implemented in manila\ni.e. share replication, share server migration and few other.\n\nSo it\u0027s difficult to anticipate bugs coming in when changes are\nstill being proposed on the active branch, and we would catch bugs\nonly when code is already merged and we see that plugin code on the\nstable branches CI.\n\nSo we introduce a new dummy DHSS\u003dTrue job to be run in every single\nchange proposed against manila-tempest-plugin and test such change\nin one stable branch. With that, we can cover more APIs and reduce\nthe risk of merging a change that will break tests on the older\nbranches.\n\nChange-Id: I79ef199c5fe8b3aa37a3932d5dd48b54dd9b7a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b4909700a779a97b2365f1c6c97174760ae3674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b4909700a779a97b2365f1c6c97174760ae3674"}]},"branch":"refs/heads/master"},"071b2b22f7dcb24bb3b7853e519f456eb9b877eb":{"kind":"REWORK","_number":2,"created":"2023-01-26 12:06:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/79/871779/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/871779/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/871779/2"}}},"commit":{"parents":[{"commit":"e4f923feb7f76d63dfbf50366314cc1209437d45","subject":"Merge \"[RBAC] Add Tests for list share services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4f923feb7f76d63dfbf50366314cc1209437d45"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-01-25 19:24:31.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-01-26 12:06:05.000000000","tz":0},"subject":"Add dummy dhss job to stable/zed","message":"Add dummy dhss job to stable/zed\n\nWe currently have tests being run on stable branches for every\nchange we propose on manila-tempest-plugin. That has helped us to\nprevent bugs and to see how changes are behaving with the older\nbranches.\n\nWe have jobs using the LVM driver, which is great, and provides us\na good API coverage but more importantly good scenario tests\ncoverage. However, the LVM driver has limitations while testing and\ncan not run a lot of the most recent features implemented in manila\ni.e. share replication, share server migration and few other.\n\nSo it\u0027s difficult to anticipate bugs coming in when changes are\nstill being proposed on the active branch, and we would catch bugs\nonly when code is already merged and we see that plugin code on the\nstable branches CI.\n\nSo we introduce a new dummy DHSS\u003dTrue job to be run in every single\nchange proposed against manila-tempest-plugin and test such change\nin one stable branch. With that, we can cover more APIs and reduce\nthe risk of merging a change that will break tests on the older\nbranches.\n\nChange-Id: I79ef199c5fe8b3aa37a3932d5dd48b54dd9b7a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/071b2b22f7dcb24bb3b7853e519f456eb9b877eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/071b2b22f7dcb24bb3b7853e519f456eb9b877eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
