)]}'
{"id":"openstack%2Fmanila-tempest-plugin~872048","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ibf8c885529098417e8ff9b417bf172218f8d684a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"wallaby-pin-tempest","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-07 23:38:43.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-08 01:02:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf8c885529098417e8ff9b417bf172218f8d684a","subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","status":"MERGED","created":"2023-01-27 20:34:48.000000000","updated":"2023-02-08 01:02:30.000000000","submitted":"2023-02-08 01:02:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872048-wallaby-pin-tempest","meta_rev_id":"d35232a363827f456c8266c44d2fdf87277c827b","_number":872048,"virtual_id_number":872048,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 01:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-02-02 14:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-07 23:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-02-07 23:38:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 22:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-29 03:51:49.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-01 15:51:05.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-02-01 15:55:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-01 17:54:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-02 14:06:56.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"0b8281d436511f1d0f5b15847a2998002511f27a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 20:34:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f898041019be2f50e9230104ca83226931091a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:17:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad376e89e887445d885cf9407118825f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2be7cbff2c453cb74b3912747c096d : SUCCESS in 3m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/490425f0545d4790842cc26cf0b369da : SUCCESS in 1h 00m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b0be308b93b347a7a4e2b2fdc6bb69a6 : FAILURE in 1h 40m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e1c1395453244c478e4eeac81c6569e3 : SUCCESS in 1h 02m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6b95d8a364f9450aa1f0c9afe59eb4be : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/62517c5b39fe4c54a25bcd2cc6b22e19 : FAILURE in 18m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/73e2699ae82148569da1b39f8b6eaf9a : SUCCESS in 1h 12m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6cc13628b147409c832d11149bfd6d11 : SUCCESS in 24m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/788074d613854be9b701de6284685472 : SUCCESS in 49m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3604246a68ac4fed93d01fbfb2a971a4 : SUCCESS in 57m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8260802679da46f2a53a4a465a7d4bda : SUCCESS in 1h 15m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c33557edfc1c43758520b14949ff3403 : FAILURE in 21m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a9141b44af4444168b382574d9af9258 : SUCCESS in 21m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9718335154804dfabdc6fe74fdffddc3 : SUCCESS in 42m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6801c06b9890464f9874852fc480efdf : FAILURE in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7f5d842582439ee574ca6a5e6983d4932a45acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 21:26:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"121a06a0fdb175eced862af491c6733b878e2cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 21:34:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec150ed1b0ffae43ae574026848d501462650a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:13:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb83ab48876f4969b0f3bd517fae4331\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef46c036cc4e492098d58c8a7f8e78c0 : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a398a73f2e644f60a5dc1d3e70fec215 : SUCCESS in 1h 19m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/98effcc658004b6894996f6a775eaaee : FAILURE in 2h 23m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/221e22bb9a7f4d0881d65077279fd63f : SUCCESS in 1h 16m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/43f9b44da52540da89f24a97d9e0e4ad : SUCCESS in 1h 03m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/69dda0df86e6452b91581936b4e8eec8 : SUCCESS in 1h 31m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a69e4828e690408397c8de3a89a655d1 : SUCCESS in 55m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/86ce81bfe9a649deb29956e9b13c73ec : SUCCESS in 25m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f1faaa3328714678b2237d72e9220dee : SUCCESS in 25m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c3e78c7f00154cfe81bd245516570e66 : SUCCESS in 1h 28m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2c09f1044f314866b44b0448853cda6a : SUCCESS in 49m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c0d4114a0876429086969a1ceb0f80d1 : SUCCESS in 22m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3f7b716926c54d819a8f72457b26d497 : SUCCESS in 40m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/50531f2ae9a849eeb90348f82beb3601 : SUCCESS in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e7c9989c2688121334885863de189cb0cd5e85a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-29 03:51:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/872048/3/upstream-check/manila-tempest-plugin-ontap-dhss/070be64/ : FAILURE in 1h 37m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/872048/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/1fdfbc8/ : FAILURE in 1h 14m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e8c3115e6d32b0914e0739021ab317be2aac1ba9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-01 03:15:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d42a3c31b6b7fb0683daba04ef629f5050a9e323","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 15:51:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd2cace2ce3e8d86d60c64d1d5a47b30d0740fe1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 15:55:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e78fa9d68685517b3f56c688bc9ec6d0849b81b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 15:55:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"90dd7801fb92e580b20a0ce4283b245426a78340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:55:56.000000000","message":"Patch Set 4: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":4},{"id":"fec6cd511695cf48e29f6e3ad2135e64ab49d16b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 15:56:17.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b7677606fcb3253064fe62c44fb557b327dd0a47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:56:40.000000000","message":"Patch Set 5: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":5},{"id":"9ace81331a29cf1db69df8410fe9b26f0e0f3ca8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 15:56:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fecc7c195fa594ae592b0f303d4046aa797c2bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 16:41:37.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6adb82bece4fe86c05f358f6abc9278661b2536c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:42:03.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  expected dict for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-lvm-fips-wallaby\n      description: |\n        Test the LVM driver (DHSS\u003dFalse) in a FIPS-enabled environment\n      parent: manila-tempest-plugin-lvm-fips-base\n      vars: *manila-tempest-plugin-pinned-version\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-wallaby-jobs.yaml@master\", line 104, column 3","accounts_in_message":[],"_revision_number":6},{"id":"6801504ec0bad9f85be3d7888b5a0a28684d8e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 16:43:46.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7ca2deb367bd8e9ed48ffdd371421127bd0b3854","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 16:44:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58cbe5a4f1ec15fefc698dbd3fce8e5e13b2b518","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 16:55:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"23d8f0ee34a6962a84bda0231543b1ebb027f634","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-01 17:19:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f599393cda2caa2de48d19f8f306a6101953d913","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-01 17:54:45.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/872048/7/check/DellEMC_Unity_Manila_dhss_true/b75af6f/DellEMC_Unity_Manila_dhss_true/1166 : FAILURE in 46m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/872048/7/check/DellEMC_Unity_Manila_dhss_false/b9ddb28/DellEMC_Unity_Manila_dhss_false/1265 : SUCCESS in 1h 10m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"af5d0c302d576f1810a82e16c5d37923e3b1c3c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:42:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35e405502d84000817a3a0fd4e12420\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8386783f8f07409c89dafbfc9e1f563d : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6cbcb8a6eb7948e7ba4248b637acd29a : SUCCESS in 1h 06m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/425c9b25ccf748cfad28cef400cfbb31 : FAILURE in 2h 37m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/091adef2fa434b00820b892b345b0d7a : SUCCESS in 1h 22m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aa556139fedb4b298eadf8fd6b01e091 : SUCCESS in 1h 15m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6db9094c057e4c3b9270f319120a54a2 : SUCCESS in 1h 37m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2194578240074a039afcd729d3e04a2f : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e5ab90f469144f280b8827bafa0e019 : SUCCESS in 34m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ec0ce61171044f397110d656893ca58 : SUCCESS in 39m 55s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/41f702e797374a77b6c930de3617b595 : SUCCESS in 1h 22m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/faa1d4095f634197a3fe488a32859ff2 : SUCCESS in 1h 18m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cecfd2cf7afe4e70a880c0f6d5d20bc0 : SUCCESS in 29m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4a58b8b60f7a4f5493a4eb8fe6c230b5 : SUCCESS in 55m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4ef254920f824287ad57c2f9b46aa6ef : SUCCESS in 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1c0f0472870cf267c6fd4cba2cfe268a38de48b9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-01 22:54:02.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/872048/7/check/DellEMC_Unity_Manila_dhss_true/3cca843/DellEMC_Unity_Manila_dhss_true/1167 : FAILURE in 50m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/872048/7/check/DellEMC_Unity_Manila_dhss_false/069f39e/DellEMC_Unity_Manila_dhss_false/1266 : SUCCESS in 1h 11m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"5c68ae6f3afc7710049762cde0ce8463a4600242","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-02 00:52:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/872048/7/upstream-check/manila-tempest-plugin-ontap-dhss/f38fd39/ : TIMED_OUT in 3h 09m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/872048/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/4312922/ : FAILURE in 1h 12m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"a9e236db4abe8a5f0f25d91a79f61eb73c7319c9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-02 14:06:56.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7c309976f2f418c815cd37bb3f060717370c1f4d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-07 23:38:43.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3eaac98e3d412e2047e47429e44b88f3c73f2dcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 23:39:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"63e11eeaf8fbe425c7b2023dee4720da33a25b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:02:30.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c6d086fa2a4c9f9507170e587696c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5cca5e2fcb4feeada3418e030911e3 : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8488c94d8b964b419ee136a7e923dfa5 : SUCCESS in 1h 18m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/648ccd2d636341c5afa397cd96ef4808 : SUCCESS in 20m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71786f398b3648a0817b30a6a7d0eafe : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":7},{"id":"d35232a363827f456c8266c44d2fdf87277c827b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:02:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"667bcbe7a4803b1a6077a9b1bef1209eb5271831","revisions":{"2cc080effb52326aeba669825a03b55f9f76c27b":{"kind":"REWORK","_number":1,"created":"2023-01-27 20:34:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/872048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/1"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.9.0\n(for ddt version compatible with stable/wallaby constraints)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.9.0/requirements.txt#L7\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.9.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cc080effb52326aeba669825a03b55f9f76c27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cc080effb52326aeba669825a03b55f9f76c27b"}]},"branch":"refs/heads/master"},"d288e225f877f1b8e3b54b87b6777f1a5f2512a3":{"kind":"REWORK","_number":2,"created":"2023-01-28 21:26:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/872048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/2"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 21:26:13.000000000","tz":-360},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.9.0\n(for ddt version compatible with stable/wallaby constraints)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.9.0/requirements.txt#L7\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.9.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d288e225f877f1b8e3b54b87b6777f1a5f2512a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d288e225f877f1b8e3b54b87b6777f1a5f2512a3"}]},"branch":"refs/heads/master"},"b0cf069145b4c9709d2ba16544267bf3c3ea482a":{"kind":"REWORK","_number":3,"created":"2023-01-28 21:34:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/872048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/3"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 21:34:01.000000000","tz":-360},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.8.0\n(for ddt version compatible with stable/wallaby constraints and tempest\nversion)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.8.0/requirements.txt\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.8.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0cf069145b4c9709d2ba16544267bf3c3ea482a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0cf069145b4c9709d2ba16544267bf3c3ea482a"}]},"branch":"refs/heads/master"},"a595e6d03b92088710388a89783877e61fa45918":{"kind":"REWORK","_number":4,"created":"2023-02-01 15:55:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/48/872048/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/4"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-01 15:55:05.000000000","tz":0},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.8.0\n(for ddt version compatible with stable/wallaby constraints and tempest\nversion)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.8.0/requirements.txt\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.8.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a595e6d03b92088710388a89783877e61fa45918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a595e6d03b92088710388a89783877e61fa45918"}]},"branch":"refs/heads/master"},"c51c2398d1b091389ca80b639c7880082d9d07e2":{"kind":"REWORK","_number":5,"created":"2023-02-01 15:56:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/48/872048/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/5"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-01 15:56:17.000000000","tz":0},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.8.0\n(for ddt version compatible with stable/wallaby constraints and tempest\nversion)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.8.0/requirements.txt\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.8.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c51c2398d1b091389ca80b639c7880082d9d07e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c51c2398d1b091389ca80b639c7880082d9d07e2"}]},"branch":"refs/heads/master"},"8ab639a36c4f0c038210a002f7e62f973ab295ea":{"kind":"REWORK","_number":6,"created":"2023-02-01 16:41:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/48/872048/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/6"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-01 16:41:37.000000000","tz":0},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.8.0\n(for ddt version compatible with stable/wallaby constraints and tempest\nversion)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.8.0/requirements.txt\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.8.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ab639a36c4f0c038210a002f7e62f973ab295ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ab639a36c4f0c038210a002f7e62f973ab295ea"}]},"branch":"refs/heads/master"},"667bcbe7a4803b1a6077a9b1bef1209eb5271831":{"kind":"REWORK","_number":7,"created":"2023-02-01 16:43:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/48/872048/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/872048/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/872048/7"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:29:36.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-01 16:43:46.000000000","tz":0},"subject":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin","message":"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\n\nAs stable/wallaby is in EM phase, and currently broken\nwith the Tempest master. We are pinning tempest there (depends-on).\n\nWe are using the tempest 29.0.0 on stable/wallaby (depends-on) and\ncompatible tag for tempest 29.0.0 is manila-tempest-plugin 1.8.0\n(for ddt version compatible with stable/wallaby constraints and tempest\nversion)\n- https://github.com/openstack/manila-tempest-plugin/blob/1.8.0/requirements.txt\n- https://github.com/openstack/requirements/blob/stable/wallaby/upper-constraints.txt#L80\n\nOtherwise it fail for ddt incompatible version\n- https://zuul.opendev.org/t/openstack/build/4dad3a22fe5d485a96722cb6d68da63d\n\nPreparing the new set of jobs with manila-tempest-plugin pin to 1.8.0\nwhich will be used to test the stable/wallaby.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/871864\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: Ibf8c885529098417e8ff9b417bf172218f8d684a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/667bcbe7a4803b1a6077a9b1bef1209eb5271831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/667bcbe7a4803b1a6077a9b1bef1209eb5271831"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
