)]}'
{"id":"openstack%2Fmanila-tempest-plugin~872260","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I23f592aa0616dd000c91366056b19f73d4a92fa5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2004230","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-20 20:18:28.000000000","reason":"Change was submitted"},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2023-10-20 20:18:28.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-20 20:18:28.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-10-20 20:18:28.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-10-20 20:18:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23f592aa0616dd000c91366056b19f73d4a92fa5","subject":"Change the expected status response from Forbidden to NotFound","status":"MERGED","created":"2023-01-31 06:57:10.000000000","updated":"2023-10-20 21:18:50.000000000","submitted":"2023-10-20 20:18:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872260-bug/2004230","meta_rev_id":"73c6543280742cb4223bab9580dc7ce877dc8cde","_number":872260,"virtual_id_number":872260,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-20 20:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2023-10-20 17:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2023-10-20 17:47:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 10:10:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 19:03:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-02 20:57:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-09-14 08:11:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-22 07:19:33.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-25 23:59:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-17 14:41:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-10-17 18:28:42.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2023-10-17 22:58:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-10-18 02:11:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"5b6b3a306d07b038dd7e250c0f477c6804c805e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-31 06:57:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba1a7572a73381bdcb648e4f2557e5ff9862f505","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-31 07:39:10.000000000","message":"Topic expected_status removed","accounts_in_message":[],"_revision_number":1},{"id":"21fdb6a81740b5ed732f6ae05aec11dc11321a31","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-31 07:39:29.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"97c842800d5077f2d4e07887bc1a3ad1320ad145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:10:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a12bc93ea134950a68a66799fc2488a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6f4a35bd8443069580981f013fe463 : SUCCESS in 3m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e16d85296c3433a816643bd8dc030bb : SUCCESS in 1h 16m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e9c82e9e9e7642599c51a3380867f4af : TIMED_OUT in 3h 03m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/18f4d85187a9414aa759582dc5c4ec76 : SUCCESS in 1h 42m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1841557f09da45abb88636068e11cb47 : SUCCESS in 1h 30m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f7efd33247624edebeadfed8ad33a111 : FAILURE in 1h 53m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b3525211eced4d9495ee5fe982a3373b : SUCCESS in 1h 23m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aff8f2562db9469ea0f4be791042d213 : SUCCESS in 36m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/826430d141804783956762aa0d8e7519 : SUCCESS in 40m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d273beb622aa4e02bb04f320a4d359cb : SUCCESS in 1h 28m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4c92633125d14b10bc79d0d67df13323 : SUCCESS in 1h 20m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/51a053464ac64c2f9ac5f1f01103d164 : SUCCESS in 1h 39m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e4c4788dc9df4394af7fc8594cbfd044 : FAILURE in 33m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3cfd94e350184423aef7567e3be81b6a : SUCCESS in 1h 00m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6efa848880954d6f89f41bb50dc2261e : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64fc9cc978fb96af8abb55fb9f1146aad7ccd29a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-31 19:03:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/1/check/DellEMC_Unity_Manila_dhss_true/5b0c4a3/DellEMC_Unity_Manila_dhss_true/None : NOT_REGISTERED\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/1/check/DellEMC_Unity_Manila_dhss_false/a655e5b/DellEMC_Unity_Manila_dhss_false/1256 : SUCCESS in 1h 02m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d37e23d552494f882314336ecd2a52ed4dad54b9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-02 20:57:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f92bc7e370604d6b0f31feb09f2b2deab7a72432","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-02 21:11:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db10df11919769ceb111c9b28d65296ad43b41c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-02 21:15:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57e4badbe22e5f91f336e3572facf1d972b58e6b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-02 22:20:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/2/check/DellEMC_Unity_Manila_dhss_true/7b8d119/DellEMC_Unity_Manila_dhss_true/1175 : FAILURE in 1h 00m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/2/check/DellEMC_Unity_Manila_dhss_false/0ca4726/DellEMC_Unity_Manila_dhss_false/1274 : FAILURE in 1h 04m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"71cd69b781b11947e5ec2b12b55fce63f8dc51f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 06:35:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff8b67239e9d4c129bd17ab28a14c266\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3827a9419341b9bde044145a84ddae : SUCCESS in 3m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee5ea2ac30ae48c8a550474d2227e28c : SUCCESS in 1h 21m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/db5593d913834e3e8c280354bc4038a0 : SUCCESS in 2h 03m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7c0ff1078ee8472287499c1a97bff51a : SUCCESS in 1h 15m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a4bf126a1fd54d8aa0c2ba7d1cd09b09 : SUCCESS in 2h 00m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/91a7113ab4a54072a76054bc7eab9449 : RETRY_LIMIT in 3h 01m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/afbc6e639f674e96a648f05160c802eb : SUCCESS in 1h 13m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/af7b1e2b95254e61bb8072b9a63f24b9 : SUCCESS in 58m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d6288a4c0174e9f9b79e5f9b9a22cf4 : SUCCESS in 39m 22s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9819fce7a8eb4220977ad6dd8a2da90c : SUCCESS in 1h 19m 18s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b8d224f059e847029b2e6c5238d199b6 : FAILURE in 1h 10m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e4b883741d4a443487647332553a95c9 : FAILURE in 37m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0fdfd76f3d4943a3a76ea2e550d8d4d0 : FAILURE in 29m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dde707d0f2324f2087c62fb03aca6a58 : SUCCESS in 53m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7a3d8164360d4682973c3b7eeb2593b4 : SUCCESS in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54297b853219aa2fd17edcfc9867c586bb3f71d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-08 10:20:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db0d32ce6321d8aeb8407c9cf15efbb8a0edcd73","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 11:19:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/3/check/DellEMC_Unity_Manila_dhss_true/1e69fda/DellEMC_Unity_Manila_dhss_true/1200 : FAILURE in 58m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/3/check/DellEMC_Unity_Manila_dhss_false/f7e7d72/DellEMC_Unity_Manila_dhss_false/1299 : SUCCESS in 54m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b89091a490c97cf5da3c7db0c45b4ab80e5d547a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:20:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28963447438b42c08e5c8a75cf8694df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b18d2f029d84e78a1663e56e9834967 : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/14671b405e3a4932b132958ea50cc5c6 : SUCCESS in 1h 15m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/648ed61c0f7849f3bfcb7fa52039dd13 : SUCCESS in 1h 54m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/71d2b42af12d4670ab9a5e9c9764e4e2 : SUCCESS in 1h 19m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/00aecad8f77f445b9e9fe93f392821ca : SUCCESS in 1h 04m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/13e0a7affe0f4ed3abde294c474544a6 : FAILURE in 1h 27m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0cce039573e1445ba79385de7feeccba : SUCCESS in 1h 07m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5223bc0de1704f8ba3e100f82665cd9e : SUCCESS in 29m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2ecc0393befd4097967bd30b57a8c956 : SUCCESS in 30m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b5d30f1571364b91871e24cf76daedfd : SUCCESS in 1h 15m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/84cfc7a5fdb0475bb84116f049c6bb49 : SUCCESS in 1h 37m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5e676b093d83465ab5fd207310f3a9ef : FAILURE in 25m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d857063682b6488f9a2460db5186367d : SUCCESS in 47m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/952c4c5cf88e481e89eb84031ee2ac48 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41fe06c4ae78e984bfc862c7529a328c2c3b0eab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-23 10:58:45.000000000","message":"Topic expected_status removed","accounts_in_message":[],"_revision_number":3},{"id":"d398d2828dc65d36a89f750a011c8d43cc76bd1a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-23 10:58:52.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":3},{"id":"e8acf938113e0c2e9d5e90a960f22cfaf8f42b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-23 11:08:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"98576c9b57bf9286d1749d7176e0339b4df2b60e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-23 12:00:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/4/check/DellEMC_Unity_Manila_dhss_true/ce7b562/DellEMC_Unity_Manila_dhss_true/1306 : FAILURE in 51m 01s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/4/check/DellEMC_Unity_Manila_dhss_false/600a886/DellEMC_Unity_Manila_dhss_false/1406 : FAILURE in 41m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"15768e5a9d7cfd62417084bb4db783c86e4475c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 13:08:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6660ee54ded8466eb66308c0ca22ca48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e67e8b8f49a4b589174e628655324c7 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8941e14959974658b49f240307e7e9d8 : SUCCESS in 1h 01m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a2e8709da5bb4f2e807adde9c4a6de44 : SUCCESS in 1h 58m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7dcaecf409094a6085e6824e9b1068ed : SUCCESS in 1h 28m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/79c312d282e043b987d1c84f854f61d6 : FAILURE in 12m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/893be16866de416fadb0a3097a975109 : FAILURE in 11m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4979091897634de09ada90c6ace7ccce : SUCCESS in 1h 00m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/41ae0b2a69744a699225bbd9edd825c2 : SUCCESS in 25m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2bb5d6b6cfc44b48a04d9d50cc61e8e8 : SUCCESS in 44m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5366179a0caf476da9eba0996a158175 : SUCCESS in 1h 25m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/448e590e839340baa072a12e610413f5 : SUCCESS in 1h 18m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5078fabb82ac40a082fcbdd108c274cd : FAILURE in 21m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1564efcf112d4543adbe9eade988d508 : SUCCESS in 51m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9520ad37004f42eb8a85223ef322df86 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2c99ab6590922af88f668f3580291e1f34e9611","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 07:05:08.000000000","message":"Topic expected_status removed","accounts_in_message":[],"_revision_number":4},{"id":"9f71bcbb4acf384a3f5df2edbf5f91e7cb9bffb9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-30 07:05:23.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"ffc25079c86350abdd838eef2c0609189d969bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 20:04:46.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ce8d79c3080d38449f1b2fa5a1069a9612f50241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 20:05:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"899b5cced67cad3e7b34899d481753beb2801c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 22:45:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c18a37b2c0824c56a09552fb211a1a42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640e8c19c0444925805428f3ecdf1618 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/af4fb33e83b645a5b6d23631326a51be : SUCCESS in 1h 17m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8ca1eeadcf2c414993e109a4476540de : FAILURE in 2h 30m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47b8ce7c277f4c6aba2b28f9610748d0 : SUCCESS in 47m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f33552b818dc438a818ff3dc147a36a9 : SUCCESS in 52m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d6d0316a13d1420f9809c7a607e370ed : FAILURE in 12m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d5f93075c04c4400acefba3c6448bb20 : SUCCESS in 46m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a01bacd279a2491a9fb5e2947e51f9f1 : SUCCESS in 24m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ca8f2b651935411ea180ec707a986465 : SUCCESS in 29m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2f0d1afb4d0b49ac8c687120ca9d4fc4 : SUCCESS in 1h 00m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/317ab963173041e597d4b4cada06254b : SUCCESS in 1h 13m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f1d1ad76298b42a19e92133862b1ab51 : FAILURE in 33m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f6851230a00d46479b23670669f2a65b : SUCCESS in 56m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0cbe1528edf447eca9fb5aa4c2fae716 : SUCCESS in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ebb4ca5052e57d5a77a7d1d60ce196bbba4a22f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-13 03:06:55.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/6/check/DellEMC_Unity_Manila_dhss_true/d945ef8/DellEMC_Unity_Manila_dhss_true/112 : SUCCESS in 3h 34m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/6/check/DellEMC_Unity_Manila_dhss_false/5b585cb/DellEMC_Unity_Manila_dhss_false/122 : SUCCESS in 1h 15m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6a9c32ebab37148b55e5031f8b0cddee8c7f0770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-14 07:36:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7f10f6b3e0f09099fe1ac8ebb75e9bb3749bb04f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-14 08:11:49.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/7/check/DellEMC_PowerFlex_Manila_dhss_false/ee123e3/DellEMC_PowerFlex_Manila_dhss_false/375 : SUCCESS in 31m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"eeae771518510752e05502c3f889268d4f95aa12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:11:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c853e400c6e446698e74a00d58de06d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2c8ac461394cb38b91f2f8ab3cc51d : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32a56a8dd6bc4e9b977da19e5f74e187 : SUCCESS in 41m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/de5a983512a24c418fea2b62ae2ff80e : FAILURE in 39m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ae50797212bc4a199e1a1a1370182c4f : FAILURE in 27m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/12449e1425fd4af4ac0a0fcde53de1e9 : SUCCESS in 1h 12m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/060f9541bb334b9cb2e504bfcfdf4a24 : SUCCESS in 1h 34m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d972ce8befc74f3596b871d15b41109a : FAILURE in 1h 11m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46a32c01ff6b45f4b965f5c833a36ee7 : SUCCESS in 28m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0d44aba80eaf4542b267933b7e710804 : SUCCESS in 47m 35s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/661818cf6a1f49c8b09cf2121c6b2381 : SUCCESS in 48m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8d85572a1d794742b8f3e2ecd1850c14 : SUCCESS in 34m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4fc0f56668544f5191a4dda6eb1af101 : FAILURE in 23m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d42e46cb3ba4e2689036c437297073f : SUCCESS in 51m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/397a9c0b99fc4883a68942099fe69456 : FAILURE in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a07d38e16b16afc735b8586082fbf89f05f6b299","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-14 21:07:17.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/7/check/DellEMC_Unity_Manila_dhss_true/13ea7a6/DellEMC_Unity_Manila_dhss_true/589 : SUCCESS in 2h 59m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/7/check/DellEMC_Unity_Manila_dhss_false/b62a40f/DellEMC_Unity_Manila_dhss_false/615 : SUCCESS in 1h 08m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"602a23bfb470788b24dfa497d26a4645deaceb33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 03:47:22.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"369d0b173dba9962605cc3fd042760891881bfdb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 04:27:04.000000000","message":"Topic secure-rbac removed","accounts_in_message":[],"_revision_number":8},{"id":"4a569d7138572403f80480ae5766d0f99617d621","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 04:27:15.000000000","message":"Topic set to bug/2004230","accounts_in_message":[],"_revision_number":8},{"id":"44d947e5d9ee762f0d794f79603a504af33185f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 04:37:58.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"b5526180811fa39bd71d8e49c642512fc224db3a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 05:39:02.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/9/check/DellEMC_PowerFlex_Manila_dhss_false/0d020d1/DellEMC_PowerFlex_Manila_dhss_false/436 : FAILURE in 29m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"ba0253a9686cedb2476bfad56e5b4179c6a5fd1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 06:47:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5be42418157949b18c1fe2a6541c2ccd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10757ef8c421462dbd6043eb421bf958 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99726ae8a3ce4b88869b1b0bc0ab226f : SUCCESS in 32m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/28c20d93057c427d8e12b909f871a9b5 : SUCCESS in 2h 03m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/293d3385c9664003a8eb6f38a1ad5c45 : SUCCESS in 1h 14m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5996da68ae7c41ad99500531a7164f1a : FAILURE in 43m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7508b3340c364091bc3b133c7a945200 : FAILURE in 1h 10m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c9d9d0db1345425b861ee1369049ecd7 : FAILURE in 1h 09m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e32f24a87d524c0fa9ea57cc64c008a8 : FAILURE in 25m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d3631eebeeeb48e1a811cb7f760c7e7e : FAILURE in 48m 20s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2771f3d6b3c94aaf983f98332a42fa9b : SUCCESS in 33m 59s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e85bea022a1c411da78c4cf86a64deee : SUCCESS in 39m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e3e89c23e1804d8a9d0217d2930dd453 : FAILURE in 28m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b34f1e91d3704ac6a0008fc864212a2b : FAILURE in 47m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1410f2ab11194fc78c224ea417299a28 : FAILURE in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"490ee5490a752bfb1dc3fd368ecd0c87c9b688b1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-22 02:25:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"089f233673f9dd520b940c17d080eaf4a022198b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 04:39:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415e80f481044705a30fdd02c0c5cd3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370146c1eb9842dcb386a1c86a22cf22 : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6aaf87ce7c6b46bf9929a26cb1400334 : SUCCESS in 44m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c18b84a62e744a891627274e8985310 : SUCCESS in 1h 58m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e9d6160b00644d9aa447dbd9e81df212 : SUCCESS in 1h 01m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bd8f0dd96edb44619759c289ef599137 : SUCCESS in 1h 06m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c7bb71e649ae44d9adebc41e78f7c7f6 : SUCCESS in 1h 31m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4120785fc1f74155b70156ad54a6aa31 : FAILURE in 48m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5c9d58c1510c4ab29fa66d6e4f3b00ba : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4925fafc300c45509823b73bb6d383cb : SUCCESS in 40m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/119721c376b643d684dfa1addb76a7ce : SUCCESS in 38m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/00c13436286e45f7ad63124280e656d4 : SUCCESS in 35m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5016dbc2d0904da1873b5ab60da056b9 : SUCCESS in 45m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/abed9f7accaf4ecf88cebe5eed6614c2 : FAILURE in 55m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d06a77aa558943cb8d1468d574a01209 : FAILURE in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"58066ea2b12bbd267c6164b417b02ebaa83df971","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-22 07:19:33.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/872260/9/upstream-check/manila-tempest-plugin-ontap-dhss/5f239e1/ : RETRY_LIMIT in 8m 41s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/872260/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/a975682/ : SUCCESS in 2h 37m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"b15ec066604d34e2b5ef261381501f801937ee56","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-25 23:32:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5aa4b39c5b52b272287120b935b2b0cdf439cc51","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-25 23:59:30.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"c1d5ff37ce1c80708a7c91743bace469517de0df","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-25 23:59:30.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"993d04b09cb42b97ed763b0fefc5d720f8562f79","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-25 23:59:31.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"df103d1689d5d0682a2762b52b69425cf43a8dff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-17 14:41:59.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b3d6ca3bf148c2faf81c6a4361d3a9a73b372b9b","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-10-17 18:28:42.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1dc3d307fdd4e95314f6792bce5cddeac57cb4b2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-18 14:15:05.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d44947fa35195b3002c394408507a307cf288ecf","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-10-18 14:16:28.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ec1a67dae44f971742af48f12b497545814e070c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-19 04:15:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3a3c972bce1214dba8292192c64258578fb208d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-20 17:47:35.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":10},{"id":"eea8328c0a4d5a261c591de49fb9aecc1d736331","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-20 17:47:57.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"32acb36b32750e17e72ea2308430a4cbc7a798ab","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-20 18:19:28.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_PowerFlex_Manila_dhss_false/2f6cfa3/DellEMC_PowerFlex_Manila_dhss_false/585 : SUCCESS in 31m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"695c686c97dbab8ad8f1e03637c8987c43814672","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-20 19:22:32.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_Unity_Manila_dhss_true/226ee93/DellEMC_Unity_Manila_dhss_true/687 : FAILURE in 1h 34m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_Unity_Manila_dhss_false/f1998d1/DellEMC_Unity_Manila_dhss_false/724 : SUCCESS in 1h 04m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"379dde9201af5eff22f8f375591f3f2aeee5569c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 19:29:41.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f28c7bd96714742b176c470327429cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8668485d6c1844719ba094e7b6c77907 : SUCCESS in 3m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8022b880f21a44a28fb416268dd47541 : SUCCESS in 32m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/63b87b6058734b03866936f12c3019c8 : SUCCESS in 1h 40m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/81f56fba74774de889587179492965d5 : SUCCESS in 59m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d661b416b945402ba6f9b9374b15b801 : SUCCESS in 1h 09m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/74f012ea6459455ba714c60e8e3cf031 : SUCCESS in 1h 08m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/52db55eeba8a4c8492b6cde99d0020a5 : SUCCESS in 1h 18m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/afbd816b31964f50a8647df5e38c9b7f : SUCCESS in 41m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7b1663dd8e92488db65738ead22480e0 : SUCCESS in 38m 27s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/39a02811e26943f49da0791010d4e1e8 : SUCCESS in 46m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/00d6ed0f13714a099c9d582e2ada6264 : SUCCESS in 44m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9afac5f076c34a739aef4453237a847b : SUCCESS in 37m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d36fcab5c0bb46aab5c4645ae6ad6f48 : FAILURE in 56m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/386edfde60d741798bcd1ac2ca2e6ac3 : FAILURE in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e0cd83275ac49c65d4d3d8e2e49c1223c470ed28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 19:30:11.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4b8aae59a59646fcf7a2f023d83d32356a71b827","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-20 19:44:16.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_PowerStore_Manila_dhss_false/85e1fcc/DellEMC_PowerStore_Manila_dhss_false/252 : FAILURE in 14m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"373ec34faf25e2ae41da2672dd18baf79469c0c6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-20 20:03:37.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_PowerFlex_Manila_dhss_false/0692df5/DellEMC_PowerFlex_Manila_dhss_false/586 : SUCCESS in 33m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"99ec1e5c736edc05fcca5a89101db80de8151520","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:18:27.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0a6ff4e2dd43dbb779ca2fde513819\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3f3a55e0d640d188c6a39db43023ef : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5163202cf2a34f269d9e85e08a5cc1db : SUCCESS in 43m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4c37046eae0245159950397385d67c8a : SUCCESS in 39m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff2f0f1de9b54e6980760ebf4250dddd : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":10},{"id":"83ab7269e9520a92c76c5dea74cad3f3fcba65b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:18:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"0a3fef354ee911bc08a3e78f18b875093a78373d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-20 21:01:06.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_Unity_Manila_dhss_true/d652cfb/DellEMC_Unity_Manila_dhss_true/688 : FAILURE in 1h 31m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/872260/10/check/DellEMC_Unity_Manila_dhss_false/a4de638/DellEMC_Unity_Manila_dhss_false/725 : SUCCESS in 1h 04m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"73c6543280742cb4223bab9580dc7ce877dc8cde","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-20 21:18:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/872260/10/upstream-check/manila-tempest-plugin-ontap-dhss/02aa00c/ : FAILURE in 1h 32m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/872260/10/upstream-check/manila-tempest-plugin-ontap-no-dhss/365dbd0/ : SUCCESS in 1h 47m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2c06e81bfdbe72297b9fbc0c8d571c0ea51ab443","revisions":{"6ca2b0c6941f6bc56d77403b4b28311caae1be35":{"kind":"REWORK","_number":1,"created":"2023-01-31 06:57:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/872260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/1"}}},"commit":{"parents":[{"commit":"2521fbf86a2aa001e5c82dc9fd36d39646b37a3b","subject":"Add tests replica create with \u0027share-network\u0027 option.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2521fbf86a2aa001e5c82dc9fd36d39646b37a3b"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ca2b0c6941f6bc56d77403b4b28311caae1be35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ca2b0c6941f6bc56d77403b4b28311caae1be35"}]},"branch":"refs/heads/master"},"588cedfe89cff832a7fe8af3d622d867de8357d4":{"kind":"REWORK","_number":2,"created":"2023-02-02 21:15:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/872260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/2"}}},"commit":{"parents":[{"commit":"2521fbf86a2aa001e5c82dc9fd36d39646b37a3b","subject":"Add tests replica create with \u0027share-network\u0027 option.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2521fbf86a2aa001e5c82dc9fd36d39646b37a3b"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-02 21:14:38.000000000","tz":120},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/588cedfe89cff832a7fe8af3d622d867de8357d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/588cedfe89cff832a7fe8af3d622d867de8357d4"}]},"branch":"refs/heads/master"},"1461a5e528a4e6fe1c734b89dad5f637fa7e6a43":{"kind":"REWORK","_number":3,"created":"2023-02-08 10:20:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/872260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/3"}}},"commit":{"parents":[{"commit":"2521fbf86a2aa001e5c82dc9fd36d39646b37a3b","subject":"Add tests replica create with \u0027share-network\u0027 option.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2521fbf86a2aa001e5c82dc9fd36d39646b37a3b"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-08 10:20:32.000000000","tz":120},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1461a5e528a4e6fe1c734b89dad5f637fa7e6a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1461a5e528a4e6fe1c734b89dad5f637fa7e6a43"}]},"branch":"refs/heads/master"},"50ab048025fbe918d6352d4269b3750fc79b3e84":{"kind":"REWORK","_number":4,"created":"2023-02-23 11:08:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/872260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/4"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-23 11:08:43.000000000","tz":120},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50ab048025fbe918d6352d4269b3750fc79b3e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50ab048025fbe918d6352d4269b3750fc79b3e84"}]},"branch":"refs/heads/master"},"294fd64fa32d5d28353f8674db6d2432a517df9c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-05-12 20:04:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/60/872260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/5"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-12 20:04:46.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/294fd64fa32d5d28353f8674db6d2432a517df9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/294fd64fa32d5d28353f8674db6d2432a517df9c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7cdfbfc98a208e919e777db40e6048401ed94ac6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-12 20:05:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/60/872260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/6"}}},"commit":{"parents":[{"commit":"797aeb1303d6b7f80679ff75c35b14028130ec84","subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/797aeb1303d6b7f80679ff75c35b14028130ec84"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-12 20:05:03.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cdfbfc98a208e919e777db40e6048401ed94ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cdfbfc98a208e919e777db40e6048401ed94ac6"}]},"branch":"refs/heads/master","description":"Rebase"},"6b6238ee1efb721fbcde155e5e8ac2af5760c190":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-09-14 07:36:02.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/872260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/7"}}},"commit":{"parents":[{"commit":"d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328","subject":"Merge \"[RBAC] Add share groups tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2023-09-14 07:36:02.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b6238ee1efb721fbcde155e5e8ac2af5760c190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b6238ee1efb721fbcde155e5e8ac2af5760c190"}]},"branch":"refs/heads/master","description":"Rebase"},"c5b9811e6cf8b6db65423da19e2e0146be9980f2":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-09-21 03:47:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/60/872260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/8"}}},"commit":{"parents":[{"commit":"d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328","subject":"Merge \"[RBAC] Add share groups tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 03:47:22.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nDepends-On: I5b1f1ce517efed000f17b1e0901e183a1913ba9f\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5b9811e6cf8b6db65423da19e2e0146be9980f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5b9811e6cf8b6db65423da19e2e0146be9980f2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0a48eeacb380cbf5ccf355f999bf8545c35ffc20":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-09-21 04:37:58.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/60/872260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/9"}}},"commit":{"parents":[{"commit":"d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328","subject":"Merge \"[RBAC] Add share groups tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:37:58.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nDepends-On: I5b1f1ce517efed000f17b1e0901e183a1913ba9f\nDepends-On: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a48eeacb380cbf5ccf355f999bf8545c35ffc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a48eeacb380cbf5ccf355f999bf8545c35ffc20"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2c06e81bfdbe72297b9fbc0c8d571c0ea51ab443":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-10-20 17:47:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/60/872260/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/872260/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/872260/10"}}},"commit":{"parents":[{"commit":"d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328","subject":"Merge \"[RBAC] Add share groups tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a4d27f49cc1d94df5ff5f4b240c1b0143f5328"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-31 06:04:24.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-20 17:47:35.000000000","tz":0},"subject":"Change the expected status response from Forbidden to NotFound","message":"Change the expected status response from Forbidden to NotFound\n\nIn some share metadata operations performed on a share resource\nin another project by member and reader users, the response\ncode is 403/Forbidden but actually it should be 404/Not Found.\n\nRelated-Bug: #2004230\nChange-Id: I23f592aa0616dd000c91366056b19f73d4a92fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c06e81bfdbe72297b9fbc0c8d571c0ea51ab443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c06e81bfdbe72297b9fbc0c8d571c0ea51ab443"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
