)]}'
{"id":"openstack%2Fmanila-tempest-plugin~872583","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If160c0c987520fc413eefbc477d3051ba7f5e9fb","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-12 20:05:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If160c0c987520fc413eefbc477d3051ba7f5e9fb","subject":"[DNM] testing manila change with tempest change","status":"ABANDONED","created":"2023-02-02 18:03:56.000000000","updated":"2023-05-12 20:05:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5699c5bb887a39453dde1f1298e6f3c23ff05b44","_number":872583,"virtual_id_number":872583,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 19:14:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-02 22:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d4049e66a622eaf983ad627d0322dbf0f08186d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-02 18:03:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"181a83a47eac3e4265991c0ab289eb3500f21632","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-02 18:05:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"bc57e824f8545db36e38f5a0a78ee28f43a98601","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-02 18:05:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d4602d2e6656a5227df65884f8ac16a93022aea4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-02 19:14:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/872583/1/check/DellEMC_Unity_Manila_dhss_true/8f5cce7/DellEMC_Unity_Manila_dhss_true/1173 : FAILURE in 1h 09m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/872583/1/check/DellEMC_Unity_Manila_dhss_false/c7a3cbb/DellEMC_Unity_Manila_dhss_false/1272 : SUCCESS in 58m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8c74a36d84584c462ad32365469dc0e96a996e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 22:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53d95362d88c4b0dad1534df4e74815e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ecf3726cdd94f5d852cbb064715e619 : SUCCESS in 1m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/60665141dc7b450283441d7cf010a308 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08052f0bb5134ed8b901fc9b5e580a6b : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6ef499dac0b846c4b3d0331fc8a320e0 : FAILURE in 1h 40m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dbc67a267ec94229864de78667f27110 : FAILURE in 2h 24m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/41f0edb9fe804a69b93e12e44a47e8e5 : SUCCESS in 1h 11m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/475b3674c0a5411b9373567a72135896 : SUCCESS in 1h 11m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ec068ce86c38444bb3ddc75203b3446b : SUCCESS in 1h 20m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8da356b5dfbd475892b64c8e6c78b1c0 : FAILURE in 1h 03m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2614558d4ebe40bd9c09d7ae0e298e15 : SUCCESS in 40m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d813ca647e43479a8587e29ed523e4ed : SUCCESS in 47m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/759a3aa8a7574419ba3c332814f7a1fd : SUCCESS in 1h 05m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/16aabd09482140bbab4534449529f241 : SUCCESS in 1h 08m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/017ba65d179f4a15a6bae48393a77d84 : FAILURE in 25m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3bb50ce7823d465283ae75fc0597dc2a : FAILURE in 30m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ca337f85e4a4e3095b5908a8a11e112 : SUCCESS in 1h 02m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/184a387b3f78446b87da6f191efeca8f : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14cf908cab69df8d5abc296a8991291d993099aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 19:44:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5699c5bb887a39453dde1f1298e6f3c23ff05b44","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 20:05:35.000000000","message":"Abandoned\n\nParent was updated to depend on the manila change directly..","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8795e20ea712a2425f120e74aea88f1af39873b0","revisions":{"40f57801b098603fcc35d0edb83722ddbb5dd394":{"kind":"REWORK","_number":1,"created":"2023-02-02 18:03:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/83/872583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/872583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/1"}}},"commit":{"parents":[{"commit":"6ca2b0c6941f6bc56d77403b4b28311caae1be35","subject":"Change the expected status response from Forbidden to NotFound","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ca2b0c6941f6bc56d77403b4b28311caae1be35"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-02 18:03:56.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-02 18:03:56.000000000","tz":0},"subject":"[DNM] testing manila change with tempest change","message":"[DNM] testing manila change with tempest change\n\n\nChange-Id: If160c0c987520fc413eefbc477d3051ba7f5e9fb\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40f57801b098603fcc35d0edb83722ddbb5dd394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40f57801b098603fcc35d0edb83722ddbb5dd394"}]},"branch":"refs/heads/master"},"8795e20ea712a2425f120e74aea88f1af39873b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-12 19:44:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/83/872583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/872583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/872583/2"}}},"commit":{"parents":[{"commit":"50ab048025fbe918d6352d4269b3750fc79b3e84","subject":"Change the expected status response from Forbidden to NotFound","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50ab048025fbe918d6352d4269b3750fc79b3e84"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-02-02 18:03:56.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-12 19:44:33.000000000","tz":0},"subject":"[DNM] testing manila change with tempest change","message":"[DNM] testing manila change with tempest change\n\n\nChange-Id: If160c0c987520fc413eefbc477d3051ba7f5e9fb\nDepends-On: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8795e20ea712a2425f120e74aea88f1af39873b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8795e20ea712a2425f120e74aea88f1af39873b0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
