)]}'
{"id":"openstack%2Fmanila-tempest-plugin~872685","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ica4a6ff8f655d372baee73c2daab8517188655aa","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ddt_named_data","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-02-14 20:59:19.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-14 19:19:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-02-07 06:19:25.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"Ica4a6ff8f655d372baee73c2daab8517188655aa","subject":"Revert \"Use ddt.named_data decorator to avoid random test names\"","status":"MERGED","created":"2023-02-06 13:39:38.000000000","updated":"2023-02-14 20:59:19.000000000","submitted":"2023-02-14 20:59:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"revert_of":855813,"submission_id":"872685-ddt_named_data","meta_rev_id":"9e38bca7ab0a9d445119b44a850bd31d6ad8abc0","_number":872685,"virtual_id_number":872685,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 20:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-02-14 19:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-02-14 19:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-09 18:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-02-14 19:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"CC"},{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-02-06 13:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 19:05:08.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"a9e89f4041faf6ba4977c04726fae7ea62d26f1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-06 13:39:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23000f3296ec557368c208503d264faaf4683837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 13:39:59.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 872685,1","accounts_in_message":[],"_revision_number":1},{"id":"4c7fda15ea835ccff224f0ec8b0cb0c8c724261a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-06 13:43:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c11612bdf7788424a1684998ab3c8618fb039e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:29:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce50259d37b457196a8a5e3744aa646\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7749ecc33b72492a938f617a76762576 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0b2548fda7470e97950e54a6bd656f : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f0277451578b476989ea9d66d9efec84 : SUCCESS in 1h 00m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9d6a949eee9e425dbc35bf983a5328b1 : SUCCESS in 1h 44m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/21848283c22c4ff6b1f5ebb87d71db48 : SUCCESS in 1h 04m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/218c8b071d704311b85f920c6fc68943 : SUCCESS in 1h 00m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7533c4e208b64bdaaff9d7359fc7404b : SUCCESS in 1h 18m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/30abbd2b37eb452e8abeb22854b98fc0 : SUCCESS in 55m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d36fbd33cdf24a48a335adf45abecdb3 : SUCCESS in 27m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fff9a5dfcfd44f32a8b45f9b63bfb183 : SUCCESS in 33m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6d7fae30c42349bd89d44571d8abbdc4 : SUCCESS in 1h 04m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a60b12c130ff4ae98ebb3f6cf83e8bc6 : SUCCESS in 1h 10m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/81146094fc0249d2b87722f8dc0516af : FAILURE in 26m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/68703f82bd9748fdb1115528da9feb93 : SUCCESS in 34m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9df8ce63529e4b26b9000340f040e7f2 : SUCCESS in 40m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cd8956f3e807418b966036859d734b17 : SUCCESS in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"053609d78406fa3df5a4b653c21ae3a810038e22","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 18:07:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/872685/2/check/DellEMC_Unity_Manila_dhss_true/b04b224/DellEMC_Unity_Manila_dhss_true/1189 : FAILURE in 1h 06m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/872685/2/check/DellEMC_Unity_Manila_dhss_false/9c12126/DellEMC_Unity_Manila_dhss_false/1288 : SUCCESS in 59m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d459ba72936716821fb1ff86a17867fda894c30b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-07 06:19:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1eda2d50a791602f23406aa1995794a9798e07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:27:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e85d8310e2f42ac90456e937f8b554b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2db170b348464e6481b3da6161333758 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56a60b603e14cff931f9cd864a2a847 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7f74845818ab491381439196deca4086 : SUCCESS in 1h 01m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/05cb7814aa244bf0802bcc65b85d3bee : SUCCESS in 1h 43m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/be4401f17a904cf8924deae6d7e63cfe : SUCCESS in 1h 00m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bbc5c50532c24c3d91da8d2673f7c3a1 : SUCCESS in 1h 29m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d77cf0af99a2411cab191dc902543870 : FAILURE in 1h 38m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/79d38ffd05ea481e8daa49809e444cdd : FAILURE in 59m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6af5b1e02e0345dc83bc9bfada6f0066 : SUCCESS in 35m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/728736a3bdbe434998a42f59123f28eb : SUCCESS in 31m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2c664786cc0e46b2acff02593be0ad6a : SUCCESS in 2h 02m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2aee78d93942461d8b3b953e3cdd6b2a : SUCCESS in 57m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9c7fec698fae408ab968bd33325d2478 : FAILURE in 42m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2167cf11cabf45729238420c17288a4d : SUCCESS in 21m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b2f96162dd1d4ea7be7ee749e8f849ab : SUCCESS in 51m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bfb1b879d39b46248b40ad057d0f6e46 : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b719fccde752baefc716f4b1056d6772876fb42","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-09 15:35:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73c6e741afbf7800975b9206d9b7b59569afee63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 18:04:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dab57ce31ec444c9c860b5a2dced4e9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/725a6725f2b4437e9376523867f989df : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8745a573f87841158035b692b9c77cb7 : SUCCESS in 3m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5713d080f16b41d0a2a0e6e7a601ff0f : SUCCESS in 1h 01m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a9eacf78a004e958ada640b167686e3 : SUCCESS in 2h 25m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ef772b15ca7144cf836e2ede8917c741 : FAILURE in 1h 08m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d405fe016b7d4198b171b27f6176a60d : SUCCESS in 1h 05m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b607801589f4c34a933ff6db3100616 : SUCCESS in 1h 33m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/45c32475ff714016a03395484708c92e : FAILURE in 1h 22m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2c00490091b64df0807c3d20f60f470a : SUCCESS in 31m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c78c09b6a8f64535a4ce31630d7689e1 : SUCCESS in 43m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/53c98478ac92412c84410dff8978373f : SUCCESS in 1h 25m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d8ef7abb3dcc473fbd2a6173081018d0 : SUCCESS in 1h 40m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/628083dd8bae4b15b8284dfc2867993e : SUCCESS in 41m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/23d44cadb859428abed4469929e3e501 : SUCCESS in 47m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/602cde7803284f21a89fe23ea1ffca85 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"390eca232da8376d9abc6ed396c0c3717d849d26","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-09 18:12:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ff4081c1daaa33e23a30e03033b3a3e6d1f2da7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-09 20:11:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/872685/2/upstream-check/manila-tempest-plugin-ontap-dhss/b396759/ : SUCCESS in 1h 38m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/872685/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/40f9f7b/ : RETRY_LIMIT in 8m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"34b4d1bd39d0155e7ebf95e84b3679bd330eb886","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-14 19:05:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8c113f45ecff26cb2d4797fc729d3cffc35be70","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-14 19:19:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72d535c4ca37ec32d780aa87fb76d3069ca7eb69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:19:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3599c70e6c8acef31be7d2c07d5c086d7c2e121b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:59:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1063307805440b09f17cbfbe3d7d312\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef731ed2ddaa4b9eb449ae804a204195 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6fd383d28e4e708ac0dac7e97a6cc5 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c0891b7d7e2446408bfe10bd92cf42ab : SUCCESS in 1h 21m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5f4930fa989f4934895ec91c0c6d3af2 : SUCCESS in 34m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ae33283f235f4812afd14317a615f608 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9e38bca7ab0a9d445119b44a850bd31d6ad8abc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:59:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"add7740f9b461d97e8e5a0a576009236852984af","revisions":{"cf6018f0d1787f2a3c68fab44437089a81e75896":{"kind":"REWORK","_number":1,"created":"2023-02-06 13:39:38.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/85/872685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/872685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/1"}}},"commit":{"parents":[{"commit":"4b26d79ba76792f53f77f3eceaf49b24e0e1578f","subject":"Use ddt.named_data decorator to avoid random test names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b26d79ba76792f53f77f3eceaf49b24e0e1578f"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-02-06 13:39:38.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-02-06 13:39:38.000000000","tz":0},"subject":"Revert \"Use ddt.named_data decorator to avoid random test names\"","message":"Revert \"Use ddt.named_data decorator to avoid random test names\"\n\nThis reverts commit 4b26d79ba76792f53f77f3eceaf49b24e0e1578f.\n\nReason for revert: This can be backwards compatible. We want to revert it for now.\n\nChange-Id: Ica4a6ff8f655d372baee73c2daab8517188655aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf6018f0d1787f2a3c68fab44437089a81e75896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf6018f0d1787f2a3c68fab44437089a81e75896"}]},"branch":"refs/heads/master"},"add7740f9b461d97e8e5a0a576009236852984af":{"kind":"REWORK","_number":2,"created":"2023-02-06 13:43:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/85/872685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/872685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/872685/2"}}},"commit":{"parents":[{"commit":"2521fbf86a2aa001e5c82dc9fd36d39646b37a3b","subject":"Add tests replica create with \u0027share-network\u0027 option.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2521fbf86a2aa001e5c82dc9fd36d39646b37a3b"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-02-06 13:39:38.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-02-06 13:43:14.000000000","tz":-180},"subject":"Revert \"Use ddt.named_data decorator to avoid random test names\"","message":"Revert \"Use ddt.named_data decorator to avoid random test names\"\n\nThis reverts commit 4b26d79ba76792f53f77f3eceaf49b24e0e1578f.\n\nReason for revert: This can be backwards compatible. We want to revert it for now.\n\nChange-Id: Ica4a6ff8f655d372baee73c2daab8517188655aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/add7740f9b461d97e8e5a0a576009236852984af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/add7740f9b461d97e8e5a0a576009236852984af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
