)]}'
{"id":"openstack%2Fmanila-tempest-plugin~873065","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I69fd7a3747986ce287c3a59cdc0f374fdd5b0099","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-02-08 16:37:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69fd7a3747986ce287c3a59cdc0f374fdd5b0099","subject":"[RBAC] Add skip condition for share manage","status":"MERGED","created":"2023-02-08 08:30:13.000000000","updated":"2023-02-08 16:37:45.000000000","submitted":"2023-02-08 16:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873065-secure-rbac","meta_rev_id":"30dfd4d07d930b2b6d9a49f63ba8527cfcc2ee96","_number":873065,"virtual_id_number":873065,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 16:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":2,"date":"2023-02-08 14:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":1,"date":"2023-02-08 14:24:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-08 09:35:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-08 10:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 14:24:39.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"90a9bac279048e81e783447161257941cecf832e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-08 08:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d3db6526669f41c40a82dc2dc2a0ce8cb4a054a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-08 08:30:45.000000000","message":"Topic skip_check_manage removed","accounts_in_message":[],"_revision_number":1},{"id":"c95854eef3bc316f1397264db31aad063ada1263","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-08 08:31:10.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"dde41465085e351ce99b04c6392fd9a3aa8e6827","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 09:35:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/65/873065/1/check/DellEMC_Unity_Manila_dhss_true/5888e94/DellEMC_Unity_Manila_dhss_true/1199 : FAILURE in 1h 05m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/65/873065/1/check/DellEMC_Unity_Manila_dhss_false/2774da4/DellEMC_Unity_Manila_dhss_false/1298 : SUCCESS in 1h 00m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7794cbb5f83c89b1ea87087f0f33937a0a1b51b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ebcf4599b874f3eb027c74a7b0d4710\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77db475ab67e41f78bed9387694e0220 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/27dfa620617b4ee88d75e6d802bc2ab1 : SUCCESS in 1h 20m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b889aec812ec47529e28b8915bbf0169 : SUCCESS in 1h 43m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da9b74c4917c42eaaabffd9ef43a83cc : SUCCESS in 48m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7165f2216315411482efa693e1e3d301 : SUCCESS in 1h 41m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c77f002776754935b170ba1861ccd075 : FAILURE in 1h 31m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/81d0272172e240feb27d69f09cb45a80 : SUCCESS in 1h 23m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47b5899661ff4f53b75f25be34bf30c5 : SUCCESS in 38m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1babfd106ec14d1b8822c8e5f4ca9467 : SUCCESS in 33m 01s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/96fd7faaf40546dda76eb5eaa2ab22bb : SUCCESS in 1h 24m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5b815582001f42ed9a15e8484f7f03cb : SUCCESS in 54m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f9bd6e5ece224d558f317594bafb0a4e : SUCCESS in 24m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/793190a91fb24b8da9a0c0b8820e84b9 : SUCCESS in 1h 03m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ddfba6268b654f8ab08adb4548a0c388 : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f43e5087f69457ccd3e99eca39b5998387f27a5b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 12:27:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/65/873065/1/check/DellEMC_Unity_Manila_dhss_true/1c34e5d/DellEMC_Unity_Manila_dhss_true/1201 : FAILURE in 1h 08m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/65/873065/1/check/DellEMC_Unity_Manila_dhss_false/1a6e65c/DellEMC_Unity_Manila_dhss_false/1300 : SUCCESS in 52m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"696157cc8212ec1a5bc6c3673336a51f6ba9fd9c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-08 14:24:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3428d5b1a30a946cd754d9c425bda9039f07f88b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-08 14:24:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"443b7f619d2478184cb4f39cc1d329c0ebe9eca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 14:25:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"879e8275ab09ee243d70104763e429cfc8b691a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:37:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2f1f9638f764e2ab75db96f92d389fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73e3e66a3384915a4d67904bd49b0e9 : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74df3afc87be4509a7933655a2a593df : SUCCESS in 1h 49m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c4cfc10f06b442a9953d94e16f79496f : SUCCESS in 32m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9683231ca7cb432cb0f2f9ca0b7feb21 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"30dfd4d07d930b2b6d9a49f63ba8527cfcc2ee96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:37:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72c3228a39ae4b6de1ece2a6bab5ec4f8829ddc8","revisions":{"72c3228a39ae4b6de1ece2a6bab5ec4f8829ddc8":{"kind":"REWORK","_number":1,"created":"2023-02-08 08:30:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/65/873065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/873065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/873065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/873065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/873065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/873065/1"}}},"commit":{"parents":[{"commit":"a93fc98451be5d8c9b4ab3cf7e78f9c300d0f3ec","subject":"Merge \"Prepare the stable/wallaby jobs with pinned manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a93fc98451be5d8c9b4ab3cf7e78f9c300d0f3ec"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-08 08:25:10.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-08 08:25:10.000000000","tz":120},"subject":"[RBAC] Add skip condition for share manage","message":"[RBAC] Add skip condition for share manage\n\nIn case a backend doesn\u0027t support this feature\nthe tests will fail.\n\nChange-Id: I69fd7a3747986ce287c3a59cdc0f374fdd5b0099\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72c3228a39ae4b6de1ece2a6bab5ec4f8829ddc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72c3228a39ae4b6de1ece2a6bab5ec4f8829ddc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[]}
