)]}'
{"id":"openstack%2Fmanila-tempest-plugin~874377","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I99ed8136266541731df157b25566e73255bca3bd","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-02-21 13:23:26.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-02-21 17:19:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I99ed8136266541731df157b25566e73255bca3bd","subject":"[RBAC] Generate unique arguments for access rule creation","status":"MERGED","created":"2023-02-20 10:09:54.000000000","updated":"2023-02-21 17:19:17.000000000","submitted":"2023-02-21 17:19:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874377-secure-rbac","meta_rev_id":"3628c74e9a5b598b8d7be39dbbd7bf7e4487eb41","_number":874377,"virtual_id_number":874377,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 17:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-02-20 23:26:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-02-21 13:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-02-21 15:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 13:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 13:01:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-20 23:26:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-20 23:26:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"9e80f830177cebb535df051a45a53a73489dc339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 10:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b28c453e8390ce8b0352da335da02a8e312e6f34","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 10:10:46.000000000","message":"Topic refactor_test_rules removed","accounts_in_message":[],"_revision_number":1},{"id":"a7a3d9388862ceb45b63d6eee677d7532f136c37","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 10:10:54.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"d74e6fba2a77b36e5fc9e98cb0cf80a934e0cda0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 11:00:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b18c246a0eb5b57de53d0e25a59039413c64bf1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 11:01:04.000000000","message":"Topic refactor_test_rules removed","accounts_in_message":[],"_revision_number":2},{"id":"d7a12880120bff5703542748e01513d271debdac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-20 11:01:17.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"34029c02fa671ae06c3d3d3c556ccf882d730c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:01:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b25f5ef51084cf6ac7f8a0179a7b374\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c3c8bd19914aa8a1f42ed499a93918 : SUCCESS in 5m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/87aea05a4f244fc3a61286ace434d054 : SUCCESS in 1h 07m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/da7b9c71b8f14c2385204b2166ef26cd : FAILURE in 1h 57m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/51c7f3bdb12f491cbf2675329587879e : SUCCESS in 1h 11m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9e278662546c4e5e8463490008825948 : SUCCESS in 1h 15m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e169f4b10b0a44369eaf27ec20d6b14a : SUCCESS in 1h 53m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bf3e29476b5c4df58a1d6072f06d222b : FAILURE in 1h 07m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc9aad949080420cb712dfa856e26327 : SUCCESS in 26m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/53bceeb1fd194e09a84ad24aad1f59b2 : SUCCESS in 43m 45s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/62cd5fd033e54a269fc89878234651b5 : SUCCESS in 1h 23m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/233046cb77ba46bb99d6a168eb4fd062 : SUCCESS in 1h 22m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/232ba528273b4956b0ad38a370b3a9ab : SUCCESS in 46m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/03b1f707a97c479095e59902e552edfb : SUCCESS in 1h 16m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6c33ea7ab5684ca2b20c75fc4cb3d8ed : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22f848021277993d0f76b70155801e3f1ec199c4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-20 13:01:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/874377/2/check/DellEMC_Unity_Manila_dhss_true/dab8345/DellEMC_Unity_Manila_dhss_true/1272 : FAILURE in 1h 07m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/874377/2/check/DellEMC_Unity_Manila_dhss_false/08bbbe6/DellEMC_Unity_Manila_dhss_false/1372 : SUCCESS in 1h 01m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"30df80fd1f8101b56b75473c25e921678e50a467","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-20 23:26:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"940a62223db62c6ff29ef19c2dbaf74acae44994","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-02-21 13:23:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e8a7a53423611fd74700042c8e6e6744c59166a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-21 15:24:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd620f011b9d215e0fc1c3ffd79a9d6f492da9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:25:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a0448e2642b1f169585d5d72e14f244366860af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:19:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabca449adbc48b9bbe090d2358affab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c23cc3ed4b4120b3b49b35d853459c : SUCCESS in 3m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9bc43cab18d1490387fea15536aae202 : SUCCESS in 1h 28m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7997a0e3fa3b467099696f06b907d911 : SUCCESS in 22m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8dab3c6b34294ec88e978219fa6375b3 : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3628c74e9a5b598b8d7be39dbbd7bf7e4487eb41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:19:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c586d4965e396436ac56f0aa0344cf140a1447f2","revisions":{"7a81e79ca1460621c110b9f65069260767d0783b":{"kind":"REWORK","_number":1,"created":"2023-02-20 10:09:54.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/874377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/874377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/1"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-20 09:56:21.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-20 09:56:21.000000000","tz":120},"subject":"[RBAC] Generate unique arguments for access rule creation","message":"[RBAC] Generate unique arguments for access rule creation\n\nIt\u0027s prefered to use unique value for \u0027access_to\u0027 variable.\nIt prevents granting the same value of \u0027access_to\u0027 to the\nsame share which will lead to 400 execption.\n\nChange-Id: I99ed8136266541731df157b25566e73255bca3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a81e79ca1460621c110b9f65069260767d0783b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a81e79ca1460621c110b9f65069260767d0783b"}]},"branch":"refs/heads/master"},"c586d4965e396436ac56f0aa0344cf140a1447f2":{"kind":"REWORK","_number":2,"created":"2023-02-20 11:00:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/874377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/874377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/874377/2"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-20 09:56:21.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-02-20 11:00:47.000000000","tz":120},"subject":"[RBAC] Generate unique arguments for access rule creation","message":"[RBAC] Generate unique arguments for access rule creation\n\nIt\u0027s prefered to use unique value for \u0027access_to\u0027 variable.\nIt prevents granting the same value of \u0027access_to\u0027 to the\nsame share which will lead to 400 execption.\n\nChange-Id: I99ed8136266541731df157b25566e73255bca3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c586d4965e396436ac56f0aa0344cf140a1447f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c586d4965e396436ac56f0aa0344cf140a1447f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
