)]}'
{"id":"openstack%2Fmanila-tempest-plugin~875660","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic958ac7e9f6f4b9267a06e5c813bcabe8a1ed9a1","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-17 19:59:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic958ac7e9f6f4b9267a06e5c813bcabe8a1ed9a1","subject":"DNM - Testing manila-tempest-plugin CI","status":"ABANDONED","created":"2023-02-28 12:47:05.000000000","updated":"2023-10-17 19:59:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7793d546d69b312041969981633077474d5b1faf","_number":875660,"virtual_id_number":875660,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-28 21:48:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 13:39:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-28 21:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9c3ec1d76daf39b5b2a18671faa8a1e642a3ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-28 12:47:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33afd9751687e542ec2524df00c3987db638239d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-28 13:39:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/875660/1/check/DellEMC_Unity_Manila_dhss_true/52c8dfa/DellEMC_Unity_Manila_dhss_true/1322 : FAILURE in 51m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/875660/1/check/DellEMC_Unity_Manila_dhss_false/ee76c60/DellEMC_Unity_Manila_dhss_false/1422 : SUCCESS in 51m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9aec38863281d54bbcafc4093cc189ac6bc9abf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 21:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e75a3cd1d774b5fb56ee24c3c56ad60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7e6c7ff3e8448fad100f89d9e263cd : SUCCESS in 2m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c39f4b26a69421d83b14fa13812b3cd : SUCCESS in 1h 07m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2fdfefb73b534e80aea1d35d586f6de6 : SUCCESS in 2h 00m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c1e33c4e04b468fa6659599cb420ea7 : SUCCESS in 1h 04m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6696825731fc45afad3b9683d092519a : SUCCESS in 1h 28m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e18bbc9d5094458ad0048dbb1bb5386 : SUCCESS in 1h 51m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/47350e7c09854b759ccf50fb904c700d : SUCCESS in 1h 00m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1303935ce14f4d8c8fa65cb3ae388004 : SUCCESS in 27m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3455e338319b4d85b39f9ad1880ffb88 : SUCCESS in 43m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8e25818134f644d6965e998fc62bc852 : SUCCESS in 1h 24m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/712ace91e5a5450eafd1a636c66e725f : SUCCESS in 1h 20m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/be32a6995d504600a388500e337f9957 : SUCCESS in 39m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/085282a16c394d44b09240514e9419dd : SUCCESS in 51m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/335f6ca7b2b44409b17135b972e34186 : SUCCESS in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dfcd62f021f2fa8e925c8fca9734801c2006aae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-28 22:50:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/60/875660/1/check/DellEMC_Unity_Manila_dhss_true/55508a4/DellEMC_Unity_Manila_dhss_true/1324 : FAILURE in 51m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/60/875660/1/check/DellEMC_Unity_Manila_dhss_false/5183532/DellEMC_Unity_Manila_dhss_false/1424 : SUCCESS in 1h 00m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7793d546d69b312041969981633077474d5b1faf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-17 19:59:08.000000000","message":"Abandoned\n\nTesting was done, we can abandon this now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b330a132b65b443d7a14071d6241c149d6bbcb1d","revisions":{"b330a132b65b443d7a14071d6241c149d6bbcb1d":{"kind":"REWORK","_number":1,"created":"2023-02-28 12:47:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/60/875660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/875660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/875660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/875660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/875660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/875660/1"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-02-28 12:46:38.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-02-28 12:46:38.000000000","tz":-180},"subject":"DNM - Testing manila-tempest-plugin CI","message":"DNM - Testing manila-tempest-plugin CI\n\nChange-Id: Ic958ac7e9f6f4b9267a06e5c813bcabe8a1ed9a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b330a132b65b443d7a14071d6241c149d6bbcb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b330a132b65b443d7a14071d6241c149d6bbcb1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
