)]}'
{"id":"openstack%2Fmanila-tempest-plugin~875981","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I05607883c34bca586c795c7efab1df8d0edd6cff","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2004212","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-03-13 23:48:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05607883c34bca586c795c7efab1df8d0edd6cff","subject":"Correct share group tests cleanup order","status":"MERGED","created":"2023-03-01 19:29:33.000000000","updated":"2023-03-13 23:48:59.000000000","submitted":"2023-03-13 23:48:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875981-bug/2004212","meta_rev_id":"a767bd0852751b8bfad630730319a415414402c6","_number":875981,"virtual_id_number":875981,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 23:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-03-07 00:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-03-13 22:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 00:26:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-02 02:32:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-03 19:02:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"da5afd79872c90189dddddf5469ccbb2b3121840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-01 19:29:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a290f9829950493cc090be5a10338ae5195aeb8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 00:26:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/81/875981/1/check/DellEMC_Unity_Manila_dhss_true/cd2553c/DellEMC_Unity_Manila_dhss_true/1332 : FAILURE in 50m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/81/875981/1/check/DellEMC_Unity_Manila_dhss_false/a2f6e56/DellEMC_Unity_Manila_dhss_false/1432 : SUCCESS in 57m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b523f19df6922556ae7640a6f279e3a4023505b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:32:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18b6528ec69e4825bd831dfdf4704901\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1327aa54a21c427aadcc305c792d010b : SUCCESS in 4m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8262577824ce4decb714f8ccffe5cf28 : SUCCESS in 1h 37m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a2f44eb0a22647a2920895107f6514ee : SUCCESS in 2h 04m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f93c60ccea244485a29bb41031a89f96 : SUCCESS in 1h 10m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3c1ca5ffaaec4340aa05918a5926e24b : SUCCESS in 1h 14m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1a3689b3be524545b5dadd96ba6b840f : FAILURE in 31m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fd8d0b204f3346a1a856c8c27ea9e588 : FAILURE in 1h 17m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44042136a6af4e5cb072c66236f369a8 : SUCCESS in 37m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1e65c466ed5b46989999303df0a4dc58 : FAILURE in 31m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5f58d4cd160e4679bdd3e558eca25d99 : SUCCESS in 1h 31m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/edd19b2e484a4ef5807c6e9bd8cba0ac : SUCCESS in 1h 17m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b1a6ff9f9e274a57bbd29f0b792d7325 : SUCCESS in 29m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cd3c6381d6cf4b8cbf8f894196a5583e : SUCCESS in 44m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/795dbeebdc014eb3aac6227cc724bd88 : SUCCESS in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82896ba7b38f14be980eeb07aa9f83ed885a6ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-02 11:35:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aef3c5bfc3ca72038288cfaefff4f4cfa8e682c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-02 12:10:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"003295f1c54ef1795a896b8a7d7970a5a9ff8869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-02 13:12:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ecf62a476383d849bbc396ec584593b11614119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:13:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f00fac7d409446bba412eaad7536c62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c2c08b2c3e44ce9e4208f197004369 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/41dd8a2c488d4be1b618abd0315d6409 : FAILURE in 1h 31m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/347bcaaf728f4f9788f395247755170b : SUCCESS in 1h 58m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cab53f208547428ab050547b62eca950 : SUCCESS in 1h 12m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/72967b5a45014106a99862b291fbc273 : SUCCESS in 1h 38m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b423ea165c143baaf7ba1fc0c590f2a : FAILURE in 1h 43m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ee23c0768c9f42bc81158a8da7764aed : FAILURE in 1h 10m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2edba12083a74d2a8b9ea9f5d97309ce : SUCCESS in 26m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/636b5513e84f491a9f8375345ff8d89d : SUCCESS in 38m 20s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ddbb1db69c4f4513b131c6328aeca3fb : SUCCESS in 1h 32m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/57de8818b3aa4df2b9cbee5d4fefa5f3 : SUCCESS in 1h 06m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/77e65dc06ffe4bcab8477456f718fbe7 : SUCCESS in 32m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ff537e871e3241eb8332d2f10c213853 : SUCCESS in 45m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3716a2b896364faeb312029749cdc2e4 : SUCCESS in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"205d6e91db10cec5486f265b20ec5923d170cefe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 16:10:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/81/875981/4/check/DellEMC_Unity_Manila_dhss_true/b13be7c/DellEMC_Unity_Manila_dhss_true/1341 : FAILURE in 51m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/81/875981/4/check/DellEMC_Unity_Manila_dhss_false/6a28d11/DellEMC_Unity_Manila_dhss_false/1441 : SUCCESS in 56m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"0125b8d03f6e31a26a36f1e52e768503245820e9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-02 16:46:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b7f41444978c9e637dfffa95fef8bcaf0ae93f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 21:33:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0be3378b7b934b06afa2562f9a45ea89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018e6486e2404b78969ab06344928ad3 : SUCCESS in 3m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b722581fbd54e1f8acc35fd25e1495f : SUCCESS in 1h 08m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bf23a98a56334e60a8027a2784d02b4c : FAILURE in 2h 02m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6ea54ec96874b7bbc75e0fc3f4f0da2 : SUCCESS in 1h 08m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9a2ed1706c3a4e1688bffb5619c938b1 : FAILURE in 1h 44m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/27381fd23b294774b89ccade672677ef : SUCCESS in 1h 36m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9d85b510c714bc2b32b9d4b6d5bc08a : FAILURE in 1h 03m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a50d57ab7e0644c781331257f2c77605 : SUCCESS in 28m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e23a661ff97242ea8b90cd83d82282fe : SUCCESS in 37m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c5400f7c19434a909bf61de119825ee8 : FAILURE in 1h 20m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/742546974afc412299648f8b4082e50d : SUCCESS in 1h 15m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e44ca8d8699c48ee904752bb23dc1a25 : SUCCESS in 30m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7551e311bde14d1583f500bd55d5630a : SUCCESS in 50m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/408258c38a1543749628da465e6cd1e7 : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59753623c66ececd020560888483b7e77a841581","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-03 19:02:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0cfa18576dbb1d8d18de77a42c4299bb6442e238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-06 18:58:38.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5058379f75b5ff9dfd392ab2b5a6cb60cad2a3bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-06 19:56:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/81/875981/5/check/DellEMC_Unity_Manila_dhss_true/be9526b/DellEMC_Unity_Manila_dhss_true/1356 : FAILURE in 52m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/81/875981/5/check/DellEMC_Unity_Manila_dhss_false/b53b3c5/DellEMC_Unity_Manila_dhss_false/1457 : SUCCESS in 58m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"efe9de6ce8f7f9b514becd005d3182f18cb39d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 22:22:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb0c5005af8842e3bd2eba474079c60a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816994aa63ed4a66b097ada38139d695 : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/844fb066e4ca468288807579598e7f6b : SUCCESS in 1h 25m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ec8512310eb64801922c9c406d2128b6 : TIMED_OUT in 3h 02m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4bc429cc9ab84a23a3355963dad58dfe : SUCCESS in 1h 24m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a2ee55a4621247e1b7b017b5cfa07110 : SUCCESS in 58m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e0fbeb0a56ba4bd69468b8705ee8ec1e : SUCCESS in 1h 14m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dd82455d8f184d1bbe62f6be6956d6b2 : SUCCESS in 1h 03m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4cefb44049294a0d91bb5de56960d4e0 : SUCCESS in 35m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/706fc597a0974639a74a011c3b31dbf5 : SUCCESS in 31m 42s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0c8e643d299e45baa0ef7c137007c4ea : SUCCESS in 1h 00m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e945479d3cc94030bfa350625c2a02c6 : SUCCESS in 1h 07m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b9a327b893134c1983519a484bf6ca60 : SUCCESS in 19m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e489165a91a6467caf55cd6783334bb5 : SUCCESS in 42m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/495316ce19384addb002abab49a00216 : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d7fefd372e4fa3583e1468760ac42513ba577f21","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-06 23:17:15.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/81/875981/5/check/DellEMC_Unity_Manila_dhss_true/776c378/DellEMC_Unity_Manila_dhss_true/1358 : FAILURE in 51m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/81/875981/5/check/DellEMC_Unity_Manila_dhss_false/bf2c3c0/DellEMC_Unity_Manila_dhss_false/1459 : SUCCESS in 54m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"ba7e4e7456b4fb3c5a8902f9a4fddf6f264a0455","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-07 00:23:28.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec9480d9d4c56f5bcf642eb1e65eba7c098ecb68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-07 00:23:46.000000000","message":"Topic set to bug/2004212","accounts_in_message":[],"_revision_number":5},{"id":"29cb7f304c035c3e117516aa07d6c9da533de2d1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-13 22:54:34.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dde7027f398d75eb9619eb181b16bdcc073e517e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 22:55:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8bde3f0d466cbd086499cf4150f949828613c878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:48:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a5fd34dfa8450b98b2d788405bc31a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8e1ba55e3e248d7a85631d99e612a01 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/38a236ad187d422ebf23e5254f86f4f5 : SUCCESS in 48m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3d552d00e07645f5a6de8300c71aae60 : SUCCESS in 24m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3373b4f866cd4c44b532049aedae2a0b : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a767bd0852751b8bfad630730319a415414402c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:48:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dab53f8e2b6d84114288ccf3c63751be8236cc57","revisions":{"c0e9fbbb6f6f0a8dbd27753e3b6915bca4e7fabb":{"kind":"REWORK","_number":1,"created":"2023-03-01 19:29:33.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/81/875981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/875981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/1"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"subject":"WIP Fix share group tests","message":"WIP Fix share group tests\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/872330\nChange-Id: I05607883c34bca586c795c7efab1df8d0edd6cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0e9fbbb6f6f0a8dbd27753e3b6915bca4e7fabb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0e9fbbb6f6f0a8dbd27753e3b6915bca4e7fabb"}]},"branch":"refs/heads/master"},"6a4eefe967725003071800b59ea3e6487680d0dc":{"kind":"REWORK","_number":2,"created":"2023-03-02 11:35:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/81/875981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/875981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/2"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-02 11:35:26.000000000","tz":-180},"subject":"WIP Fix share group tests","message":"WIP Fix share group tests\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/872330\nChange-Id: I05607883c34bca586c795c7efab1df8d0edd6cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a4eefe967725003071800b59ea3e6487680d0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a4eefe967725003071800b59ea3e6487680d0dc"}]},"branch":"refs/heads/master"},"d6e90e7025f893a8e311206485dfb5a7e4ba6ba9":{"kind":"REWORK","_number":3,"created":"2023-03-02 12:10:23.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/81/875981/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/875981/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/3"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-02 12:10:11.000000000","tz":-180},"subject":"WIP Fix share group tests","message":"WIP Fix share group tests\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/872330\nChange-Id: I05607883c34bca586c795c7efab1df8d0edd6cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6e90e7025f893a8e311206485dfb5a7e4ba6ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6e90e7025f893a8e311206485dfb5a7e4ba6ba9"}]},"branch":"refs/heads/master"},"b3466b8d8fb3a3d68079ab22d234b6b7fce0e5b7":{"kind":"REWORK","_number":4,"created":"2023-03-02 13:12:06.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/81/875981/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/875981/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/4"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-02 13:11:43.000000000","tz":-180},"subject":"Correct share group tests cleanup order","message":"Correct share group tests cleanup order\n\nSome share group tests were taking advantage of a bug in Manila.\nShare network subnets were being allowed to be deleted even when\nits share servers were tied to a share group. With that bug being\nfixed, we need to ensure that the share groups are deleted before\nthe share network subnet alongside the share servers are.\n\nRelated-Bug: #2004212\nDepends-On: https://review.opendev.org/c/openstack/manila/+/872330\nChange-Id: I05607883c34bca586c795c7efab1df8d0edd6cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3466b8d8fb3a3d68079ab22d234b6b7fce0e5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3466b8d8fb3a3d68079ab22d234b6b7fce0e5b7"}]},"branch":"refs/heads/master"},"dab53f8e2b6d84114288ccf3c63751be8236cc57":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-03-06 18:58:38.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/81/875981/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/875981/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/875981/5"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-03-01 19:28:05.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-03-06 18:58:38.000000000","tz":0},"subject":"Correct share group tests cleanup order","message":"Correct share group tests cleanup order\n\nSome share group tests were taking advantage of a bug in Manila.\nShare network subnets were being allowed to be deleted even when\nits share servers were tied to a share group. With that bug being\nfixed, we need to ensure that the share groups are deleted before\nthe share network subnet alongside the share servers are.\n\nRelated-Bug: #2004212\nChange-Id: I05607883c34bca586c795c7efab1df8d0edd6cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dab53f8e2b6d84114288ccf3c63751be8236cc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dab53f8e2b6d84114288ccf3c63751be8236cc57"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
