)]}'
{"id":"openstack%2Fmanila-tempest-plugin~881533","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If7bca173830d8d0912873d073d64c6251714817f","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-09 05:38:20.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-05-09 05:38:20.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-09 04:25:02.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"If7bca173830d8d0912873d073d64c6251714817f","subject":"[CI] Reduce mysql memory in all test jobs","status":"MERGED","created":"2023-04-26 02:49:59.000000000","updated":"2023-05-09 05:38:20.000000000","submitted":"2023-05-09 05:38:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881533","meta_rev_id":"8f623639f4dd2363da3aadc3cc66541c7f1180fa","_number":881533,"virtual_id_number":881533,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 05:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-05-08 18:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-09 04:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-09 04:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 06:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-26 06:19:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-26 09:17:02.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-08 18:41:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-05-08 22:39:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-05-08 22:39:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"17f1110ef9ac8f6e10d518c6a14847d93250ef54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-26 02:49:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1831573d841ff2396bf7fe8894d0a9d827eb3834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 06:00:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788a2b22993f4dbd8f134ef56e6fb511\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e250a15bf3fa4ef6b58eddc0c4af7c77 : SUCCESS in 3m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a9d1fb8e4a7141bf9ac9cea0284d29a9 : SUCCESS in 1h 20m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f01ec27e1ad4319b1da58ad4653c0f4 : TIMED_OUT in 3h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d319a6fee7564bf2aaa00d8f3e87d059 : FAILURE in 2h 41m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1e2a3a9480ef40c0948ec0b2cf40ac83 : SUCCESS in 52m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9ab80ae660934a29862931cab609c820 : FAILURE in 1h 44m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5accfdac53054d46ba89b1d4a8bf9023 : FAILURE in 1h 07m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1daaec91d1b24a61bed4e7a178b5f7ab : SUCCESS in 44m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1b16e3073a2743f1aa565214065bc167 : SUCCESS in 31m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6b39c03378cf4083a1ccd3070f252cd3 : SUCCESS in 1h 02m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c8680accf0b34420b0e98a1309f1d696 : SUCCESS in 1h 17m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/50167f7fd83144d5aaf1589e6ac77974 : SUCCESS in 33m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa74772b39e147bd98421b82b9de285d : FAILURE in 52m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/93c62ba79ae643d9b17655ba58c668ab : SUCCESS in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5eb72102ef269885225e4563f6b80476fd882233","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-26 06:19:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/33/881533/1/check/DellEMC_Unity_Manila_dhss_true/33fb9d9/DellEMC_Unity_Manila_dhss_true/52 : SUCCESS in 3h 29m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/33/881533/1/check/DellEMC_Unity_Manila_dhss_false/4b81750/DellEMC_Unity_Manila_dhss_false/51 : FAILURE in 1h 19m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8660a2df6160343f5edfd945f8d8254c3d45def9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-26 09:17:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/881533/1/upstream-check/manila-tempest-plugin-ontap-dhss/a1a65ea/ : FAILURE in 3h 09m 15s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/881533/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/063866c/ : SUCCESS in 2h 55m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"77b77e57cf4b57c8670ff77df5fcc449052f01f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-26 13:36:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9cb28d7238de1136f364faa3aaaeb9d91deeac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:58:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc740b3dd8c48d6a46b13e38b49f166\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777741c68aaf4b298a10baa6c400ecdd : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c95c558f762d43f6adaa9e280831f796 : SUCCESS in 1h 04m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/80cab15171de4760b450260cfd9b102d : FAILURE in 2h 21m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cc29ea1e1df54656acca7fca76cc292f : SUCCESS in 1h 08m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bd1b9d06065a42819d3c84137880dfff : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/36029a1ebf164ce68577605142309da4 : FAILURE in 1h 18m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1bc35d80ebce42de838db2e33f4b1ccc : SUCCESS in 1h 09m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e066b00afc3147c7a0419af12a390d44 : SUCCESS in 27m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/899ade8ed42b4c289f12a0ace5e797bb : SUCCESS in 44m 00s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b55e9fcbf4744a0099a7491370b6361f : SUCCESS in 1h 26m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9c187171c0ba4afc8c42567625961dda : SUCCESS in 1h 14m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c643d76aa3d3417fa2d1e75c9e0fb01d : SUCCESS in 33m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/98194a4515d140fba55c4eac1ec3d99e : SUCCESS in 48m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ef39ce63d6114ca5a044193120417915 : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bd564c67d9a335cb0e4886ea0d21ccda57c8b23","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-26 18:05:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/881533/1/upstream-check/manila-tempest-plugin-ontap-dhss/af17dd1/ : FAILURE in 2h 05m 38s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/881533/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/c6cd357/ : SUCCESS in 1h 43m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"121f0601c080e923b990e185692d57a54966192b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-26 18:39:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/33/881533/1/check/DellEMC_Unity_Manila_dhss_true/d703e65/DellEMC_Unity_Manila_dhss_true/54 : SUCCESS in 2h 41m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/33/881533/1/check/DellEMC_Unity_Manila_dhss_false/e3e4ef8/DellEMC_Unity_Manila_dhss_false/53 : SUCCESS in 1h 15m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6850e9da577645ce19e0ddb5500da5c3c3b3629e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-08 18:41:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a691524c7f43599c7c4f9c2bfa8f5dd82c44722","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-09 04:25:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59e569232e55cb867fb3629cb2e6e8a1b9bad2ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 04:25:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52ced87bea5132ea77638f02e6e46c8073eb6bb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 05:38:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c55106014b4003a13e3e028228432c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51302bca3a047938fcaf04aff71f9a5 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/daa8dc4f794e4d68939b13c71d5b038a : SUCCESS in 1h 11m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/543a480874434b60a042af34f6af5cb7 : SUCCESS in 25m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bbff3d76766e4fcd93232bbe47e837d1 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8f623639f4dd2363da3aadc3cc66541c7f1180fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 05:38:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82b1e6f64bcf19f4092592fb74d19f48ab46d343","revisions":{"82b1e6f64bcf19f4092592fb74d19f48ab46d343":{"kind":"REWORK","_number":1,"created":"2023-04-26 02:49:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/881533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/33/881533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/881533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/881533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/881533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/881533/1"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 02:49:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 02:49:38.000000000","tz":-420},"subject":"[CI] Reduce mysql memory in all test jobs","message":"[CI] Reduce mysql memory in all test jobs\n\nChange-Id: If7bca173830d8d0912873d073d64c6251714817f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
