)]}'
{"id":"openstack%2Fmanila-tempest-plugin~881676","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I66ac85b735566f8f7dd75234eb8e0be7773fee3a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"unpin-focal","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-09 11:46:57.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-05-09 10:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-05-09 06:50:21.000000000","reason":"\u003cGERRIT_ACCOUNT_6413\u003e replied on the change","reason_account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-09 11:46:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66ac85b735566f8f7dd75234eb8e0be7773fee3a","subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","status":"MERGED","created":"2023-04-26 23:52:16.000000000","updated":"2023-05-09 11:46:57.000000000","submitted":"2023-05-09 11:46:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881676-unpin-focal","meta_rev_id":"abd59661bcf8a2fc7f86a16e1746aa134336931a","_number":881676,"virtual_id_number":881676,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 11:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":2,"date":"2023-05-09 10:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-09 06:50:21.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":1,"date":"2023-05-09 10:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 01:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 06:24:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-27 07:40:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-04 15:13:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2023-05-04 15:13:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-05-08 22:36:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"9c943556149030d1f5ff8955de5708221d6af209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-26 23:52:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17a3548012996796c79c8a40bf371c5b76378f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 01:50:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efb8be76cc184f38b22ea11427be13cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730f1a6954d142edbe29094bb21335e4 : SUCCESS in 4m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49e87367d4484fb2871fb712cc148bee : SUCCESS in 1h 15m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/958029e63bdd4c0eaf89f72e9a639feb : SUCCESS in 1h 54m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e8540ab8809a4fb6a0cd1b2697365aaf : SUCCESS in 1h 20m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/60f52ba592d248e5bffcf2b442ea9b26 : FAILURE in 23m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/efdaa86137a54f17ba2dcdd4392c961a : FAILURE in 1h 40m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8fbf40c584c4c9ebc59ac99f3397a52 : FAILURE in 1h 17m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3908b9fea9de484ba1a800acdcbeb9c2 : SUCCESS in 34m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/926e692f1632406cb44f1f9d1f51cbc7 : SUCCESS in 42m 20s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/bb3038c233034560847d883acc0ba716 : FAILURE in 1h 25m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d6fffceed8b6472081b1cb716d3bb3e7 : SUCCESS in 1h 10m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/780cdd5a2124449f90488cef07fa9be3 : SUCCESS in 22m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/684a4c9756fb4400aeb2481e98af60c8 : SUCCESS in 35m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c706de948b964bc39fb5554bf6ec8712 : FAILURE in 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d492eb63af7fce9786076b37c2f55ab080c7bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-27 02:36:25.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"003b6dbd9d045728b482f06a7ad27c14cc186bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 04:24:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4887993d8979482da83ff9cc58749caa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae4eda361974f5080dac8f4718e9acc : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0ed1c240e28b489fa60bca43fa958291 : SUCCESS in 1h 18m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0b6931a0a33942d0bd8f594dc0e76c10 : SUCCESS in 1h 45m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/43cc02d8acd74694bccf874f776429c7 : SUCCESS in 1h 09m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/73ececa3be1f444da5c137caa3d004d8 : FAILURE in 1h 25m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/67fbdba098574cc9aee42ec5d52749cc : FAILURE in 7m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de5a37d8aed34431abbd460f095dbb67 : FAILURE in 1h 14m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9526caba0d2d40109defef302e4de3a6 : SUCCESS in 37m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/28771ac9313d446fb5fc6b0e9136ddd9 : SUCCESS in 40m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/478d8f9ab0fb4f06a25e2a1736d9845c : SUCCESS in 1h 28m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3a24f07034c241989b659d87a7d3539f : SUCCESS in 1h 06m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/544f3cc7ecc348d98c29e49349674096 : SUCCESS in 22m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d38c1847611a4597a5be35ff4f4640f2 : FAILURE in 49m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fd47d1c2e8354bb0b7be5f8f1c339842 : SUCCESS in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b966aa501d310788d2fc49a6c1d6f0038f3ac232","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-27 06:24:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/76/881676/2/check/DellEMC_Unity_Manila_dhss_true/89ff4b4/DellEMC_Unity_Manila_dhss_true/57 : SUCCESS in 3h 47m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/76/881676/2/check/DellEMC_Unity_Manila_dhss_false/04356ef/DellEMC_Unity_Manila_dhss_false/56 : SUCCESS in 1h 12m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"099ad2e7bb873388b9f8eeee0f78d947bbdd50b2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-27 07:40:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/2/upstream-check/manila-tempest-plugin-ontap-dhss/b2a3d63/ : TIMED_OUT in 3h 14m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/13fd959/ : SUCCESS in 2h 10m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c754a7bd0296e7fe5fa21e485d6d2db41f9353b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-27 19:23:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"18b80ea5ebe424899d04e5cb529fd52621619e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 21:55:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bba6527d14184e3a861049ea4c2a21d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6101578aef48e884e9e08aead24b8b : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dc74564dd39b4a038eba90c9e81ee889 : SUCCESS in 1h 24m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a7b06b481f54bbc8dd9aab31ea6c4dd : FAILURE in 2h 25m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/59a24b10229442c794d31c1ea941471b : SUCCESS in 1h 16m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/16cfe5abd43149deb413765deadc5e21 : SUCCESS in 45m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/71f3ccfd463248059501fc8e9b9c4d82 : FAILURE in 5m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/448aec6104eb425b8d72fc7480f0f43e : FAILURE in 2h 01m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a525fa05ff5f4af8b9d94c8ed8d8818c : SUCCESS in 40m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4582905c5e12422bbbc4485cf5d1fa96 : FAILURE in 28m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3da3038531a141659e28e8ccf30e31ee : SUCCESS in 1h 19m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f48c0963a1c049b8b86016904fcb0eb2 : SUCCESS in 49m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cc679f55d1b84878a201eab3762c2f3a : SUCCESS in 34m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/28700e994bf244c596c19d9e650d688e : SUCCESS in 1h 47m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1647d9c5292c48fea1fcae30705cc47f : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6a45949ebea97582eec67b9ba2c93f2f5eb1a86","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-27 22:06:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ddbb80cdee51c7981d258f909bc88996372ac0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 01:31:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2fa1f3eab2461abd1f26dcaf6d075b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0d930b2a3a4088a0c5b4f1b3fcb758 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/031618bc405b4b7191bb641e379aa5f4 : SUCCESS in 1h 12m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fa3e4a26f8fd4c95999a9df4a92b575c : SUCCESS in 2h 40m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3f67de0df01b49008358aeff8d7bd2da : TIMED_OUT in 3h 06m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ae383b30f1e744a397e4ceea45e96adc : FAILURE in 1h 20m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/881eb0b1cba8465ea55b20b9bc76d55d : FAILURE in 18m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fd167132809c4144acedb9fe7e255436 : FAILURE in 1h 14m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/782f0bec04c743f999873e5ed825f1e4 : SUCCESS in 37m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/76d43744a16643fa9db2a257791c4d0f : SUCCESS in 40m 49s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4a6ef578a537493389174618e5b4356d : SUCCESS in 1h 28m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e9588d26084346059825141d92e1d0f2 : SUCCESS in 1h 21m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f7339230c8c4496eb18bdca62ee1027d : SUCCESS in 32m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ead0e9c9e7e45ad867e47418504fdc4 : SUCCESS in 1h 01m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5df7652bea1f43d7b3a1f8102eaaa3c5 : SUCCESS in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ffad6dd1c74cb9d600c170bc9a1cbfffb40b3c72","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-28 02:46:19.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/76/881676/3/check/DellEMC_Unity_Manila_dhss_true/0a0d793/DellEMC_Unity_Manila_dhss_true/61 : SUCCESS in 4h 17m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/76/881676/3/check/DellEMC_Unity_Manila_dhss_false/9e2d043/DellEMC_Unity_Manila_dhss_false/60 : SUCCESS in 1h 05m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"7308b65fab72388609633275452fcf4ad4e544f3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-28 04:43:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/3/upstream-check/manila-tempest-plugin-ontap-dhss/01376fd/ : FAILURE in 2h 46m 16s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/aedb816/ : TIMED_OUT in 3h 10m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b5bb7bedbfd8dc1ae42f59f12cc9a7368cb7ef26","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-04 15:13:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91eaa86555806fe151a418f7fb06bca44eed8bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:05:03.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/865315 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"89016e558d4482beaa82e27fcd0241b962cba5f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-08 14:54:11.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"133b939a72ea5c5d7d3785122c33aa71c7ea0f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 16:53:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e55d0ed2764e20b6da96b60605c865\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c0e5d84ed0479cb5557661b45d1c82 : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/575389a686ad4b3ebaf5256640257faf : SUCCESS in 1h 08m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7186ca6798ee4b05a5227fa0d9e9a31d : SUCCESS in 1h 56m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3b39ed53b83f45d0bf00f55947936403 : SUCCESS in 1h 21m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/73d30fe366624316bc0b0b83008d3449 : SUCCESS in 59m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/819dcb931e6f4d73a1d897eb3b4f1240 : FAILURE in 14m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bbf803458b2347b586a6f9fe10e2586d : SUCCESS in 1h 05m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0a4d404309374567b3ef113a3180e1b0 : SUCCESS in 39m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fcd0b3f8a0a6407f82d5ca3158e33483 : SUCCESS in 47m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/19141105245b4d06b6651f0c4253e07d : SUCCESS in 1h 29m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/70a33542c5de417cac1810942de2209e : SUCCESS in 1h 07m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d4af64707c4d40b8ad480df454e11c4f : SUCCESS in 25m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/570788556c8b402aaa4c2862c56d0412 : FAILURE in 51m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/955b0fdd9f114a32aefbec83c23c0116 : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aa713bef67535dd149640f54e807a37a93ddf164","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-08 19:02:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/4/upstream-check/manila-tempest-plugin-ontap-dhss/1573ad6/ : SUCCESS in 1h 29m 24s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/881676/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/22a572d/ : FAILURE in 2h 07m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"bcb4cca596d8912eefd807d9c0e06fee840cf155","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-08 22:27:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/76/881676/4/check/DellEMC_Unity_Manila_dhss_true/d2d860a/DellEMC_Unity_Manila_dhss_true/87 : SUCCESS in 3h 04m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/76/881676/4/check/DellEMC_Unity_Manila_dhss_false/b7e8ff6/DellEMC_Unity_Manila_dhss_false/86 : SUCCESS in 1h 01m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e0d1797215212642fef25f8c2757d582e7f80542","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2023-05-09 06:50:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"718d453b97620938b7d92a34619773a49b0cf800","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-09 10:22:42.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec3f56e9c8f3416d6ba8566a3cf12e0188d5de90","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-09 10:22:59.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6a28e33447fe9a8be12a608cadc5c7ef42f2084a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:23:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91d55dd73eb70a7dda7d4323b4d4076510f64d3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:46:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f54a45530024e2cb099d7999d56c033\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c570b7655a411b98f96fbffca351b5 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/891df7422de14b4f9f5da4f334fca48d : SUCCESS in 1h 18m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/317179fa6ba1433b95286d5d25e99e25 : SUCCESS in 34m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1ce62892ec5a47cf85bbcca2c0170ac0 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":4},{"id":"abd59661bcf8a2fc7f86a16e1746aa134336931a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:46:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"797aeb1303d6b7f80679ff75c35b14028130ec84","revisions":{"ea659a258df0fb246d17efbfa6adf70b82298d89":{"kind":"REWORK","_number":1,"created":"2023-04-26 23:52:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/881676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/881676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/1"}}},"commit":{"parents":[{"commit":"82b1e6f64bcf19f4092592fb74d19f48ab46d343","subject":"[CI] Reduce mysql memory in all test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 23:52:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 23:52:00.000000000","tz":-420},"subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","message":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish\n\nWe are trying to drop testing the trunk branch (2023.2)\non Focal Fossa.\n\nChange-Id: I66ac85b735566f8f7dd75234eb8e0be7773fee3a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea659a258df0fb246d17efbfa6adf70b82298d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea659a258df0fb246d17efbfa6adf70b82298d89"}]},"branch":"refs/heads/master"},"03ccb9a1bb6d50dd8f92f73ca20999e5fd6b23a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-27 02:36:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/881676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/881676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/2"}}},"commit":{"parents":[{"commit":"82b1e6f64bcf19f4092592fb74d19f48ab46d343","subject":"[CI] Reduce mysql memory in all test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 23:52:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-27 02:36:25.000000000","tz":0},"subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","message":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish\n\nWe are trying to drop testing the trunk branch (2023.2)\non Focal Fossa.\n\nDepends-On: I899822fec863f43cd6c58b25cf4688c6a3ac1e9b\nChange-Id: I66ac85b735566f8f7dd75234eb8e0be7773fee3a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03ccb9a1bb6d50dd8f92f73ca20999e5fd6b23a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03ccb9a1bb6d50dd8f92f73ca20999e5fd6b23a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2e81d8662639a884c5838d631cca309bb5fc22a2":{"kind":"REWORK","_number":3,"created":"2023-04-27 19:23:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/881676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/881676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/3"}}},"commit":{"parents":[{"commit":"82b1e6f64bcf19f4092592fb74d19f48ab46d343","subject":"[CI] Reduce mysql memory in all test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 23:52:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-27 19:23:14.000000000","tz":0},"subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","message":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish\n\nWe are trying to drop testing the trunk branch (2023.2)\non Focal Fossa.\n\nDepends-On: I899822fec863f43cd6c58b25cf4688c6a3ac1e9b\nChange-Id: I66ac85b735566f8f7dd75234eb8e0be7773fee3a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e81d8662639a884c5838d631cca309bb5fc22a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e81d8662639a884c5838d631cca309bb5fc22a2"}]},"branch":"refs/heads/master"},"797aeb1303d6b7f80679ff75c35b14028130ec84":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-08 14:54:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/881676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/881676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/881676/4"}}},"commit":{"parents":[{"commit":"82b1e6f64bcf19f4092592fb74d19f48ab46d343","subject":"[CI] Reduce mysql memory in all test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82b1e6f64bcf19f4092592fb74d19f48ab46d343"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-26 23:52:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-08 14:54:11.000000000","tz":0},"subject":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish","message":"[CI] Run cephfs-native-cephadm job on Jammy Jellyfish\n\nWe are trying to drop testing the trunk branch (2023.2)\non Focal Fossa.\n\n\nChange-Id: I66ac85b735566f8f7dd75234eb8e0be7773fee3a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/797aeb1303d6b7f80679ff75c35b14028130ec84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/797aeb1303d6b7f80679ff75c35b14028130ec84"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
