)]}'
{"id":"openstack%2Fmanila-tempest-plugin~882561","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"rbac-wallaby","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-05-22 19:29:37.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-12 20:21:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-22 19:29:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b","subject":"[DNM] Check secure rbac tests in Wallaby","status":"ABANDONED","created":"2023-05-08 12:58:37.000000000","updated":"2023-05-22 19:29:37.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6bc8b48799455b28a6a9d8f59fb564d87a28ab3c","_number":882561,"virtual_id_number":882561,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-11 22:42:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 14:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-08 16:03:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-10 17:54:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-05-11 20:45:00.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"5764b3da8de4df3c982f737dcd673b0da582951c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-08 12:58:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02030e77de7b64f1b89ded75251944e35bdb746e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 14:59:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6feb42fc365046ba85d5045436b99459\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3395a71e95244c9ba5760f5e089c9a5d : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7a05712d085d4b75869674b04590084a : SUCCESS in 1h 11m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5699db500cb141e391561b219f58342d : FAILURE in 1h 59m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d18528335c2a46fa823baf86e97a609b : SUCCESS in 1h 24m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b2b5a1b74e484b929f69b224c0a8133b : FAILURE in 6m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/88f29e0021f54c79969faa4aa02bca6c : FAILURE in 6m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/00929ffb75034dd0bc3a774c17e17b59 : SUCCESS in 1h 04m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3adfccafedf74192b79c39b445f34ece : SUCCESS in 26m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/19b86a79eefc45b2b2461c5b1afcb09d : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d47a83c0db8c59a4216c85407ac97ca1796385b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-08 16:03:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/882561/1/check/DellEMC_Unity_Manila_dhss_true/e7596c6/DellEMC_Unity_Manila_dhss_true/85 : SUCCESS in 3h 04m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/882561/1/check/DellEMC_Unity_Manila_dhss_false/9a16d55/DellEMC_Unity_Manila_dhss_false/84 : SUCCESS in 1h 03m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"87e74b990e0a81e947d5cf5ff3a010a4e5e7d297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-08 17:20:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5b096b49489fa00e2bf386d1d0ee62e4a7a6d93a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 19:27:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd28d3165bfd42dba62d31f48385b992\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343b49fef86b48d6b2755bec5a7aaccd : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de54c3b4f75f4f3ca33f2aa742daa7c6 : SUCCESS in 1h 10m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c92e6a844bc44ec2a1037e9e7f609166 : SUCCESS in 2h 01m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/71e7786d8bd94eadbaafad02eab41e2d : SUCCESS in 1h 16m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/757814ec78154afa9fc1822efa847149 : FAILURE in 17m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/079f5aeeda5f42018fb7507a439d8532 : FAILURE in 5m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d4b162330df04f23af0b61e7fd0df00f : SUCCESS in 1h 13m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e12edef4b8d5470e88717c848357422d : SUCCESS in 33m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/7da1f9e6b1bd48588fe55974958ddc5f : FAILURE in 14m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0036544b512b76760992d9041c40ffbb474f37a0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 04:30:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/882561/2/check/DellEMC_Unity_Manila_dhss_true/28c1e85/DellEMC_Unity_Manila_dhss_true/89 : SUCCESS in 3h 11m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/882561/2/check/DellEMC_Unity_Manila_dhss_false/f288ed9/DellEMC_Unity_Manila_dhss_false/88 : SUCCESS in 59m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6dbf4eef47e2034d14c4cd048844e6c496827c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-09 13:29:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"514c56f98849623bbfd2ecab2dfb737b39f9e8ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 14:12:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/882561/3/check/DellEMC_Unity_Manila_dhss_true/aa28b54/DellEMC_Unity_Manila_dhss_true/93 : FAILURE in 43m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/882561/3/check/DellEMC_Unity_Manila_dhss_false/9499c91/DellEMC_Unity_Manila_dhss_false/93 : FAILURE in 32m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"323c453aba2f6c88373f09ee3401eeac1f026488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:14:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8da838a94652444b9d00c94743fc86f3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2876c9b93a8c4bdc81985372877e6be0 : SUCCESS in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a92ae9928d74a519c3b904123d91942 : FAILURE in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dfa0b6f8ef3e4ff586030fc47a85c96e : FAILURE in 25m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/90ab2aa96aab4e1ead00d166e837ec54 : FAILURE in 27m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7a109a4101f542a08c5c9433fdd57014 : FAILURE in 40m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8c588d56e2e4488dab02c3c7834cd565 : FAILURE in 29m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d8f0c44f85694f8a9da056f07744963b : FAILURE in 5m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dcfadb5f6b3e4551861bd74319281a6e : FAILURE in 39m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/37dc6eb17f9843d78b3eab593e99b775 : FAILURE in 25m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/23a9f3e5dc0b428a9699103510301470 : FAILURE in 15m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8af167504b35f1f6fb15d79376a2a0e197a343c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-10 11:18:50.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c4ef7cfe55b98216933bff9f5ece43b9f8eef1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:59:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86dd70dfa49a4873863b0930adabc423\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b65c645f2e124d618810d82d8e82e571 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200a3318da8949d399ccb34fb9006222 : FAILURE in 2m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b03921e8676458996e724119a6bb0b1 : FAILURE in 32m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8403c4b02d13458aabca11b75b21a0d0 : FAILURE in 33m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9aef67af34214d4bb06704e6667347d3 : FAILURE in 27m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/637863874bd048338f7456eaeb6a9304 : FAILURE in 27m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/19ddfca58e604e1e990e9e166f48b514 : FAILURE in 14m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/04829f60e112449d8a130590c19b0c4d : FAILURE in 34m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/47baf7d67bab42ee87cc1b6b2efbf573 : FAILURE in 21m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/774633c9c6544f9c90d9ac518726f528 : FAILURE in 17m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c1398ab3f48148265e6ddc1875cf77ba01bb6064","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 12:07:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/882561/4/check/DellEMC_Unity_Manila_dhss_true/1d0b9ca/DellEMC_Unity_Manila_dhss_true/98 : FAILURE in 29m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/882561/4/check/DellEMC_Unity_Manila_dhss_false/c374c1d/DellEMC_Unity_Manila_dhss_false/98 : FAILURE in 32m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"56a07a175eaea5e6c4b2988dc72266ee5b5dd1ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 14:29:54.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9f92724cad203196bfd635a0b663de38adfb42fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 15:28:15.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7c238d64722391579cfeaba5063b0d802b660c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:54:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/278b2774b99748ac95a8a8d39ad18e0d\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/43bee9cb61ea4b09bb1817d77d397429 : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3fc370f9c6b293d862679910f6563f3a9d4b6161","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 17:23:47.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bc5901670385df7d4e5022a23219d1fcdb625470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 17:48:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/232abd56c4444ff0a0be7cb2a698d652\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/404a900386974b38a32f1f51d54701ca : FAILURE in 23m 19s","accounts_in_message":[],"_revision_number":7},{"id":"f38c05799eb349d7ee7f7e1909e36c8ac5ac18ec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 17:54:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c094602d28f292163351876c74701fea27040bc9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 20:33:03.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"16e67a4b60d9dfe52457b2ded0139d2bef928cbe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-10 20:34:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"911e9158524af4b5ee959ad8330982f9c5559561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 21:10:27.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc9c4f8862bb4e7b9963739fc5d9764c\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/d85285f8c7884c4d9d37418b6e14bc20 : FAILURE in 35m 42s","accounts_in_message":[],"_revision_number":7},{"id":"a210eba11c4d25149b97b86d4c693fe27f2d38de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 17:25:53.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"10b0f4ff5ea2b85ba1b017a196a820b2e5833860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 17:52:34.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"0f9418fffbfe1ab4b23de4ab63846fded35cdc93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:25:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cef274c921bc4d2baaf059147c76c9ce\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/678fe214449a42abac46fa185ea871af : FAILURE in 32m 21s","accounts_in_message":[],"_revision_number":9},{"id":"acb6382c17c4ffc0b13add076315278ce39ed840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 20:02:05.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d40ab94e714d2c2bb6d6fe4b4a58a7a6a4cac85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 20:19:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4b8277eaf4c4b77b29aaa609dcea492\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/ce31d296cdc04219a21a9c8a56e86548 : FAILURE in 15m 18s","accounts_in_message":[],"_revision_number":10},{"id":"cb7b8653503f64475a372d6513099db6af052921","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 20:45:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1bca5842fb9a8fd40a7d0cba5304809a465f99b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 20:58:15.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f22ffa5af6b4ac5961ed971a25a59b1\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/ff965c77bb7b438db38e9b4d30cf314a : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":10},{"id":"58bccfde30aed0b6ab687a2a5cbd8ecbbb924747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 20:58:56.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"047e15fa0b3444aaa0cd8d2f56617711ba7d1427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:12:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e190936a0e34007ad3d353271a36242\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/1d1ef9ffef3c43339e93bae0e3eadf18 : FAILURE in 13m 02s","accounts_in_message":[],"_revision_number":11},{"id":"72e7aff63df46511b1cadba0305e409c80883e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-11 21:13:50.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a2fb8015c6515156e31730bb296fd7eafc1a26a0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-11 21:32:01.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a7d304063e302d4e8beeb98c449ef6e8dd12c512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:41:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0613fc2f33fa4aecb8334f67fb8f103c\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/925d9d532247474eaedf508b6fbcb0bf : FAILURE in 22m 32s","accounts_in_message":[],"_revision_number":12},{"id":"e56061953e34f042fd85443477c2cf81926a3b15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-11 22:08:40.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"c8abfaf6c4a6351542f92c41ce9fe0f8e2a0a241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 22:42:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5d039253e524c628590fbf11d69d3e2\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/1f3e9e6b036345c18d2f60cbf0ddb496 : FAILURE in 29m 36s","accounts_in_message":[],"_revision_number":13},{"id":"b5c5d1eb46f7aecd4da476faabc6366f470d1df7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-12 05:04:27.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/61/882561/13/check/DellEMC_Unity_Manila_dhss_true/3a13322/DellEMC_Unity_Manila_dhss_true/110 : SUCCESS in 3h 52m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/61/882561/13/check/DellEMC_Unity_Manila_dhss_false/8f334dc/DellEMC_Unity_Manila_dhss_false/120 : SUCCESS in 1h 07m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"586d7222008f475b8c7275652450b21c11fd253b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 18:12:43.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b8a3e55c4e89e53112e9fd8b1eb2e172d1a10696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 18:31:20.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36fecea799b24c76ab63a1bce0ce971a\n\n- manila-tempest-plugin-dummy-no-dhss-rbac-wallaby https://zuul.opendev.org/t/openstack/build/79a17225059845d49b0a034f43eb7189 : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":13},{"id":"4abc4b1223da61e538f5eccec271ff19f738681b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-12 20:21:05.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6bc8b48799455b28a6a9d8f59fb564d87a28ab3c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-22 19:29:37.000000000","message":"Abandoned\n\nThe tests here worked as expected; no plans to ship a wallaby compatible manila-tempest-plugin tag any longer..","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"7077572333e08f6c1d2da75f5b350473cb6505c8","revisions":{"c5dafb9385350feae6279cad09a70416a52da206":{"kind":"REWORK","_number":1,"created":"2023-05-08 12:58:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/882561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/1"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5dafb9385350feae6279cad09a70416a52da206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5dafb9385350feae6279cad09a70416a52da206"}]},"branch":"refs/heads/master"},"99182500b043fc2d95d1a5a4903e992ef504005f":{"kind":"REWORK","_number":2,"created":"2023-05-08 17:20:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/882561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/2"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 17:20:13.000000000","tz":180},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99182500b043fc2d95d1a5a4903e992ef504005f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99182500b043fc2d95d1a5a4903e992ef504005f"}]},"branch":"refs/heads/master"},"2390cc6c66ef8859877781818c9c273c7961facb":{"kind":"REWORK","_number":3,"created":"2023-05-09 13:29:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/882561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/3"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-09 13:29:03.000000000","tz":180},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2390cc6c66ef8859877781818c9c273c7961facb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2390cc6c66ef8859877781818c9c273c7961facb"}]},"branch":"refs/heads/master"},"f6c5538e9916445f07764f08810a355b169dc3e8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-10 11:18:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/61/882561/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/4"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-10 11:18:00.000000000","tz":180},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6c5538e9916445f07764f08810a355b169dc3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6c5538e9916445f07764f08810a355b169dc3e8"}]},"branch":"refs/heads/master"},"58049c8ea34f0ed2f162474265861c2341d1aed1":{"kind":"REWORK","_number":5,"created":"2023-05-10 14:29:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/882561/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/5"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-10 14:29:54.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58049c8ea34f0ed2f162474265861c2341d1aed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58049c8ea34f0ed2f162474265861c2341d1aed1"}]},"branch":"refs/heads/master"},"7a0beb6d2c5657f0b906a7edf0bc5653de107fff":{"kind":"REWORK","_number":6,"created":"2023-05-10 15:28:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/882561/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/6"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-10 15:28:14.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a0beb6d2c5657f0b906a7edf0bc5653de107fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a0beb6d2c5657f0b906a7edf0bc5653de107fff"}]},"branch":"refs/heads/master"},"2eae8882ea0cbf47149412da499000ca4f87f5e6":{"kind":"REWORK","_number":7,"created":"2023-05-10 17:23:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/882561/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/7"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-10 17:23:47.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2eae8882ea0cbf47149412da499000ca4f87f5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2eae8882ea0cbf47149412da499000ca4f87f5e6"}]},"branch":"refs/heads/master"},"c3303e1cf3210e55665d864d2ea558a7ccf411ad":{"kind":"REWORK","_number":8,"created":"2023-05-11 17:25:53.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/882561/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/8"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-05-11 17:25:45.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3303e1cf3210e55665d864d2ea558a7ccf411ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3303e1cf3210e55665d864d2ea558a7ccf411ad"}]},"branch":"refs/heads/master"},"ac52eb8bb8f4fd60eff2548c4cf20a90ebd5a044":{"kind":"REWORK","_number":9,"created":"2023-05-11 17:52:34.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/882561/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/9"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-05-11 17:51:31.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac52eb8bb8f4fd60eff2548c4cf20a90ebd5a044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac52eb8bb8f4fd60eff2548c4cf20a90ebd5a044"}]},"branch":"refs/heads/master"},"638fe691a606dee8fcdbb682c7a96b91c24f0323":{"kind":"REWORK","_number":10,"created":"2023-05-11 20:02:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/882561/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/10"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-05-11 20:02:00.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/638fe691a606dee8fcdbb682c7a96b91c24f0323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/638fe691a606dee8fcdbb682c7a96b91c24f0323"}]},"branch":"refs/heads/master"},"b0a426c411c9a1106285e7c521734ad46183cd71":{"kind":"REWORK","_number":11,"created":"2023-05-11 20:58:56.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/882561/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/11"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-05-11 20:58:52.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0a426c411c9a1106285e7c521734ad46183cd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0a426c411c9a1106285e7c521734ad46183cd71"}]},"branch":"refs/heads/master"},"e2b824affebe6344c7353fb79c0cd65a51a54934":{"kind":"REWORK","_number":12,"created":"2023-05-11 21:13:50.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/882561/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/12"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-05-11 21:13:47.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\n\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2b824affebe6344c7353fb79c0cd65a51a54934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2b824affebe6344c7353fb79c0cd65a51a54934"}]},"branch":"refs/heads/master"},"7077572333e08f6c1d2da75f5b350473cb6505c8":{"kind":"REWORK","_number":13,"created":"2023-05-11 22:08:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/882561/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/882561/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/882561/13"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-05-08 12:57:32.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-11 22:08:40.000000000","tz":0},"subject":"[DNM] Check secure rbac tests in Wallaby","message":"[DNM] Check secure rbac tests in Wallaby\n\nDepends-On: I64b6bac06cbb9f33abac5f42b3f45e021b7a29b4\nDepends-On: I3cdcd2a3c64b38aea0d0a4fb2d4bd96270c06dc8\nChange-Id: Ia2dfd9478649e15499aa92c12ce53d30db3a6e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7077572333e08f6c1d2da75f5b350473cb6505c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7077572333e08f6c1d2da75f5b350473cb6505c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
