)]}'
{"id":"openstack%2Fmanila-tempest-plugin~882624","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6abf7a8eb275c5b29cfe29b3a0f54de76c5bb1e8","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-05-10 13:29:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-17 22:46:57.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-05-17 22:46:57.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-05-10 04:52:13.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I6abf7a8eb275c5b29cfe29b3a0f54de76c5bb1e8","subject":"[CI] Fix cephfs-nfs job by running on centos-stream-9","status":"MERGED","created":"2023-05-09 00:49:53.000000000","updated":"2023-05-17 22:46:57.000000000","submitted":"2023-05-17 22:46:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882624","meta_rev_id":"36610d814684d666ebc17d5b5627b49216e666fa","_number":882624,"virtual_id_number":882624,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 22:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-05-10 13:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-05-17 21:37:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 02:44:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 06:02:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-09 07:18:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-09 23:06:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-05-09 23:06:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-05-09 23:06:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"b8127c23add0a0dace85df530dd5465a20576415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-09 00:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1e3e71882e340b79faa24f224f784f3999d4f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 02:44:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9b073d6d32447e93a6392b71e2e0a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6b1e8288f04b9bb93eb21239e57754 : SUCCESS in 3m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/163fea272fb44aeb839040d411e53e21 : SUCCESS in 1h 15m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e4605f65d4fb4232977ffe605f55a59f : SUCCESS in 1h 53m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a148f550aca94756a3df405c696d7bd8 : SUCCESS in 1h 04m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c197748a44134ddc9be798056a9040bd : FAILURE in 5m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1f659ad557484e99843bac5b23245904 : SUCCESS in 1h 24m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e1b027bc2d91491684c61ecef496ae9f : SUCCESS in 52m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8b23680329f74c6c9dffd505123601ce : SUCCESS in 32m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7530d1b76de349cf81b7a2e5aabf0d6b : SUCCESS in 35m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b45b7b8f058a444883c7d9b9df497891 : SUCCESS in 1h 02m 37s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7eb3829fb9044b4bacadfee427a996eb : SUCCESS in 1h 15m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6786576afbe24738b32fb49b1e70dd95 : SUCCESS in 35m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ea840f6d26141b3ab9d9dd962b5f365 : SUCCESS in 55m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6bb012e7c3cd4f7cb6f77142d6bced04 : SUCCESS in 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"769191d00fb0f318a5b07714d6c5a8953ef9a75c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-09 06:02:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/882624/1/upstream-check/manila-tempest-plugin-ontap-dhss/e2dbded/ : TIMED_OUT in 3h 16m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/882624/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/eb24282/ : SUCCESS in 1h 50m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ede5a3f00310d7bee51b4320acf290bcad89278f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 07:18:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/24/882624/1/check/DellEMC_Unity_Manila_dhss_true/ba46a80/DellEMC_Unity_Manila_dhss_true/90 : SUCCESS in 2h 47m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/24/882624/1/check/DellEMC_Unity_Manila_dhss_false/b4f57c3/DellEMC_Unity_Manila_dhss_false/90 : SUCCESS in 1h 16m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc494903d94b52972fc3c022b12980848d45eb4b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-05-10 04:52:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a01c12f2a5a8d42b5bc529300b95e209b3c5a15e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-10 13:29:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"941466f509e8680ec32dff7551cbd17953d0abe9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-16 23:48:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7bbaf83eaf9c2781d5e4b74c811d6502690f8dc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-17 21:37:49.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b4990c7bef3b0cf8fc9a535895a34bb9733da32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 21:38:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85d3fb17a77147ad53950ad3597a209e183a7408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 22:46:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d602c4d85566431badb08ed87c0819f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbec25c40bc74121a2c34ad99ecb52a8 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b93fbb8e0d04477fb269bf95ab1f1c5f : SUCCESS in 48m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4379516393334b6d805fed43457278f5 : SUCCESS in 28m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff986d296edd42f5b021acb2816db1af : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"36610d814684d666ebc17d5b5627b49216e666fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 22:46:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7afdada46d249a89519e1ecc0977ff606b159f74","revisions":{"7afdada46d249a89519e1ecc0977ff606b159f74":{"kind":"REWORK","_number":1,"created":"2023-05-09 00:49:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/882624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/882624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/882624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/882624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/882624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/882624/1"}}},"commit":{"parents":[{"commit":"ad0db6359c8c51c4521ac6660c8014981b2f1dea","subject":"Merge \"[CI] Add cephadm based native cephfs test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad0db6359c8c51c4521ac6660c8014981b2f1dea"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-09 00:49:31.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-09 00:49:31.000000000","tz":-420},"subject":"[CI] Fix cephfs-nfs job by running on centos-stream-9","message":"[CI] Fix cephfs-nfs job by running on centos-stream-9\n\nPackage based installation of ceph via devstack-plugin-ceph\nis deprecated. Recently, we broke support for Ubuntu 20.04 LTS\n(Focal Fossa) [1] because we\u0027re trying to advance all OpenStack\nCI jobs to Ubuntu 22.04/Jammy Jellyfish or Centos Stream 9. The\ncommunity\u0027s plan is to retire package based installation of ceph\non devstack during the 2023.2/Bobcat release cycle, and move all\nefforts onto \"cephadm\" [2].\n\nWith CephFS/NFS, cephadm deployment works, but seems to require\nmore CI resources than a single test node can afford. So, we\u0027re\nworking in parallel to create a multi-node job [3][4] to ensure more\nstable testing without foregoing coverage. As a temporary workaround,\nwe can run the ceph-nfs job (which doesn\u0027t use cephadm) on\ncentos-9 stream.\n\nUnfortunately, ipv6 testing would have to be turned off because of\nLP bug #1998489. We\u0027ll remedy this separately.\n\n[1] https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/865315\n[2] https://docs.ceph.com/en/quincy/cephadm/index.html\n[3] https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/876747\n[4] https://review.opendev.org/c/openstack/manila/+/869235\n\nChange-Id: I6abf7a8eb275c5b29cfe29b3a0f54de76c5bb1e8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7afdada46d249a89519e1ecc0977ff606b159f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7afdada46d249a89519e1ecc0977ff606b159f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
