)]}'
{"id":"openstack%2Fmanila-tempest-plugin~887389","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270","project":"openstack/manila-tempest-plugin","branch":"master","topic":"multinode","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-13 21:25:29.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2023-09-18 22:01:20.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-09-18 22:01:20.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-09-12 18:16:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-09-12 18:28:34.000000000","reason":"\u003cGERRIT_ACCOUNT_6413\u003e replied on the change","reason_account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-09-11 08:08:24.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270","subject":"[CI] Add multinode job for cephfs/nfs with cephadm","status":"MERGED","created":"2023-06-30 15:39:25.000000000","updated":"2023-09-19 00:05:13.000000000","submitted":"2023-09-18 22:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887389-multinode","meta_rev_id":"7a629f4f2900f67686d8a630240bf797204cdaee","_number":887389,"virtual_id_number":887389,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-18 22:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-13 21:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-09-14 10:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-12 18:28:34.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":2,"date":"2023-09-11 08:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-12 18:28:34.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 15:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-30 19:42:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-30 20:43:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-08 20:32:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 20:51:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-09 16:06:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-08-15 18:54:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-09-11 00:45:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-09-11 00:45:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-09-12 17:10:30.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-09-12 17:21:37.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"a61227a2429666347d80afe778f62d7ef19971f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-06-30 15:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"200ee9430d79a50d260071cdff4d2843bb6e9a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 15:39:54.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/manila-tempest-plugin on branch master.  The error was:\n  \n    found duplicate anchor; first occurrence\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 12, column 23\n  second occurrence\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 507, column 23","accounts_in_message":[],"_revision_number":1},{"id":"fe95fb5eab2b340fd5354a60ed5300d6da1dafc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-06-30 15:47:58.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f5595737f5c9c9d3ba3118ab801c0d833491964f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-06-30 17:08:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6c9e34f18f3bf295e3b99f5160d34bdd9e1f90c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 18:04:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfeca447461492d9510ad3262ddd612\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d1c3fc30ee458a92fd56af8ecaf5b4 : SUCCESS in 2m 45s\n- manila-tempest-plugin-multinode-cephadm https://zuul.opendev.org/t/openstack/build/af8aaa622c9b4257b4e8921f1fe856a0 : FAILURE in 53m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a149e0ac21434fd4293e73429acd51ab8984d121","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-30 19:42:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/3/check/DellEMC_Unity_Manila_dhss_true/8d817db/DellEMC_Unity_Manila_dhss_true/197 : SUCCESS in 2h 34m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/3/check/DellEMC_Unity_Manila_dhss_false/e46fee2/DellEMC_Unity_Manila_dhss_false/211 : SUCCESS in 57m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6cc6f8e2eb3801b4cbb5f52a4df7552ad446e392","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-30 20:43:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/3/upstream-check/manila-tempest-plugin-ontap-dhss/6c167d5/ : RETRY_LIMIT in 8m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/5c2a430/ : SUCCESS in 2h 37m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"cc31039ae4df3b8da4d217de71d5eceed3775ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-08-08 19:29:44.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7381b35c5002d7ff5f4e331eaa74aae14aa2ec55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 20:19:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501246139d02470aa9e97556c92f0053\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01259c196bde4048b09e118cee9c69f3 : SUCCESS in 2m 36s\n- manila-tempest-plugin-multinode-cephadm https://zuul.opendev.org/t/openstack/build/70d1015fc00c4fa3b09c417fc8fe1954 : FAILURE in 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12e329080e9b681ee2b13fee4fc3da29ee5202b4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-08 20:32:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/4/check/DellEMC_PowerFlex_Manila_dhss_false/40e01bb/DellEMC_PowerFlex_Manila_dhss_false/176 : FAILURE in 32m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"1479b466633349595ca52b80c246ba14b32e85df","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-08 20:51:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/4/check/DellEMC_PowerStore_Manila_dhss_false/e6111c8/DellEMC_PowerStore_Manila_dhss_false/57 : FAILURE in 31m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"afd3a2675f52090ad7d447a24d3fc9926824840c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-08 22:12:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/4/upstream-check/manila-tempest-plugin-ontap-dhss/c58875b/ : FAILURE in 1h 51m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/3fbc704/ : FAILURE in 1h 48m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7d3f80062e0269254cc202f612b3bccee14a0a0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-09 01:25:33.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/4/check/DellEMC_Unity_Manila_dhss_true/6d911a5/DellEMC_Unity_Manila_dhss_true/337 : SUCCESS in 3h 00m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/4/check/DellEMC_Unity_Manila_dhss_false/f4ccafd/DellEMC_Unity_Manila_dhss_false/361 : SUCCESS in 1h 17m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"87eebd710d4e0a0c48b16839058ad1af3e0a67f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 16:06:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3cc2a63e18bb03333fab5bfb48b7b14205007bcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-08-11 13:39:00.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d30b8171e59c38f58f8d029e630e2887f6de0dc6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-11 14:11:22.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/5/check/DellEMC_PowerFlex_Manila_dhss_false/120d1d2/DellEMC_PowerFlex_Manila_dhss_false/199 : FAILURE in 32m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"11f00eaed1dbe2b86155e0e3d6be00e0291fd4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 14:48:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d748d723fb234f79a3d17aab7a2d6e6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee150d18b99403c9f9d4ccfab0211b7 : SUCCESS in 2m 43s\n- manila-tempest-plugin-multinode-cephadm https://zuul.opendev.org/t/openstack/build/ce0003c814004623989c7b92c058cc57 : FAILURE in 1h 08m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ccfdca042c4fb820c5aaf1f740ad49bfe1cc9a9a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-11 15:20:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/5/check/DellEMC_PowerStore_Manila_dhss_false/5e9ef7f/DellEMC_PowerStore_Manila_dhss_false/66 : FAILURE in 31m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"3e51db1c23564fa77ff2e2a83380f93541cef823","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-08-11 15:34:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"da92ae874215295fc3c4f36771f34c18710b417d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-11 15:46:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/5/check/DellEMC_PowerFlex_Manila_dhss_false/e5dfc41/DellEMC_PowerFlex_Manila_dhss_false/202 : FAILURE in 31m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"fc87c92030383b94d898af2ac030c28d6c970728","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-11 16:30:25.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/5/check/DellEMC_Unity_Manila_dhss_true/a04c395/DellEMC_Unity_Manila_dhss_true/352 : SUCCESS in 2h 51m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/5/check/DellEMC_Unity_Manila_dhss_false/e7b6831/DellEMC_Unity_Manila_dhss_false/376 : SUCCESS in 1h 03m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"319df47fabdb59159cd91d5a719e95d31e917b87","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-11 17:14:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/5/upstream-check/manila-tempest-plugin-ontap-dhss/eac4ea8/ : FAILURE in 2h 06m 54s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/1a35286/ : FAILURE in 2h 24m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"3977cefa4c6c6ee471873943c94dbfa0d21f4238","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-11 19:25:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ba96f0cd8c8440e70b5c919fc7949a849cb4af2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-11 19:27:36.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8ed2d38ca3c86ee2a10e6d7f27b7fba662e5521a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-11 19:40:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/6/check/DellEMC_PowerFlex_Manila_dhss_false/7cf5784/DellEMC_PowerFlex_Manila_dhss_false/207 : FAILURE in 12m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"2f20b76883a327b9c925b6db6eca52c93a448cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 20:35:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f6fe00d6e14a8795721fb480857218\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db074e1575b454c80f2601f4c3ae06a : SUCCESS in 2m 33s\n- manila-tempest-plugin-multinode-cephadm https://zuul.opendev.org/t/openstack/build/b72468773f554a00ace4b37cf8eca628 : FAILURE in 1h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b74b15fed0d542b09eedadccad2c1791a9aaaa7a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-11 20:47:49.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/6/check/DellEMC_PowerFlex_Manila_dhss_false/94b823b/DellEMC_PowerFlex_Manila_dhss_false/208 : FAILURE in 12m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"fc65a7bd0a2e54ddcdbb2b2efcfbffdedf83a596","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-11 20:48:31.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/6/check/DellEMC_PowerStore_Manila_dhss_false/143c6ed/DellEMC_PowerStore_Manila_dhss_false/70 : FAILURE in 13m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"290b62f645302202878beacc37ce3c831ab2031d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-11 22:19:02.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/6/check/DellEMC_Unity_Manila_dhss_true/47f1907/DellEMC_Unity_Manila_dhss_true/357 : FAILURE in 21m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/6/check/DellEMC_Unity_Manila_dhss_false/df4c3a7/DellEMC_Unity_Manila_dhss_false/381 : FAILURE in 12m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"96c48813ca63b6005a73ac57bb4ab201d59bf950","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-11 22:33:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/6/upstream-check/manila-tempest-plugin-ontap-dhss/c0ec893/ : FAILURE in 1h 56m 53s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/f53b176/ : RETRY_LIMIT in 9m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"34d9aff5c599d3d62647d8dbd15d39c6b6137ec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 18:54:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b40eca98376a905f0e8a53672ba0ac0ca0951642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 18:55:08.000000000","message":"Patch Set 7: Verified-1\n\n(5 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Invalid Ansible variable name \u0027group-vars\u0027 for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-multinode-cephfs-nfs-cephadm\n      description: Test CephFS NFS (DHSS\u003dFalse) in a Multinode devstack env\n      parent: manila-tempest-plugin-multinode-base\n      required-projects:\n        - openstack/devstack-plugin-ceph\n        - openstack/neutron-dynamic-routing\n      vars:\n        tempest_concurrency: 2\n        # turning off some tests due to exhaustion of disk space\n        # https://bugs.launchpad.net/manila/+bug/2009083\n        tempest_exclude_regex: \"\\\n          (^manila_tempest_tests.tests.scenario.test_share_extend.TestShareExtendNFSIPv6.test_create_extend_and_write)|\\\n          (^manila_tempest_tests.tests.scenario.test_share_shrink.TestShareShrinkNFSIPv6.test_create_shrink_and_write)\"\n        devstack_plugins:\n          devstack-plugin-ceph: https://opendev.org/openstack/devstack-plugin-ceph\n          neutron-dynamic-routing: https://opendev.org/openstack/neutron-dynamic-routing\n        devstack_services: *devstack-with-ovs # LP 1940324\n        devstack_localrc:\n          # NOTE(gouthamr): LP#1940324 prevents bgp usage with OVN, use OVS\n          Q_AGENT: openvswitch\n          Q_ML2_TENANT_NETWORK_TYPE: vxlan\n          Q_ML2_PLUGIN_MECHANISM_DRIVERS: openvswitch\n          IP_VERSION: 4\n          SERVICE_IP_VERSION: 4\n          USE_PYTHON3: true\n          MYSQL_REDUCE_MEMORY: True\n          MANILA_USE_DOWNGRADE_MIGRATIONS: false\n          MANILA_ALLOW_NAS_SERVER_PORTS_ON_HOST: true\n          MANILA_DEFAULT_SHARE_TYPE_EXTRA_SPECS: \u0027snapshot_support\u003dTrue create_share_from_snapshot_support\u003dTrue\u0027\n          MANILA_CONFIGURE_DEFAULT_TYPES: true\n          SHARE_DRIVER: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_ENABLED_BACKENDS: cephfsnfs\n          ENABLED_SHARE_PROTOCOLS: NFS\n          MANILA_SERVICE_IMAGE_ENABLED: True\n          MANILA_OPTGROUP_cephfsnfs_share_driver: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_OPTGROUP_cephfsnfs_driver_handles_share_servers: false\n          MANILA_OPTGROUP_cephfsnfs_cephfs_protocol_helper_type: NFS\n          MANILA_OPTGROUP_cephfsnfs_cephfs_conf_path: /etc/ceph/ceph.conf\n          MANILA_OPTGROUP_cephfsnfs_cephfs_auth_id: manila\n          MANILA_OPTGROUP_cephfsnfs_cephfs_nfs_cluster_id: cephfs\n          MANILA_CEPH_DRIVER: cephfsnfs\n          NEUTRON_CREATE_INITIAL_NETWORKS: false\n          MANILA_RESTORE_IPV6_DEFAULT_ROUTE: false\n          CEPHADM_DEV_OSD: true\n          TARGET_DEV_OSD_DIR: /opt/stack\n          CEPH_LOOPBACK_DISK_SIZE: 60GB\n          CEPHADM_DEPLOY: True\n          REMOTE_CEPH: False\n          ENABLE_CEPH_MANILA: True\n          ENABLE_CEPH_NOVA: False\n        devstack_local_conf:\n          test-config:\n            $TEMPEST_CONFIG:\n              share:\n                default_share_type_name: default\n                run_ipv6_tests: true\n                run_share_group_tests: false\n                multitenancy_enabled: false\n                backend_names: CEPHFSNFS\n                enable_protocols: nfs\n                capability_storage_protocol: NFS\n                image_password: manila\n        group-vars:\n          subnode:\n            devstack_localrc:\n              CEPHADM_DEPLOY: True\n              REMOTE_CEPH: True\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 507, column 3","accounts_in_message":[],"_revision_number":7},{"id":"18702cacdccd6c17b196b2f6c385b442816ce8b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 18:56:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"413b51a2bb345404ef6e2c5d51d796777f26612e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-15 19:09:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/8/check/DellEMC_PowerFlex_Manila_dhss_false/4ba8139/DellEMC_PowerFlex_Manila_dhss_false/225 : FAILURE in 12m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"1542209af1b102970759ef944db44529aed966bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-15 19:21:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/8/check/DellEMC_Unity_Manila_dhss_true/13b26d8/DellEMC_Unity_Manila_dhss_true/374 : FAILURE in 24m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/8/check/DellEMC_Unity_Manila_dhss_false/d985fa2/DellEMC_Unity_Manila_dhss_false/399 : FAILURE in 12m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"1554701afbc509207992157c8ff18be67921ad04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:06:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40db3431889a4eb8a9dca12d24099e8a\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/141c2cae65a54952a846b8bbf15cd719 : FAILURE in 1h 03m 59s","accounts_in_message":[],"_revision_number":8},{"id":"437960a2ae1f59596b3d055fde6141c8366945c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 21:46:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"01d935fef0a8ccd125bb60bceaf7468ab5c8d6f0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-15 22:07:38.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/9/check/DellEMC_Unity_Manila_dhss_true/cb6cc81/DellEMC_Unity_Manila_dhss_true/380 : FAILURE in 20m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/9/check/DellEMC_Unity_Manila_dhss_false/16d6240/DellEMC_Unity_Manila_dhss_false/405 : FAILURE in 12m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6ccb25b0d25b727a3d97d2d6ca3fb0ca104bf452","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 22:43:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"49324abe52980d4ecdcbae7cbe421406cca85a57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 23:31:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d98b278349954f36929de57112a67342\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d5c9090c8ada465c944e009999768d60 : FAILURE in 1h 38m 27s","accounts_in_message":[],"_revision_number":9},{"id":"5a90b97d87f2ed5267b473d677be86daaafd76ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-16 05:02:22.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0398add498fdf64e73e8d86afc426145c0196c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 05:02:52.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Invalid Ansible variable name \u0027group-vars\u0027 for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-multinode-base\n      abstract: true\n      description: |\n        Base job for testing multinode with Manila. Manila is enabled in\n        the controller node; and we have an additional compute node.\n      parent: tempest-multinode-full-py3\n      timeout: 10800\n      irrelevant-files: *irrelevant-files\n      required-projects: *manila-tempest-required-projects\n      vars:\n        tox_envlist: all\n        tempest_test_regex: manila_tempest_tests\n        tempest_concurrency: 8\n        tempest_plugins:\n          - manila-tempest-plugin\n        devstack_services:\n          cinder: false\n          s-account: false\n          s-container: false\n          s-object: false\n          s-proxy: false\n          horizon: false\n          tls-proxy: true\n        group-vars:\n          controller:\n            devstack_plugins:\n                manila: https://opendev.org/openstack/manila\n        devstack_localrc:\n          USE_PYTHON3: true\n          MYSQL_REDUCE_MEMORY: True\n          MANILA_USE_DOWNGRADE_MIGRATIONS: false\n          MANILA_INSTALL_TEMPEST_PLUGIN_SYSTEMWIDE: false\n          MANILA_ALLOW_NAS_SERVER_PORTS_ON_HOST: true\n          MANILA_SHARE_MIGRATION_PERIOD_TASK_INTERVAL: 1\n          MANILA_SERVER_MIGRATION_PERIOD_TASK_INTERVAL: 10\n          MANILA_REPLICA_STATE_UPDATE_INTERVAL: 10\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 495, column 3","accounts_in_message":[],"_revision_number":10},{"id":"c2ff970df420182d735909c805bb38790bd824aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-16 05:23:20.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/10/check/DellEMC_Unity_Manila_dhss_true/a71b751/DellEMC_Unity_Manila_dhss_true/384 : FAILURE in 20m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/10/check/DellEMC_Unity_Manila_dhss_false/da0b34d/DellEMC_Unity_Manila_dhss_false/409 : FAILURE in 12m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"ed34dd6af1b92e19d8f77f958151947e0d8a9af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-16 05:49:03.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"c4e8e806aee94cdabadb90c8562a2d62c6ae6348","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-16 06:20:28.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/11/check/DellEMC_Unity_Manila_dhss_true/151bb3c/DellEMC_Unity_Manila_dhss_true/386 : FAILURE in 20m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/11/check/DellEMC_Unity_Manila_dhss_false/1bd699f/DellEMC_Unity_Manila_dhss_false/411 : FAILURE in 12m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"45b3c388aca00568f60fd3ecb6096dc199c71920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 06:43:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63c4ca066ab140c69afebdc286ce21f5\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/48624e4d3f1048f08839a3edcd967686 : FAILURE in 48m 26s","accounts_in_message":[],"_revision_number":11},{"id":"fc7d691d748924ad98a603b7f0eed9512b18324f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-16 18:52:27.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6326bf4de4007605b8626e9daea0870aa6f50544","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-16 19:15:12.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/12/check/DellEMC_Unity_Manila_dhss_true/f6efe1a/DellEMC_Unity_Manila_dhss_true/396 : FAILURE in 22m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/12/check/DellEMC_Unity_Manila_dhss_false/579f3de/DellEMC_Unity_Manila_dhss_false/421 : FAILURE in 12m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"a19a3feb52d13ebcff58aa6555ea1704db8fa545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 19:56:28.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cac92c7e0c114d9a8efdea0dc236164d\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/818d4907e08a406991525062dc3db00c : FAILURE in 56m 31s","accounts_in_message":[],"_revision_number":12},{"id":"cf3cc7493415c8f78daa82b61066c9254937e8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-16 20:07:31.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"d45b7a43d15f5d9e21685a63f269f4be44e1ebb8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-16 20:36:05.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/13/check/DellEMC_Unity_Manila_dhss_true/ebabee7/DellEMC_Unity_Manila_dhss_true/398 : FAILURE in 20m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/13/check/DellEMC_Unity_Manila_dhss_false/65e6b2c/DellEMC_Unity_Manila_dhss_false/423 : FAILURE in 12m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"4db1720640f4d4183d4abfb5ddf1a7d3088bcb38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 21:11:19.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50b3128441ad4887917e8a7afc5baab2\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/31a14a9b64e44ce6bedf4c885fded972 : FAILURE in 1h 02m 24s","accounts_in_message":[],"_revision_number":13},{"id":"c84cf56d70518bc2f6930e83bb6472e92a8a73c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-16 21:33:50.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"9787eacc7c8c8ea02a60ec9423f5138696fd5d92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-16 21:58:10.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/14/check/DellEMC_Unity_Manila_dhss_true/1fe9c44/DellEMC_Unity_Manila_dhss_true/399 : FAILURE in 24m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/14/check/DellEMC_Unity_Manila_dhss_false/c7ea171/DellEMC_Unity_Manila_dhss_false/424 : FAILURE in 12m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"3854696e69b9b605eda56fa0e548124dabb9af7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 22:53:06.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87d8ddf5682b4c818add34e7d92c3738\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6a973b306f214084be5f7caf63436e7e : FAILURE in 1h 11m 22s","accounts_in_message":[],"_revision_number":14},{"id":"2f6fd67b6cd3dd3be13dd82190755260a4a87e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 00:03:56.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"f9095202cd37e5a598319b21e08af03ed2662bd1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 00:33:24.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/15/check/DellEMC_Unity_Manila_dhss_true/9e55035/DellEMC_Unity_Manila_dhss_true/402 : FAILURE in 24m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/15/check/DellEMC_Unity_Manila_dhss_false/f59507f/DellEMC_Unity_Manila_dhss_false/427 : FAILURE in 12m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"5c8e43fc48bee4f3a20c0b62e9831825c4657a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 00:52:12.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc096c4e944e41489e30dd338d89f3d3\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e369b71c31534f0988766525d144f28b : FAILURE in 44m 17s","accounts_in_message":[],"_revision_number":15},{"id":"8941aef4c8e516845a95185120ea6cbd7382f1b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 05:52:39.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"43b4bda4b77f9623cfd07ce385d2c28753c59400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 06:01:59.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38a17e12335a4ceda7576c20dbf34ff5\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/073262738bad44af9f9226216099196e : FAILURE in 7m 57s","accounts_in_message":[],"_revision_number":16},{"id":"170827ad3b559b285875da4e4985e42d9db6c4cd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 06:19:20.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/16/check/DellEMC_Unity_Manila_dhss_true/7cbcd77/DellEMC_Unity_Manila_dhss_true/403 : FAILURE in 26m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/16/check/DellEMC_Unity_Manila_dhss_false/1229c8d/DellEMC_Unity_Manila_dhss_false/428 : FAILURE in 12m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"2f4b691b96ce7ecba0828df609f5d08264c52e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 18:40:26.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"71a0962f5f10fcd7c3ee70d14ef56faca687230c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 19:08:18.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/17/check/DellEMC_Unity_Manila_dhss_true/e64f911/DellEMC_Unity_Manila_dhss_true/410 : FAILURE in 27m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/17/check/DellEMC_Unity_Manila_dhss_false/4d585a3/DellEMC_Unity_Manila_dhss_false/435 : FAILURE in 12m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"12e58e245ec3bd2dbb350690063507301ab275b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:20:27.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/422f42dae49340e8a26f4bdd44ee6f7a\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/be026e25d34a48e598f2c11c63bbe48b : FAILURE in 29m 38s","accounts_in_message":[],"_revision_number":17},{"id":"972d7e5940d54bf65b5092a53690ce8e7cae23ad","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 19:26:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8eb943d310556380a17cfe177afac0408ad3078d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:41:26.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5def0b3160df45c49d44362e0ffa5c02\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6dec698fa4b44031a006f4d69d207e51 : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":17},{"id":"bb8966b2db8be52c6daf215c04ad3749bb026b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 23:12:19.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"4f4991b46d5074921ddd9da1b3b20a0aea574eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 23:51:50.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85b449441f924496bae72dc343e412fd\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/58ad6365d4ca48bca43c5d5057ae1ca5 : RETRY_LIMIT in 19m 18s","accounts_in_message":[],"_revision_number":18},{"id":"d014a12462af71d8a609134bcf1cf636bd55f0c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 23:52:40.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/18/check/DellEMC_Unity_Manila_dhss_true/3b0ca18/DellEMC_Unity_Manila_dhss_true/414 : FAILURE in 24m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/18/check/DellEMC_Unity_Manila_dhss_false/6c9eedd/DellEMC_Unity_Manila_dhss_false/439 : FAILURE in 11m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"55b873241686c494bb689b759e42cd77bdaca3ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-07 18:34:25.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"39f54692c76c55f4e51b4c8500ae0e29b7300b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:55:08.000000000","message":"Patch Set 19: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77abec814a51451db1563f976fd1463b\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dfe8afd0a9a649e9b960abec7bead81f : FAILURE in 19m 49s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":19},{"id":"5c45c70574f556ff479fb06fe4ef57503be2b5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-07 19:05:38.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"33df59062ae25a602ee3c39ddb2012423aebd241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 19:28:12.000000000","message":"Patch Set 20: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49233c7229e44b49b16a6bc1c5775c57\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c05fe2a356654cf69fc7b51381a922de : FAILURE in 21m 08s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":20},{"id":"61bc6862497bb8596aa3ef44101c6af4e66595b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 19:38:20.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/20/check/DellEMC_PowerFlex_Manila_dhss_false/1d4c10d/DellEMC_PowerFlex_Manila_dhss_false/333 : SUCCESS in 32m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"8ee70f79093d592f766cc7c288416231d9a96fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-07 20:09:59.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"88775c2562589ab3b49c757588588df48cf5c5aa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 20:41:20.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/21/check/DellEMC_PowerFlex_Manila_dhss_false/d5586b8/DellEMC_PowerFlex_Manila_dhss_false/334 : SUCCESS in 31m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"389cda5ed32a6076af977dd1fb8c807daeafcb0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 20:50:25.000000000","message":"Patch Set 21: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd517a5f708f4cafaf80c949ef3b04e9\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e3f29019e8dc4919882c09be0d533713 : FAILURE in 39m 04s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":21},{"id":"4ed7145b4568a6f3e0417b0b8df9245adf3e769a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-07 23:17:25.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"85ee331e2495026ac00cc7ca347da64899f69e27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 23:49:47.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/22/check/DellEMC_PowerFlex_Manila_dhss_false/125bfd1/DellEMC_PowerFlex_Manila_dhss_false/335 : SUCCESS in 32m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"b9418211f2c8afc3c5b1136f1ae2b0da3d17988c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:06:15.000000000","message":"Patch Set 22: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7f0ac4e57fd4a7ab016974259f8b595\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bec70f48459c40f096c185c9b8556b96 : FAILURE in 39m 32s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":22},{"id":"992b1229ea83d89a87369a8c2a8491d6e9a2f3f1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-08 00:20:25.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"923aee21c252f39a2671213eae95643ef1af8e4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-08 01:01:58.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/22/check/DellEMC_Unity_Manila_dhss_true/0d3a2af/DellEMC_Unity_Manila_dhss_true/558 : FAILURE in 51m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/22/check/DellEMC_Unity_Manila_dhss_false/6c7165b/DellEMC_Unity_Manila_dhss_false/580 : ABORTED in 1h 44m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"465e752d6bfd1689c04d3f63901d6f17bcc07d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 01:22:52.000000000","message":"Patch Set 22:\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d27c290607e4ee2b7dea953b32b3fb7\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7e196fe565c842fd8d713bb3cd9addcc : FAILURE in 56m 51s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":22},{"id":"1599f2c799d537bdda01b66d87cb17c7affee2ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-08 16:32:18.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"7a5326c5ffad17537281d7b179f6a274fcb42e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 16:41:39.000000000","message":"Patch Set 23: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/876747 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"8858d899ddc664fabda87320914c91c36b76e712","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-08 16:41:40.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"88753d310f1759416b91cac2577cbcf0e6418fb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-08 16:41:40.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"17557fbb985e61b592b898bbe7af66ab9f0d49b7","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-08 17:25:50.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c3c047996de2abd7111e5a56230b3fc8d5584b24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 18:38:52.000000000","message":"Patch Set 23: Verified+1\n\n(7 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3db12c55edb493db1d3e0154b4dcf08\n\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4db7ec312a0f42df8fd7781109bd4ba0 : SUCCESS in 1h 11m 35s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":23},{"id":"90d64931bbcf2feaf8ef35ff942e62d6d8717df2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-08 19:06:49.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/23/check/DellEMC_PowerFlex_Manila_dhss_false/39c51e8/DellEMC_PowerFlex_Manila_dhss_false/343 : SUCCESS in 27m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"a9005a78f0487e374bd046ed8e7d3acb482bd99f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-08 19:17:17.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/23/check/DellEMC_PowerStore_Manila_dhss_false/bfcca1c/DellEMC_PowerStore_Manila_dhss_false/145 : SUCCESS in 38m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":23},{"id":"3e2f5513a8197ab7b85f8cc3842a019cec38e70e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-08 21:26:15.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/23/check/DellEMC_Unity_Manila_dhss_true/c7c6a80/DellEMC_Unity_Manila_dhss_true/563 : FAILURE in 55m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/23/check/DellEMC_Unity_Manila_dhss_false/8ca63e0/DellEMC_Unity_Manila_dhss_false/585 : FAILURE in 2h 47m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"c99a41e019d25e11edc6f2190fa584e6b1c77ad0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-08 21:56:27.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/23/upstream-check/manila-tempest-plugin-ontap-dhss/c9828f2/ : TIMED_OUT in 3h 16m 13s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/23/upstream-check/manila-tempest-plugin-ontap-no-dhss/eb19e42/ : SUCCESS in 2h 00m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"c7163019f03623c1ac15d47fec1d0a7969d179ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-09 16:15:13.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"683b21a70a0927e11df238d05286b4676da01dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-09 16:16:00.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"36bb00c0a870a33ab4ef2c1833412b0d87802cbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-09 16:18:31.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"826091db5e0c37d03cf046ecdd3338d251825000","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-09 16:44:12.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_PowerFlex_Manila_dhss_false/c1d45c1/DellEMC_PowerFlex_Manila_dhss_false/347 : SUCCESS in 25m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"6de7be9edcb641dc754d637adba24a0d2eedde1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 18:43:03.000000000","message":"Patch Set 26: Verified+1\n\n(7 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63ca4a91d86453cb220401358d2b582\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95812cd9e1124279b4567107583a08e2 : SUCCESS in 5m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5e087f485fb242a99bb993cee96e4503 : SUCCESS in 42m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a0ce78d27a994a80b4821be7c4cd4f52 : FAILURE in 51m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a73a7820c86640b3af948b524afd9779 : FAILURE in 31m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3d4f329398514ae5af400e4b2edd0f2e : SUCCESS in 44m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6e4552521207454885d0faae507b1e9b : SUCCESS in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5773fe1efafc40c08f47e08db68f511b : FAILURE in 59m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/870c65cf8dc842f5ada0997e0a61aee4 : SUCCESS in 24m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe606f1be39846b48550a4ff70d9d6df : SUCCESS in 27m 06s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7b4f25ceadff418f968dbd3d1983c3b3 : SUCCESS in 29m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9147555880e94cc680beab5e1c64f717 : SUCCESS in 39m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a81e6ff653dc4842b55a3ae167a3dae4 : SUCCESS in 21m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78a5e1f4b318499db5c8e39006d279d6 : SUCCESS in 1h 06m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6b8fb2f65b6d4e1b987e7d0f90213da0 : FAILURE in 22m 38s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":26},{"id":"428addc68f94510133c2d61d59c3a813878e77d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-09 19:06:05.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_Unity_Manila_dhss_true/3e97597/DellEMC_Unity_Manila_dhss_true/567 : FAILURE in 58m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_Unity_Manila_dhss_false/abba38a/DellEMC_Unity_Manila_dhss_false/589 : FAILURE in 2h 47m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"65d1f8c4b684d062e66981c2c8e128a5f6c6527d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-09 19:16:53.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_PowerFlex_Manila_dhss_false/1fdf924/DellEMC_PowerFlex_Manila_dhss_false/348 : SUCCESS in 33m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"5c6d3a70b3b1cc14f0ea488c22bfbbae545941d9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-09 19:20:47.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_PowerStore_Manila_dhss_false/a0dcfc1/DellEMC_PowerStore_Manila_dhss_false/146 : SUCCESS in 37m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":26},{"id":"6d928397043e36ccc2ac15c09a56db1c7f3c22bc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-09 21:04:05.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/26/upstream-check/manila-tempest-plugin-ontap-dhss/d5636e8/ : RETRY_LIMIT in 8m 27s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/26/upstream-check/manila-tempest-plugin-ontap-no-dhss/25532df/ : SUCCESS in 2h 19m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"ff15017112d451a87193ec3e2d26563409e6093e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-11 08:08:24.000000000","message":"Patch Set 26: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"cf0d20bf111611ca64892419ebd967993bd1cbe8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-11 14:12:43.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"5c3427bfd1dd5921c45d111934f67a6fd1ed7d06","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2023-09-12 17:21:37.000000000","message":"Patch Set 26: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":26},{"id":"3552b3899d35372989b3810af7f60313fed8e9b3","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-09-12 18:16:03.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"6e635c0d446363dc536d0990636ade8340f07227","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 18:26:07.000000000","message":"Patch Set 26:\n\n(4 comments)","accounts_in_message":[],"_revision_number":26},{"id":"90a3369cdc1b529102e209f213c56d24d92f79a8","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2023-09-12 18:28:34.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"072d9f3fff97f491497125ceff22605e11f7b8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:28:46.000000000","message":"Patch Set 26: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c29b6d50e7dd4fe9ae21cee2ea5a1b10\n\n\nWarning:\n  Change 876747 in project openstack/devstack-plugin-ceph does not share a change queue with 887389 in project openstack/manila-tempest-plugin","accounts_in_message":[],"_revision_number":26},{"id":"ca72022840766f826c6cb1c42205dc76305bdb5f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 18:43:48.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"2f707933231f1264887693299f031ca7f1ca771d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-13 21:25:29.000000000","message":"Patch Set 26: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e2039d0c0c31dfb92d88c5d511f611d0a0103f9e","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-09-14 10:18:10.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"840a735d6b4584b631874278bc6c09e2a2362513","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-18 19:09:40.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"26340d4d50ec9a94329c24960f8f153cec16b3cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 21:03:37.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ddc3e133ee442e7b3f3a82ea7a55257\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0398b194ecd74af7b523f96a6a8fbc32 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7253eabdfd2f44088e9c5948fd468d42 : SUCCESS in 38m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4b8ec987e420459ba8cf829e152dcb87 : SUCCESS in 1h 52m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/80690a0666f24ceeb4e134e2fb68120d : SUCCESS in 1h 02m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bd9c443c33524b9491f0624438eea08b : SUCCESS in 54m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f1db8597538d42c486f5a3586dc6a5ed : SUCCESS in 1h 41m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/454dc39dde484cbf9f6e0c9ae026862d : SUCCESS in 1h 07m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6ac526a3ccf24e04993ddaedd73f98df : SUCCESS in 23m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f33a6b7f8f98400f9407a70463fc8ecf : SUCCESS in 37m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/89a0ff64780645cd9cee352678f88686 : SUCCESS in 46m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0c40ef1eefd44d71a7621f0b45605ff5 : SUCCESS in 47m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3378a486ac56427c871fe665510627ae : SUCCESS in 39m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7182507b3364447081f9495a972f67da : SUCCESS in 59m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/55c4050107ae415abf4d932699471b87 : FAILURE in 24m 21s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"a43501e14685f80ba7801693980ec1157a723a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 21:04:14.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"5b523645227aea4e1a12bacc29e0d2e57b015830","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-18 21:34:26.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_PowerFlex_Manila_dhss_false/d8fc3f0/DellEMC_PowerFlex_Manila_dhss_false/404 : SUCCESS in 30m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"0d0673dcc54a2ea2fed23f520847932994a9393e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-18 21:41:57.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/89/887389/26/check/DellEMC_PowerStore_Manila_dhss_false/bad51b0/DellEMC_PowerStore_Manila_dhss_false/173 : SUCCESS in 38m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":26},{"id":"45818714402de9894b9ac29170daab7d74b437d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 22:01:19.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e197de8b17d4404db80c31c5858935b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fb564d41834c719a11b9fb0c57c427 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46663d3189ee40acb2babd7362f5dd84 : SUCCESS in 48m 31s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cb377d5869e64fc18bc6a630623caf5b : SUCCESS in 40m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ba1f571c3a77425d8dbe10f2601825cd : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":26},{"id":"74b9120385562d6486c7c89b8c9d7ed8fdb5b127","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 22:01:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"7a629f4f2900f67686d8a630240bf797204cdaee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-19 00:05:13.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/26/upstream-check/manila-tempest-plugin-ontap-dhss/8e308ab/ : FAILURE in 3h 00m 46s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/887389/26/upstream-check/manila-tempest-plugin-ontap-no-dhss/22cd2b7/ : FAILURE in 2h 49m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"1bab3433f080a6688cb790f0eb916fb82ce49afe","revisions":{"3b500dd2de09fa48f156f234583bc032cfdafd41":{"kind":"REWORK","_number":1,"created":"2023-06-30 15:39:25.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/89/887389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/1"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 15:38:56.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b500dd2de09fa48f156f234583bc032cfdafd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b500dd2de09fa48f156f234583bc032cfdafd41"}]},"branch":"refs/heads/master"},"8d822f5189aa4172e3c039bc956643987d250702":{"kind":"REWORK","_number":2,"created":"2023-06-30 15:47:58.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/89/887389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/2"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2023-06-30 15:47:58.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d822f5189aa4172e3c039bc956643987d250702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d822f5189aa4172e3c039bc956643987d250702"}]},"branch":"refs/heads/master"},"5b2f088dc38d57ec40214c12c394a9e94e444958":{"kind":"REWORK","_number":3,"created":"2023-06-30 17:08:06.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/89/887389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/3"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2023-06-30 17:07:57.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b2f088dc38d57ec40214c12c394a9e94e444958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b2f088dc38d57ec40214c12c394a9e94e444958"}]},"branch":"refs/heads/master"},"628f300a7acae65c137ee882ef46fa25e8f12ae3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-08-08 19:29:44.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/89/887389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/4"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2023-08-08 19:29:44.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/628f300a7acae65c137ee882ef46fa25e8f12ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/628f300a7acae65c137ee882ef46fa25e8f12ae3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3b02894ed220f468e61139e69e51f82a4a989db3":{"kind":"REWORK","_number":5,"created":"2023-08-11 13:39:00.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/89/887389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/5"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2023-08-11 13:39:00.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b02894ed220f468e61139e69e51f82a4a989db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b02894ed220f468e61139e69e51f82a4a989db3"}]},"branch":"refs/heads/master"},"b97c47e8a22e1d73135df83186acfba7b3402420":{"kind":"REWORK","_number":6,"created":"2023-08-11 19:27:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/6"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-11 19:27:36.000000000","tz":0},"subject":"[WIP] Add multinode job","message":"[WIP] Add multinode job\n\nCreates multinode base job with cephadm\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b97c47e8a22e1d73135df83186acfba7b3402420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b97c47e8a22e1d73135df83186acfba7b3402420"}]},"branch":"refs/heads/master"},"aaba33bc097ed1cc083368251dfa5a37e3d5be1d":{"kind":"REWORK","_number":7,"created":"2023-08-15 18:54:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/7"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-15 18:54:13.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aaba33bc097ed1cc083368251dfa5a37e3d5be1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aaba33bc097ed1cc083368251dfa5a37e3d5be1d"}]},"branch":"refs/heads/master"},"4226e472d67e0d076eaece0bb7e075ec6bc60086":{"kind":"REWORK","_number":8,"created":"2023-08-15 18:56:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/8"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-15 18:56:08.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4226e472d67e0d076eaece0bb7e075ec6bc60086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4226e472d67e0d076eaece0bb7e075ec6bc60086"}]},"branch":"refs/heads/master"},"40233313e0de9c7df892a23d08d43e4f0a98a5e7":{"kind":"REWORK","_number":9,"created":"2023-08-15 21:46:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/9"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-15 21:46:44.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40233313e0de9c7df892a23d08d43e4f0a98a5e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40233313e0de9c7df892a23d08d43e4f0a98a5e7"}]},"branch":"refs/heads/master"},"10ebd7c1e5aec45350a16925ca076622c97c7a0c":{"kind":"REWORK","_number":10,"created":"2023-08-16 05:02:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/10"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-16 05:01:56.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10ebd7c1e5aec45350a16925ca076622c97c7a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10ebd7c1e5aec45350a16925ca076622c97c7a0c"}]},"branch":"refs/heads/master"},"66b84d680c6024fc42f6ceb8ec7a60c47da9c91d":{"kind":"REWORK","_number":11,"created":"2023-08-16 05:49:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/11"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-16 05:48:10.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66b84d680c6024fc42f6ceb8ec7a60c47da9c91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66b84d680c6024fc42f6ceb8ec7a60c47da9c91d"}]},"branch":"refs/heads/master"},"e4a1270b65ce352a1a4d2aa9150c305583a2568e":{"kind":"REWORK","_number":12,"created":"2023-08-16 18:52:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/12"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-16 18:52:14.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1270b65ce352a1a4d2aa9150c305583a2568e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1270b65ce352a1a4d2aa9150c305583a2568e"}]},"branch":"refs/heads/master"},"b68b043951d6c295f853bc9ef942099ebeb32233":{"kind":"REWORK","_number":13,"created":"2023-08-16 20:07:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/13"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-16 20:07:23.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b68b043951d6c295f853bc9ef942099ebeb32233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b68b043951d6c295f853bc9ef942099ebeb32233"}]},"branch":"refs/heads/master"},"cc30450b80cf1dac1d0ded2fb28547096f8f4d47":{"kind":"REWORK","_number":14,"created":"2023-08-16 21:33:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/14"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-16 21:33:44.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc30450b80cf1dac1d0ded2fb28547096f8f4d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc30450b80cf1dac1d0ded2fb28547096f8f4d47"}]},"branch":"refs/heads/master"},"f7f57463c5d5fc711de562b1e7397f2ecea18207":{"kind":"REWORK","_number":15,"created":"2023-08-17 00:03:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/15"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-17 00:03:45.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7f57463c5d5fc711de562b1e7397f2ecea18207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7f57463c5d5fc711de562b1e7397f2ecea18207"}]},"branch":"refs/heads/master"},"3e56c539614379a203a2c1813910aaa4d2d15097":{"kind":"REWORK","_number":16,"created":"2023-08-17 05:52:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/16"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-17 05:52:25.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3e56c539614379a203a2c1813910aaa4d2d15097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3e56c539614379a203a2c1813910aaa4d2d15097"}]},"branch":"refs/heads/master"},"dc277988e2fda15d857547ea04460ae447eaa43d":{"kind":"REWORK","_number":17,"created":"2023-08-17 18:40:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/17"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-17 18:40:10.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc277988e2fda15d857547ea04460ae447eaa43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc277988e2fda15d857547ea04460ae447eaa43d"}]},"branch":"refs/heads/master"},"d0cc51e689c7d092600e250d769edab01c8a71d2":{"kind":"REWORK","_number":18,"created":"2023-08-17 23:12:19.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/18"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-17 23:12:07.000000000","tz":-420},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0cc51e689c7d092600e250d769edab01c8a71d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0cc51e689c7d092600e250d769edab01c8a71d2"}]},"branch":"refs/heads/master"},"30ca62d3b9dcc985230449ac9753c2adc068181b":{"kind":"REWORK","_number":19,"created":"2023-09-07 18:34:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/19"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-07 18:34:25.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30ca62d3b9dcc985230449ac9753c2adc068181b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30ca62d3b9dcc985230449ac9753c2adc068181b"}]},"branch":"refs/heads/master"},"e0b3cbbdadbf232d5aaa4c570f4f8f530fe7a0f3":{"kind":"REWORK","_number":20,"created":"2023-09-07 19:05:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/20"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-07 19:05:38.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0b3cbbdadbf232d5aaa4c570f4f8f530fe7a0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0b3cbbdadbf232d5aaa4c570f4f8f530fe7a0f3"}]},"branch":"refs/heads/master"},"3a64e8195781b48ee5b45d778b88c70f4064f972":{"kind":"REWORK","_number":21,"created":"2023-09-07 20:09:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/21"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-07 20:09:59.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a64e8195781b48ee5b45d778b88c70f4064f972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a64e8195781b48ee5b45d778b88c70f4064f972"}]},"branch":"refs/heads/master"},"55fdb28c96eecc4d3eb1cec6b464ea9076dbdffa":{"kind":"REWORK","_number":22,"created":"2023-09-07 23:17:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/22"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-07 23:17:25.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55fdb28c96eecc4d3eb1cec6b464ea9076dbdffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55fdb28c96eecc4d3eb1cec6b464ea9076dbdffa"}]},"branch":"refs/heads/master"},"7f23212c57b659479c851fc1d74209471e022e5a":{"kind":"REWORK","_number":23,"created":"2023-09-08 16:32:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/23"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-08 16:32:18.000000000","tz":0},"subject":"[WIP] Add multinode job for cephfs/nfs","message":"[WIP] Add multinode job for cephfs/nfs\n\nA multinode job creates a controller node\nthat runs manila; and an additional compute\nnode to spread the compute VM workload around.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f23212c57b659479c851fc1d74209471e022e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f23212c57b659479c851fc1d74209471e022e5a"}]},"branch":"refs/heads/master"},"15df7e23681e707c9820ac2c63b86cc9e7f11b86":{"kind":"REWORK","_number":24,"created":"2023-09-09 16:15:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/24"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-09 16:14:37.000000000","tz":-420},"subject":"[CI] Add multinode job for cephfs/nfs with cephadm","message":"[CI] Add multinode job for cephfs/nfs with cephadm\n\nA multinode job creates a controller node\nthat runs manila with other OpenStack components;\nand an additional compute node to spread the\ncompute VM workload around.\n\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm`\nreplaces the existing `manila-tempest-plugin-cephfs-nfs`\nin the check queue for this repository. These jobs\nare different. `manila-tempest-plugin-cephfs-nfs`\nuses devstack-plugin-ceph to install ceph and nfs-ganesha\nas standalone components on the devstack host, whereas\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm` uses\ncephadm to deploy ceph, and it includes the deployment\nof an nfs service. Manila\u0027s ceph driver communicates with\ncephadm deployed nfs via ceph-mgr APIs, and with standalone\nnfs-ganesha service via DBUS APIs. Ideally we\u0027d run\nboth jobs all the time, however, since these jobs will\nrun individually against the manila repository, we\u0027ll just\npick the newer one here.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/15df7e23681e707c9820ac2c63b86cc9e7f11b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/15df7e23681e707c9820ac2c63b86cc9e7f11b86"}]},"branch":"refs/heads/master"},"c4c6e3671fded40883cc313f26ff912a715b5569":{"kind":"REWORK","_number":25,"created":"2023-09-09 16:16:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/25"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-09 16:15:54.000000000","tz":-420},"subject":"[CI] Add multinode job for cephfs/nfs with cephadm","message":"[CI] Add multinode job for cephfs/nfs with cephadm\n\nA multinode job creates a controller node\nthat runs manila with other OpenStack components;\nand an additional compute node to spread the\ncompute VM workload around.\n\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm`\nreplaces the existing `manila-tempest-plugin-cephfs-nfs`\nin the check queue for this repository. These jobs\nare different. `manila-tempest-plugin-cephfs-nfs`\nuses devstack-plugin-ceph to install ceph and nfs-ganesha\nas standalone components on the devstack host, whereas\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm` uses\ncephadm to deploy ceph, and it includes the deployment\nof an nfs service. Manila\u0027s ceph driver communicates with\ncephadm deployed nfs via ceph-mgr APIs, and with standalone\nnfs-ganesha service via DBUS APIs. Ideally we\u0027d run\nboth jobs all the time, however, since these jobs will\nrun individually against the manila repository, we\u0027ll just\npick the newer one here.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c4c6e3671fded40883cc313f26ff912a715b5569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c4c6e3671fded40883cc313f26ff912a715b5569"}]},"branch":"refs/heads/master"},"1bab3433f080a6688cb790f0eb916fb82ce49afe":{"kind":"REWORK","_number":26,"created":"2023-09-09 16:18:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/887389/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/887389/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/887389/26"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2023-06-30 14:50:11.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-09 16:18:25.000000000","tz":-420},"subject":"[CI] Add multinode job for cephfs/nfs with cephadm","message":"[CI] Add multinode job for cephfs/nfs with cephadm\n\nA multinode job creates a controller node\nthat runs manila with other OpenStack components;\nand an additional compute node to spread the\ncompute VM workload around.\n\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm`\nreplaces the existing `manila-tempest-plugin-cephfs-nfs`\nin the check queue for this repository. These jobs\nare different. `manila-tempest-plugin-cephfs-nfs`\nuses devstack-plugin-ceph to install ceph and nfs-ganesha\nas standalone components on the devstack host, whereas\n`manila-tempest-plugin-multinode-cephfs-nfs-cephadm` uses\ncephadm to deploy ceph, and it includes the deployment\nof an nfs service. Manila\u0027s ceph driver communicates with\ncephadm deployed nfs via ceph-mgr APIs, and with standalone\nnfs-ganesha service via DBUS APIs. Ideally we\u0027d run\nboth jobs all the time, however, since these jobs will\nrun individually against the manila repository, we\u0027ll just\npick the newer one here.\n\nDepends-On: I84249ae268dfe00a112c67e5170b679acb318a25\nChange-Id: I5162815b66d3f3e8cf8c1e246b61b0ea06c1a270\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bab3433f080a6688cb790f0eb916fb82ce49afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bab3433f080a6688cb790f0eb916fb82ce49afe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
