)]}'
{"id":"openstack%2Fmanila-tempest-plugin~888618","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idf71e236b1b8a2558bb4ad3de1018fa33b41877f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/allow-locking-shares-against-deletion","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-08-17 12:43:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2023-08-28 16:07:56.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-08-28 16:07:56.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-08-28 16:07:56.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-08-23 07:54:48.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"Idf71e236b1b8a2558bb4ad3de1018fa33b41877f","subject":"Test resource locks","status":"MERGED","created":"2023-07-14 19:48:17.000000000","updated":"2023-08-28 17:28:13.000000000","submitted":"2023-08-28 16:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888618-bp/allow-locking-shares-against-deletion","meta_rev_id":"201b173f76f0752ec6f233dfd6b88708a03aa780","_number":888618,"virtual_id_number":888618,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 16:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-08-25 13:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-23 07:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-08-25 13:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 20:37:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-14 21:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-15 13:26:49.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-24 16:56:59.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-07-24 17:26:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-26 22:55:45.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-17 12:43:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2023-08-17 12:43:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-08-17 12:43:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-08-25 13:28:57.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"f594caf2ea76aed33737c144051e1516e35ddda7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-14 19:48:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9352a28b95b9cb441a7e2f6d832c63bb471a15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-14 19:55:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b94963656329d7ea46916ae3265af5f75b62edd3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-14 20:37:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/2/check/DellEMC_Unity_Manila_dhss_true/fca0f5b/DellEMC_Unity_Manila_dhss_true/238 : FAILURE in 41m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/2/check/DellEMC_Unity_Manila_dhss_false/aac57e5/DellEMC_Unity_Manila_dhss_false/254 : FAILURE in 30m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d0a19b183dfadcb8a2fef42ddb3fc5add446baed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 21:12:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29257d618a1641bdba3060cf30c4ffa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e7ca7d8fdd4221ae05f074cae266d2 : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/428218765ac44042a861b34b5829be85 : FAILURE in 22m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4b779863cc11452fa9fe5f6cb21dcd96 : FAILURE in 38m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/84ae8483b538464bac7f716128f342d7 : FAILURE in 36m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1f408317c9b94b54943c9911e5255cb7 : FAILURE in 1h 07m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b2d4d2f6a81045afbef61fdc0aee4f27 : FAILURE in 24m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8cebf67e7e7149bb80df8cc20fdee2aa : FAILURE in 17m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4fa71f77853749778df95e69910bf405 : FAILURE in 23m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f0a7742c75004fb0ad53a8abb82bfcc9 : FAILURE in 28m 20s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6c65721cccf04f369333029201ebda34 : SUCCESS in 1h 15m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/29036c5b054743e2931ff86f10409991 : SUCCESS in 1h 05m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/52cca15af1094379be25cb5a3cb32d40 : SUCCESS in 23m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/eff4a992273045339c432e0e1e1b6826 : FAILURE in 18m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6991783e0dbf4d328d51587aff7e433b : FAILURE in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52bc4651d0067596966af4ab5f4fca2dea68934d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-15 00:57:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"457b8c4cf56417f1279611a022cff204feda952a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-15 01:22:34.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"fb44d194613b9b35cc4a97135b02ce58432c74b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 03:00:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9663cfbef9148e8815bf48ddd1b7e4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a28473a0c642b98a786fb6f4e22fa8 : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ae16503e5e86441ab6f5a075faf0e698 : SUCCESS in 1h 02m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/654a64738c1a4177b6a0a6f94f4d9136 : FAILURE in 31m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb7267bdf8d147b19d3ffab7047633f7 : FAILURE in 27m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d6afe5bee3c14864ae82e66408abf86d : SUCCESS in 47m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6cf8cda486e24d81a02d688e57ac391e : SUCCESS in 1h 25m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d847dbbcd6354c228109ad075f17a5fd : FAILURE in 28m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8ef7912cad549aba39e941a0e0e76c2 : SUCCESS in 38m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5ad80d8c3bcb4be2b7d3e966a9d91c6c : SUCCESS in 40m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/15db1bb8424348f1a4fd2352919c0bb5 : SUCCESS in 1h 32m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ca48d1e736d4422a937652d42100d8c5 : FAILURE in 1h 19m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/93119e7344ce48f2b96632bb8bdfe00c : SUCCESS in 35m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5ca8d4f6fd9f496298597a7ce2898730 : FAILURE in 29m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/75aedb606274409c8fc755fff3aad07a : SUCCESS in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72034f1ee58b19075df89db39971a56f52b33ff2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-15 09:31:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c53d15f65e3729c89240bf41216c3593f46fb531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 11:00:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab6936db2e64c7084bba97d0ea19e21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334cceb084554ed4a081b7cc77369ad5 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dfd29f70391044e99057b4d90d29c5ae : SUCCESS in 1h 23m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c4ec4aa05acd453aa0929580db465973 : FAILURE in 28m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6a9740fae99148cda96ca857403e0fb6 : FAILURE in 42m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/95f840f546db4c4596cf40f1819ba333 : SUCCESS in 1h 08m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f35735f12b69425fa29808a0343a7bba : SUCCESS in 50m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4fcb5f8614c545308fd73eb7fea6df10 : FAILURE in 14m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1a0f710074746a4a2e8c3ca53e8eacf : SUCCESS in 40m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c7b1430afa4246acabc9236bf3456331 : SUCCESS in 28m 22s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c18d49acd1fa4225beecbc3861b78715 : SUCCESS in 1h 21m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/418aa341220d4518b4a66e40c4ba72fa : SUCCESS in 47m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d4bdaea6314447de809d52960ba0e625 : SUCCESS in 20m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f4c9e54e901e4cabaa795b1c0310e138 : FAILURE in 15m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f4027fc5078244a2955101b985d9eb58 : SUCCESS in 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b6fba6e75d009736d6d5d3df3fc4f798134494d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-15 13:26:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/3/upstream-check/manila-tempest-plugin-ontap-dhss/4ffa8f3/ : FAILURE in 2h 13m 47s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/41f344b/ : SUCCESS in 2h 24m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"24b52e04d0690ce334ef7092d47df9a2126e8708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-24 16:55:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2ba38ee56670346a453dc367161d510b35318fb0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-24 16:56:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab9e3c7e385cc58d3c4c2a8c950822d9f9b60564","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-24 17:26:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/4/check/DellEMC_PowerFlex_Manila_dhss_false/22e83f8/DellEMC_PowerFlex_Manila_dhss_false/91 : FAILURE in 30m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"42f6cc6cfbf22063a7cd2ea6756130b97291f767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 18:56:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d04899bc31e44928816d17f0b508db3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a53c851c6a54928a5edc395a53a4615 : SUCCESS in 3m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3f20bee63f674f8eaf3865afb4f1f203 : SUCCESS in 1h 25m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9aed55b44a9049009434b753b4801b35 : FAILURE in 55m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb0a7ce4aac1426da1c17c141ab78ca7 : FAILURE in 31m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8b6311eebb6942cd92c18bf21d7a2633 : SUCCESS in 1h 35m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0ac7e1a16ba14d3b9e929cd9cf8a73cc : SUCCESS in 1h 48m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/739c6faf78db4a83865439f327a2c8a5 : FAILURE in 1h 46m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a84125219f5420da2aa93e26f6668a3 : SUCCESS in 50m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/25eff4751d444e02b1a17a47138ee6a4 : FAILURE in 31m 04s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6532fb757db24110a32a11579be63c40 : SUCCESS in 1h 43m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e9f0bb02e22a43d796f9f88233aa7702 : SUCCESS in 1h 11m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/76e759ee472f4eba91247e17cb2a41f9 : SUCCESS in 47m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fe0414d91da54d7f962e2ee605905754 : FAILURE in 1h 11m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c1585c389f2f4aee8fee26a014af9f0e : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"569cdebb691aef11d9f2a2f41bef6e467214d4da","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-26 06:10:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/4/check/DellEMC_Unity_Manila_dhss_true/dcb9b09/DellEMC_Unity_Manila_dhss_true/270 : SUCCESS in 3h 00m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/4/check/DellEMC_Unity_Manila_dhss_false/fbc9b8f/DellEMC_Unity_Manila_dhss_false/289 : SUCCESS in 1h 25m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"045eac7c3e93aa6c494db0910b33ea6c7275e47b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-26 18:52:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3bbd761e097e600776fb5383f9186c6b019d25f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 21:00:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1757cafb12284d3b8287a51816d98242\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9891d5ceab27485487d71a3e6242ceec : SUCCESS in 3m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51b0a903377c442f8334e5c61cf7e263 : SUCCESS in 1h 13m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d5361ab0bfda4315bf97b88c5a1e9e69 : FAILURE in 46m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e67b36ca91fe4c2aa9cc7b14bd5771db : FAILURE in 45m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6d3527d77f144d7998d4ba6ca52e5cf4 : SUCCESS in 1h 10m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/66a760f2ab834525a43bbeec0f7ade62 : SUCCESS in 1h 14m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f155d5ebe20844c78e7412a0f02deeac : FAILURE in 1h 05m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b0e9076c02041ac926657a05011df06 : SUCCESS in 29m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d308240ee5f14157b9f32505a997e841 : SUCCESS in 37m 38s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/404fc66a3559494fa27ef09022e68580 : SUCCESS in 1h 59m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4de85fe4ef3f41c395cd54ad26dd979e : SUCCESS in 1h 12m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fa2c8d1b203e4a53a38142689fb222e3 : SUCCESS in 37m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81c1bf9b189e435ca75e2307e9772bbb : FAILURE in 51m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bc5af8dfe21f44848e39d3ce95148c35 : SUCCESS in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"672c976ece9e63d030b81506cd5612d31aecc89f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-26 22:55:45.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"55f79e6217e26cad68e2f5f91785b97c33bb74ba","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-26 22:55:45.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"727dd88d129e797f75caf9997621de65a3119fca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-26 22:55:45.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"908339fe5e4b54b32bd7f004797d5e9c8247c837","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-26 22:55:46.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ab34087eebf639d9277b9375ac052d9a8fe84b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-26 22:57:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c52c78f09e048dc242afd07f76d0c023e7fb14c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 00:31:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3858e0abcbc42c69320dbc887f4599d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac447a4be7e42fb9e9a78058a1582d8 : SUCCESS in 5m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6e20d2220d394d21bbf5c9284f0d8c0e : SUCCESS in 1h 28m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7ec3d90b17fc492a8614cade4a7eee5d : FAILURE in 35m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee86941241b641c78b4e49b19d5b7ac7 : FAILURE in 46m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bead9900b16147d68dcb235a7e13ac8e : SUCCESS in 1h 07m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b53b40872b3747fab64793c65565ce38 : SUCCESS in 58m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/67cfa50ebeab4fc1915fae2690bcf1a1 : SUCCESS in 1h 16m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cef07dabca2140f688979bec7e0f454e : SUCCESS in 20m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0bc1b1460af9468b8f616d2b982ed05a : SUCCESS in 39m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3cf4c30f47fa42b2a422040bb0897151 : SUCCESS in 50m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e0682700e1314d65ac8ec6614274dc99 : SUCCESS in 47m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ec3e2d59017341d3a22ce8f11f75acd2 : SUCCESS in 21m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a8b35ca61803405482834849e8d188e9 : FAILURE in 56m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b5abb13dadb14b1786423567e2eedcf2 : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"01a0af06d8a39cf81134beb37350e6cd0a54eb09","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-27 01:03:55.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/5/check/DellEMC_PowerStore_Manila_dhss_false/6284851/DellEMC_PowerStore_Manila_dhss_false/19 : FAILURE in 31m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"5beec21b3c0dc7a7394cd4015ef99325991aba26","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-27 02:11:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/5/upstream-check/manila-tempest-plugin-ontap-dhss/1f51c56/ : RETRY_LIMIT in 8m 01s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/bfd4b83/ : FAILURE in 1h 38m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7c4fcbf3f8e827d2216c5bcd55a8ec079caef9cd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 16:46:48.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"7dabd2c7f0cd43ce195a995524834da16bf72a81","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-27 16:46:48.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"870a61a0582961e263af17b4c068096ccc5114d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-20 21:11:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"81d3fc6020348f502f3482fb3aff4ac8547bd91b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-20 21:45:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/6/check/DellEMC_Unity_Manila_dhss_true/65affde/DellEMC_Unity_Manila_dhss_true/431 : FAILURE in 33m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/6/check/DellEMC_Unity_Manila_dhss_false/efdf71c/DellEMC_Unity_Manila_dhss_false/457 : FAILURE in 13m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"e891792722be668d4e969778aa231b7f4cfb714f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 22:49:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936a94cc56694ad0aca8954951936ef6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9db1bd5258049dcbb37d4bb41655980 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a04e171477564d018c134acae11f39cc : SUCCESS in 59m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3b60a10b0a7c429abc6b33ab26edff11 : FAILURE in 46m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d1bc6f0a250e4091ac784ca9520f7aff : FAILURE in 28m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/35504c33c7dc4dbd84852d0843cdce2c : FAILURE in 1h 02m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/11e61851ebd2441c8db7a2a366416a01 : FAILURE in 34m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/eff1c6e9873944cf9fd8a0f4da8b4f01 : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ac7f63e51d934e78be90658f676beb16 : SUCCESS in 34m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/96e0b11797304547b04fd029d4a6c2e2 : SUCCESS in 24m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/faf513bfce3b4e8f877860d4ce5cfd1a : SUCCESS in 1h 32m 04s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/04331fabbc354b8a85799f5eb0473bab : SUCCESS in 48m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f17777afc21b4316889c35a033953eea : SUCCESS in 36m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/96f0b49fd3294407be962586c5ca9e3c : SUCCESS in 53m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e97936df5ee483ab1b0b8caea8302e0 : FAILURE in 30m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf2a62272d9fb83ad1f7b116a09babfc5c19225c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-20 23:04:05.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/6/check/DellEMC_PowerStore_Manila_dhss_false/08f8d89/DellEMC_PowerStore_Manila_dhss_false/95 : FAILURE in 14m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"3cfbfb3bf74bd3878290944903828dcd3d4e0e23","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-20 23:21:18.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/6/check/DellEMC_Unity_Manila_dhss_true/5adf952/DellEMC_Unity_Manila_dhss_true/432 : FAILURE in 31m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/6/check/DellEMC_Unity_Manila_dhss_false/0b7097b/DellEMC_Unity_Manila_dhss_false/458 : FAILURE in 12m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"5867ed8b3965ffb20b1d74d00654e938b002e037","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-21 00:34:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/6/upstream-check/manila-tempest-plugin-ontap-dhss/d18defb/ : RETRY_LIMIT in 7m 54s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/329f0cf/ : FAILURE in 1h 44m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"2ab06e5aa9d0edb69a287370eb54b76a94fdf3f1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-21 06:04:04.000000000","message":"Patch Set 6:\n\n(9 comments)","accounts_in_message":[],"_revision_number":6},{"id":"72e31589190ff290818ded0e24a0ec13b7f34f51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-21 23:53:07.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d3979af01d0949a1741f9cea5261b19c56d93cd2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 00:07:28.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"75144241f5007d04e9629a716ee812debd35654d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 00:08:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b90725f5e9ba709ea647d2f127b3ba65306d58ed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-22 00:19:34.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_true/f37163f/DellEMC_Unity_Manila_dhss_true/436 : FAILURE in 26m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_false/bfbf9aa/DellEMC_Unity_Manila_dhss_false/462 : FAILURE in 12m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"2823066f0f92bc6f1336de9d2b4933673c282926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 01:21:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d6dbd11d22450e9476ab1eb638e788\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782a81ae64a6497cab39f4f54d9379c2 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4ba90fb45dc45a48cc6c42ea636a002 : SUCCESS in 53m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c23cfb767b434dbd9c14a1a62223a980 : FAILURE in 32m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/606d54235c4146d987d407717e84e8b8 : FAILURE in 32m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bc4ae2010b7b471f91656f406627b6b3 : SUCCESS in 1h 13m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bcb0eadb8602483ea27e8b93b62f17b7 : FAILURE in 18m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/03d97ca3388841319fb8450c6f2cf299 : SUCCESS in 1h 23m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8207505b63ca48d1b9501f0c941f6afa : SUCCESS in 25m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1702292970a44e2b939d12767e0c4003 : SUCCESS in 38m 37s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/80d6429190754ad18261c749ac43410b : SUCCESS in 1h 07m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4443f1989b4e46e88a430b2028e2ef21 : SUCCESS in 1h 16m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/06f1fc48116e415b891e887f154b1aad : SUCCESS in 23m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/31721b576a3349e193a0612faf9fc10f : FAILURE in 37m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3ecdcea2fa144977bc0c90467d3432e9 : FAILURE in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f8a5f83c3aa87449b69f4115162530ec6aeb6a65","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-22 01:36:19.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_PowerStore_Manila_dhss_false/292a931/DellEMC_PowerStore_Manila_dhss_false/97 : FAILURE in 14m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"3a1232141d1d7c946c24cd742ed50a09d9273e36","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-22 01:50:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_true/91f10c7/DellEMC_Unity_Manila_dhss_true/437 : FAILURE in 28m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_false/9082729/DellEMC_Unity_Manila_dhss_false/463 : FAILURE in 12m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b874d4516d56e97b0b26cdc9acaaf2df819f5aa7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-22 03:31:43.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/7/upstream-check/manila-tempest-plugin-ontap-dhss/fdb0410/ : POST_FAILURE in 1h 42m 11s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/ad32626/ : FAILURE in 2h 08m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f6e607b7ebdb6da957918f20a198823b80631082","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-23 07:54:48.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a8ad425c740efa429ac44179c572c5e9fb162d07","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-25 13:28:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab0ff31c4f13a3dfb61308a98c854b696423d834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 13:29:28.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a53643e5f16c87eec8b126beea54438e47e38ee1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 15:20:54.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ae2fc2f143743b5bc837200a49c1085\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b540caa26054f1783c4192e68726f4c : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/067d01616bfb4e528f6c1407cbe45f88 : SUCCESS in 1h 46m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6a26de4edb143fd818959bae860e038 : SUCCESS in 37m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4a424df7fa7c4af9ae6ed81361260f1a : FAILURE in 34m 07s","accounts_in_message":[],"_revision_number":7},{"id":"4a90855daeaded9d6b4a9796ead2e941e6579a1a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-28 12:57:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d520f6a0aea50c6af9960daf64c2b33ae6c5611d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:05:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6735f233aa641c59c21c6283996302a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e566f07c6a418482e1c673bc1a56c1 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d2174e23cee4e2595b2fd50c7c746d5 : SUCCESS in 1h 33m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/884da36fee424431b052c427b7c9a46f : FAILURE in 45m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/647a1f64b40f491ab128db772d86ce70 : FAILURE in 57m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9c6e54fcaa37412db12b7b57fc5b1eaa : SUCCESS in 1h 10m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/565adef64f154b86a5d93fe7aab82497 : SUCCESS in 1h 57m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0fb1f61fe94d4eeca02cdef94e0eefde : FAILURE in 26m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0118f70c4ab642b9bcf5989306fd618f : SUCCESS in 43m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b38ce56648214b5bb596f14856f56952 : SUCCESS in 52m 07s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/179b74b5a72444319dcb34bcc2f70979 : SUCCESS in 1h 42m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/eb65ee14c3c9402583d606e64497ff76 : SUCCESS in 1h 24m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ed79e816679f45a3b135fc2fa5494d30 : SUCCESS in 36m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d64c99a09d5049acb398b5343b0d2b13 : FAILURE in 54m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c16af791b6194a7f86ba0fef5d986b52 : FAILURE in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"94daa9f1ca329906be70e8fca5a42c7414aadbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:06:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"02a02dcf7e37b83e80a8372e6464718ab82b13e5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 15:23:09.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_true/5e63c7b/DellEMC_Unity_Manila_dhss_true/478 : FAILURE in 16m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_Unity_Manila_dhss_false/9cc2c14/DellEMC_Unity_Manila_dhss_false/504 : FAILURE in 13m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"0902420e0b6662b482ec26d2839e495fd4688d1f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-28 15:41:44.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_PowerStore_Manila_dhss_false/4868652/DellEMC_PowerStore_Manila_dhss_false/118 : SUCCESS in 35m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"4352008a06e1649af805c86d9b10d78b14a723be","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 15:50:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/18/888618/7/check/DellEMC_PowerFlex_Manila_dhss_false/3efa017/DellEMC_PowerFlex_Manila_dhss_false/264 : SUCCESS in 33m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"6f5aab9808adb2ee9d83b22c3c36f302885423c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:07:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c9d2e75b0345aba4be4da589418356\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2517393b4c4b4b98848f322ed8bd87 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3b9f6c9217df4fb3959f5bd42970da4b : SUCCESS in 1h 01m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d14090723e3f4f4fb6d4de31c1323241 : SUCCESS in 37m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7280f38a7b6c425a8bd500fcd66944db : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":7},{"id":"324cf7c3a19e5c605dc3a8e9219398c9ff46b40d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:07:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"201b173f76f0752ec6f233dfd6b88708a03aa780","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-28 17:28:13.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/7/upstream-check/manila-tempest-plugin-ontap-dhss/c2eba69/ : FAILURE in 1h 17m 32s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/888618/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/f5f6146/ : FAILURE in 2h 21m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"646fcd2e23b4e0c15489a275af15a9ef9a7c0b0b","revisions":{"3b341e1ee35f16eeda440d4803be631d97138a06":{"kind":"REWORK","_number":1,"created":"2023-07-14 19:48:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/1"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"subject":"[wip] Test resource locks","message":"[wip] Test resource locks\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b341e1ee35f16eeda440d4803be631d97138a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b341e1ee35f16eeda440d4803be631d97138a06"}]},"branch":"refs/heads/master"},"7cf353a912e42e3ebf8f02ac19b98c2286b01e6b":{"kind":"REWORK","_number":2,"created":"2023-07-14 19:55:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/2"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:55:50.000000000","tz":-420},"subject":"[wip] Test resource locks","message":"[wip] Test resource locks\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cf353a912e42e3ebf8f02ac19b98c2286b01e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cf353a912e42e3ebf8f02ac19b98c2286b01e6b"}]},"branch":"refs/heads/master"},"d2643ec6d7bbaac3fce7fe0720d9e8319befaff0":{"kind":"REWORK","_number":3,"created":"2023-07-15 00:57:19.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/3"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-15 00:56:41.000000000","tz":-420},"subject":"Test resource locks","message":"Test resource locks\n\nAdd API tests for the resource locks APIs\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2643ec6d7bbaac3fce7fe0720d9e8319befaff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2643ec6d7bbaac3fce7fe0720d9e8319befaff0"}]},"branch":"refs/heads/master"},"e20307670e449c696aef9d58af5a888891436df2":{"kind":"REWORK","_number":4,"created":"2023-07-24 16:55:56.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/18/888618/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/4"}}},"commit":{"parents":[{"commit":"5bd5f54073329423bb87f0345012f40c93251c2e","subject":"[RBAC] Add share snapshot metadata tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bd5f54073329423bb87f0345012f40c93251c2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-07-24 16:55:02.000000000","tz":-180},"subject":"Test resource locks","message":"Test resource locks\n\nAdd API tests for the resource locks APIs\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e20307670e449c696aef9d58af5a888891436df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e20307670e449c696aef9d58af5a888891436df2"}]},"branch":"refs/heads/master"},"4f7d08145ac1c4c7d45fc82edd8c8727fcb6dea8":{"kind":"REWORK","_number":5,"created":"2023-07-26 22:57:58.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/5"}}},"commit":{"parents":[{"commit":"5bd5f54073329423bb87f0345012f40c93251c2e","subject":"[RBAC] Add share snapshot metadata tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bd5f54073329423bb87f0345012f40c93251c2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-26 22:57:47.000000000","tz":-420},"subject":"Test resource locks","message":"Test resource locks\n\nAdd API tests for the resource locks APIs\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f7d08145ac1c4c7d45fc82edd8c8727fcb6dea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f7d08145ac1c4c7d45fc82edd8c8727fcb6dea8"}]},"branch":"refs/heads/master"},"f226d1eea40b8f7e09dee4c535301932f7618104":{"kind":"REWORK","_number":6,"created":"2023-08-20 21:11:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/6"}}},"commit":{"parents":[{"commit":"5bd5f54073329423bb87f0345012f40c93251c2e","subject":"[RBAC] Add share snapshot metadata tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bd5f54073329423bb87f0345012f40c93251c2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-20 21:11:45.000000000","tz":0},"subject":"Test resource locks","message":"Test resource locks\n\nAdd API tests for the resource locks APIs\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f226d1eea40b8f7e09dee4c535301932f7618104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f226d1eea40b8f7e09dee4c535301932f7618104"}]},"branch":"refs/heads/master"},"646fcd2e23b4e0c15489a275af15a9ef9a7c0b0b":{"kind":"REWORK","_number":7,"created":"2023-08-21 23:53:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/888618/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/888618/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/888618/7"}}},"commit":{"parents":[{"commit":"5bd5f54073329423bb87f0345012f40c93251c2e","subject":"[RBAC] Add share snapshot metadata tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bd5f54073329423bb87f0345012f40c93251c2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 19:47:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-21 23:52:44.000000000","tz":-420},"subject":"Test resource locks","message":"Test resource locks\n\nAdd API tests for the resource locks APIs\n\nChange-Id: Idf71e236b1b8a2558bb4ad3de1018fa33b41877f\nPartially-implements: bp/allow-locking-shares-against-deletion\nDepends-On: I146bc09e4e8a39797e22458ff6860346e11e592e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/646fcd2e23b4e0c15489a275af15a9ef9a7c0b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/646fcd2e23b4e0c15489a275af15a9ef9a7c0b0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
