)]}'
{"id":"openstack%2Fmanila-tempest-plugin~890192","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I33e1312a378de50591a581ead883e34afa0d05c4","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-17 19:58:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I33e1312a378de50591a581ead883e34afa0d05c4","subject":"[DNM] Test share backups with dummy driver","status":"ABANDONED","created":"2023-08-01 10:56:00.000000000","updated":"2023-10-17 19:58:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ea98019f5c59f07a32dd8c86237c22f83a2f8bd","_number":890192,"virtual_id_number":890192,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-01 20:02:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 12:05:51.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-01 12:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 15:44:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"e964c941205e43d6e7a7b04077a3ce07131f9bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 10:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef9b78aa74cfb9b269a37f7e79f823335d704e74","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 12:05:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/1/check/DellEMC_PowerFlex_Manila_dhss_false/487479d/DellEMC_PowerFlex_Manila_dhss_false/133 : FAILURE in 30m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"88c0ec2f40eec611d205c3ea42654a2d4ddbc630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:22:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff78300291a64d38980584b833ba868a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0991aeb3c98a45f3ab44b60def27693d : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/90a28d268e49453a917e40ba0f20226a : SUCCESS in 1h 13m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f3e9741865624f11889ab6ea95783d2b : FAILURE in 28m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8a9a20289ffc487faa24c34f2e287e5e : FAILURE in 38m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e7c03b725d8740b6951c98fc40b1864d : SUCCESS in 1h 17m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d947073f03784f2f9f5b0593626ff2aa : FAILURE in 32m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/86aa51df50e641b7bc5537176f7f661b : SUCCESS in 1h 03m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/40c8272c5e1d4955825b4db887fab8b5 : FAILURE in 27m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/939a94b6c8ac4785865fa4ff85c12407 : SUCCESS in 34m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e3e95858c8c047f6a79fbc1098a88621 : SUCCESS in 1h 12m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/527ec7fbfddd469c9a803ad456c3ca6d : SUCCESS in 1h 18m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a96c910fd6e24acc91ea6242be26d710 : SUCCESS in 21m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d36f3aca1e0416fa1af3d7a5c7e917f : SUCCESS in 48m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/62da7016bc874bccb53ba5b1d09c713b : SUCCESS in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87b14422da56e0c76648f134cb79b58cfba4e0e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 13:26:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c24db32b749678fc666b9e8d06f69fd4ebf0e669","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 13:57:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/2/check/DellEMC_PowerFlex_Manila_dhss_false/4cd803a/DellEMC_PowerFlex_Manila_dhss_false/134 : FAILURE in 30m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"eef54fbe8bcf7a9bcae443c33edcbf88f7da56ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 13:59:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d70d2bece65bc9115e6fd18ac733097182f4f413","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 14:30:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/3/check/DellEMC_PowerFlex_Manila_dhss_false/672e531/DellEMC_PowerFlex_Manila_dhss_false/135 : FAILURE in 30m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"8b64824277e4cf20a60c71ba37d12cc7b5872ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 15:22:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2405746eb00b4700bd5bb1b8d23acdcb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284413372c02484dbd4a806cfaf4a302 : SUCCESS in 2m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e30db4b2194b4f0e9a31e1031a75ae41 : SUCCESS in 1h 17m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0204fae6457943b39d8e34954597e12e : FAILURE in 32m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fe212319fd1341b9a7710605f343765e : FAILURE in 44m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f11131fa25b545d1b1abaa89b54ce4f2 : SUCCESS in 55m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6e996d6d979642af8a1d0a18c5054289 : FAILURE in 42m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/661279bdd3fb4061be5c8d6da2c6244b : SUCCESS in 59m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bfbcbb08d5074ba4b65f5b8e172a3f13 : FAILURE in 28m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15966c13e2c04ba9b6b5e285a7350441 : SUCCESS in 41m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/da2a77869a5041b4aa95f275d9e1d02f : SUCCESS in 1h 14m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0183892578ea423eb5409c7854ce8331 : SUCCESS in 52m 42s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fd581ebcbbea4ec69859cdb65922590e : SUCCESS in 31m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8a42bbd73f924e9c9ae4e0cd94343b00 : FAILURE in 54m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/070c5b8d851d4adf9a5a9974a3f9066e : SUCCESS in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f697c5b5a5a28676e911a1ca0489cdb9d6e85f67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-01 15:44:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/92/890192/3/check/DellEMC_Unity_Manila_dhss_true/7a80e27/DellEMC_Unity_Manila_dhss_true/311 : FAILURE in 1h 44m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/3/check/DellEMC_Unity_Manila_dhss_false/8835152/DellEMC_Unity_Manila_dhss_false/335 : SUCCESS in 1h 09m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3216232902d4a2f1c703d3522e2cdc06ce7afb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 16:28:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"80c0b585ec188fbf0dc41ddb54bc6c97209850ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 16:59:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/4/check/DellEMC_PowerFlex_Manila_dhss_false/86d2393/DellEMC_PowerFlex_Manila_dhss_false/137 : FAILURE in 30m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"8adc65bcf07fbc92a63fbe649996ccf9e16f5a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 18:02:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df8f610d86f24579ae1dad289e16c72a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6089efe5924b4ca0a63b2d34fcfce5e1 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3bc5a36b8d8b42beaa7a3d3c4ece38bd : SUCCESS in 1h 16m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d9ddb1129d144504b9223491f5f11339 : FAILURE in 50m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/196b6fb9374d4eb58685d3c797c9c689 : FAILURE in 47m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3136693c092040e7b268206a1850cbb1 : SUCCESS in 50m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5a52fe7a0fa948598acfa90a8cea24db : FAILURE in 29m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dd4273a9e5384a3696f313ccc19cd7a8 : SUCCESS in 1h 00m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e22f929df374477e83eb89a6d13b51f7 : FAILURE in 40m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c9a9690a23e48b1a656dab890c03161 : SUCCESS in 28m 38s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1932fc29889a49228de74f9d83de28ef : SUCCESS in 1h 28m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8ee433248a6844e085eead7202d80611 : SUCCESS in 1h 09m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9aa4c9a39579481e83a0b1a04cbb7ef1 : SUCCESS in 25m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/53e6ea658b8e41c9be94aab7c18238b7 : FAILURE in 51m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dc2e287d0bc546f8b888961d918c9ab8 : FAILURE in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b27c052318ffe2febe09508e51592f91a53b8154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 18:28:40.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"75fcdc36f1acdcc28b6b1ed520db5f25fdf29363","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 18:59:34.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/5/check/DellEMC_PowerFlex_Manila_dhss_false/c6c7960/DellEMC_PowerFlex_Manila_dhss_false/138 : FAILURE in 30m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"f3bd6a331e50b244dbc90647d565ff4cde91a235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 20:02:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e884ebd93254bd48ef5e157bcfc152a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5342637e47a4c6caf8fec5201e0855c : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/84b11cfb69724d7c92af4c968714b7e9 : SUCCESS in 1h 16m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/81648981a9a041ef8d38c6f34b5bf7f7 : FAILURE in 58m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d5174d3d476b403bbaaa86769c8528d3 : FAILURE in 45m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0ead3b14e64f4b27839f4c52d2564081 : SUCCESS in 1h 19m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/647c1a878e014424951a10f4974f4c98 : FAILURE in 40m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1302b9051ca64033acf62d9502766c18 : FAILURE in 1h 26m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/835bb61c4d404c16ad50310ef199fee6 : FAILURE in 23m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3161aa71100e46209feb5232d878aa39 : SUCCESS in 44m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9de8dd5d2c0e488d84a560702e9d5bc8 : SUCCESS in 1h 21m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e575bc00097345ffa30475b62a2a8d1e : SUCCESS in 1h 14m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c6ed987b73074344b961f1f3f2a93b48 : SUCCESS in 27m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d22f1b15b8d14393b48c690e9ba83e81 : SUCCESS in 58m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5a59ecef4b244a6290b69ab3ff3b28a6 : SUCCESS in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad0bccd73ef8c637c62fb4f652215e8c4c2baac8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-01 20:32:46.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/92/890192/5/check/DellEMC_Unity_Manila_dhss_true/265e473/DellEMC_Unity_Manila_dhss_true/314 : FAILURE in 2h 03m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/890192/5/check/DellEMC_Unity_Manila_dhss_false/22221b7/DellEMC_Unity_Manila_dhss_false/338 : SUCCESS in 1h 12m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"9ea98019f5c59f07a32dd8c86237c22f83a2f8bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-17 19:58:41.000000000","message":"Abandoned\n\nChange not needed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b929bf984d95d711eff6ac203d1890e9e0c15210","revisions":{"806455e0eaab0abc104f08c2a8e0d89ad9d45c54":{"kind":"REWORK","_number":1,"created":"2023-08-01 10:56:00.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/92/890192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/890192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/1"}}},"commit":{"parents":[{"commit":"fc24f2379c32347e41f70db1162db8d88b47af63","subject":"Add share backup tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"subject":"[DNM] Test share backups with dummy driver","message":"[DNM] Test share backups with dummy driver\n\nChange-Id: I33e1312a378de50591a581ead883e34afa0d05c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/806455e0eaab0abc104f08c2a8e0d89ad9d45c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/806455e0eaab0abc104f08c2a8e0d89ad9d45c54"}]},"branch":"refs/heads/master"},"a141b69fc90606c502de7551b3e1af20942f445a":{"kind":"REWORK","_number":2,"created":"2023-08-01 13:26:59.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/92/890192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/890192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/2"}}},"commit":{"parents":[{"commit":"fc24f2379c32347e41f70db1162db8d88b47af63","subject":"Add share backup tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 13:26:49.000000000","tz":-180},"subject":"[DNM] Test share backups with dummy driver","message":"[DNM] Test share backups with dummy driver\n\nChange-Id: I33e1312a378de50591a581ead883e34afa0d05c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a141b69fc90606c502de7551b3e1af20942f445a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a141b69fc90606c502de7551b3e1af20942f445a"}]},"branch":"refs/heads/master"},"e0d3e35725c36c485ff765de5b6dbce99c4068df":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-01 13:59:47.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/92/890192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/890192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/3"}}},"commit":{"parents":[{"commit":"fc24f2379c32347e41f70db1162db8d88b47af63","subject":"Add share backup tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-08-01 13:59:47.000000000","tz":0},"subject":"[DNM] Test share backups with dummy driver","message":"[DNM] Test share backups with dummy driver\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/343980\nChange-Id: I33e1312a378de50591a581ead883e34afa0d05c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0d3e35725c36c485ff765de5b6dbce99c4068df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0d3e35725c36c485ff765de5b6dbce99c4068df"}]},"branch":"refs/heads/master","description":"Edit commit message"},"35490c34973cd5677f25819b1c94d60d62f07d0d":{"kind":"REWORK","_number":4,"created":"2023-08-01 16:28:13.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/92/890192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/890192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/4"}}},"commit":{"parents":[{"commit":"fc24f2379c32347e41f70db1162db8d88b47af63","subject":"Add share backup tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 16:28:03.000000000","tz":-180},"subject":"[DNM] Test share backups with dummy driver","message":"[DNM] Test share backups with dummy driver\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/343980\nChange-Id: I33e1312a378de50591a581ead883e34afa0d05c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35490c34973cd5677f25819b1c94d60d62f07d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35490c34973cd5677f25819b1c94d60d62f07d0d"}]},"branch":"refs/heads/master"},"b929bf984d95d711eff6ac203d1890e9e0c15210":{"kind":"REWORK","_number":5,"created":"2023-08-01 18:28:40.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/92/890192/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/890192/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/890192/5"}}},"commit":{"parents":[{"commit":"fc24f2379c32347e41f70db1162db8d88b47af63","subject":"Add share backup tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-08-01 10:55:26.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-08-01 18:28:34.000000000","tz":0},"subject":"[DNM] Test share backups with dummy driver","message":"[DNM] Test share backups with dummy driver\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/343980\nChange-Id: I33e1312a378de50591a581ead883e34afa0d05c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b929bf984d95d711eff6ac203d1890e9e0c15210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b929bf984d95d711eff6ac203d1890e9e0c15210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
