)]}'
{"id":"openstack%2Fmanila-tempest-plugin~893651","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-06 20:11:53.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-09-06 19:24:09.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-09-06 20:11:53.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-09-06 20:11:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916","subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","status":"MERGED","created":"2023-09-04 13:19:27.000000000","updated":"2023-09-07 10:28:57.000000000","submitted":"2023-09-06 20:11:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893651","meta_rev_id":"1ae7f341c16ae83f52ccd2d9930cda45566d14eb","_number":893651,"virtual_id_number":893651,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 20:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2023-09-06 19:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2023-09-06 19:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 13:43:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-09-04 16:20:06.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-04 23:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 23:44:31.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-06 02:23:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-06 09:27:37.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2023-09-06 09:27:37.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2023-09-06 09:28:06.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-09-07 10:28:57.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"9ccc3392557868f20e991f5a88350344f3149834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 13:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5bf7af8c9c1b82cacd2863a472313792d307ddb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 13:43:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/1/check/DellEMC_Unity_Manila_dhss_true/4b69d7f/DellEMC_Unity_Manila_dhss_true/507 : FAILURE in 24m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/1/check/DellEMC_Unity_Manila_dhss_false/e745f6d/DellEMC_Unity_Manila_dhss_false/533 : FAILURE in 14m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"06f28d3b872252a295c21261e91f3e5adcb73585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 13:55:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"44e592c1da3e45a1b7cec61ae724c7abba167f16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 14:05:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"81c4044b8a77310be7dc9fa3b30d367d0c2fa4ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 14:17:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a80e9587e4106b8727818c97ccd6724c34e06042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 14:19:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4632fa97782f613fedd741d1d225d7f645df90e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 14:34:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d63a868d01a075a8b2795cd86b4c91f57f80ad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 14:50:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fb86d108013f2832e5e033ef4fe7df7cd417acd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 15:04:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7e9c2d9119f5c7445bb59365d8fa5a46a0eb12cd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 15:50:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/8/check/DellEMC_Unity_Manila_dhss_true/7996c3e/DellEMC_Unity_Manila_dhss_true/515 : FAILURE in 23m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/8/check/DellEMC_Unity_Manila_dhss_false/83bc72d/DellEMC_Unity_Manila_dhss_false/541 : FAILURE in 14m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"1d5e077906b275185809c5b874c3ab01ed28135b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 16:20:06.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/8/check/DellEMC_PowerFlex_Manila_dhss_false/016920f/DellEMC_PowerFlex_Manila_dhss_false/303 : SUCCESS in 42m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"d09eb788ce33256827efa7eaf27ee9dbf122ce84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 17:04:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2bf43b58c010652ad5937cb23ceea983dc3e42e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 17:10:02.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cb6b9eac9a2fa5ed70a54ea57e06ac2776bc6bd4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 17:19:05.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/10/check/DellEMC_PowerFlex_Manila_dhss_false/76b5ae6/DellEMC_PowerFlex_Manila_dhss_false/305 : FAILURE in 8m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"fc620af12e4938d8518b4d98468cf6f823a86349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 17:25:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"426ec418619a9c21fc8b112879bedfd7ab202beb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 17:34:34.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/11/check/DellEMC_PowerFlex_Manila_dhss_false/332e713/DellEMC_PowerFlex_Manila_dhss_false/306 : FAILURE in 8m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"c45f83e13bd139a8893216187fd2c892f66e84a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 17:50:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ce53ed709d5ba9e11062bbd2d53210fa9adba375","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 17:59:20.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/12/check/DellEMC_PowerFlex_Manila_dhss_false/437b545/DellEMC_PowerFlex_Manila_dhss_false/307 : FAILURE in 8m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"9bc8b94fbd2bf224ef02a4bf47f51467a207cc70","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 18:09:39.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/12/check/DellEMC_Unity_Manila_dhss_true/9725f30/DellEMC_Unity_Manila_dhss_true/519 : FAILURE in 19m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/12/check/DellEMC_Unity_Manila_dhss_false/09469d9/DellEMC_Unity_Manila_dhss_false/545 : FAILURE in 8m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"4ead19ee962aaf1699486e492ac3bb516cf6b5a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 18:11:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c9eab7c769bb4563d6d37befa898e223e18c7cca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 18:20:41.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/13/check/DellEMC_PowerFlex_Manila_dhss_false/5de3723/DellEMC_PowerFlex_Manila_dhss_false/308 : FAILURE in 9m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"e2bbed604a52f9182792f5af939631be6c366ff0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 18:30:28.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/13/check/DellEMC_Unity_Manila_dhss_true/67aab88/DellEMC_Unity_Manila_dhss_true/520 : FAILURE in 18m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/13/check/DellEMC_Unity_Manila_dhss_false/dd37b10/DellEMC_Unity_Manila_dhss_false/546 : FAILURE in 8m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"878cfc8fd09596cbec9bba31f639c2132b1e1ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 18:54:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9c4c5025327a97ef52bcc0d562654536b47144a3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 19:04:06.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/14/check/DellEMC_PowerFlex_Manila_dhss_false/12261e5/DellEMC_PowerFlex_Manila_dhss_false/309 : FAILURE in 8m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"fbbd50509500a7e7c9fefd2b989b662fe61f060e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 19:27:13.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/14/check/DellEMC_Unity_Manila_dhss_true/c35fe89/DellEMC_Unity_Manila_dhss_true/521 : FAILURE in 32m 01s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/14/check/DellEMC_Unity_Manila_dhss_false/ba1dbf5/DellEMC_Unity_Manila_dhss_false/547 : FAILURE in 8m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"39aec75053275622f5d361b767fdcd43ec841663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-04 22:34:23.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"36dbe1a938fbace70cfaa83891900992af96f402","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-04 22:43:45.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/15/check/DellEMC_PowerFlex_Manila_dhss_false/ce3e2c3/DellEMC_PowerFlex_Manila_dhss_false/310 : FAILURE in 9m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"ae916e6f6221bead49562ef1c5e0b55b303f5d4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-04 23:00:38.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/15/check/DellEMC_Unity_Manila_dhss_true/f5c3934/DellEMC_Unity_Manila_dhss_true/522 : FAILURE in 25m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/15/check/DellEMC_Unity_Manila_dhss_false/ad34bde/DellEMC_Unity_Manila_dhss_false/548 : FAILURE in 8m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"ffe976b6f86b11896545441d3e45a93292d61315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 23:56:04.000000000","message":"Patch Set 15: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b199c3b2d5e44057bba839be883b97d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0079543806044aafa4f6da3a9da4b9fe : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9eb715e06e5443369fba7e4ea3d4bf9d : FAILURE in 46m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4a369eca6cb741b2bff453e142a01777 : FAILURE in 46m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b84bbc7a915b4cf48f7f58dc9297abb6 : FAILURE in 43m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/83e7f11affec4b9591227a34e8257bba : SUCCESS in 1h 07m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6fa73df65947464c9a56d616d23765ac : SUCCESS in 1h 17m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b2a314fe57b45638e3f764cbca96b97 : SUCCESS in 1h 14m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97a6d33a07cf419dab87bf4cbc606012 : SUCCESS in 37m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c8942252b5bf40218c8d72462036a2bc : SUCCESS in 43m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/651c6add55324ed8bb35805ea336fa2d : FAILURE in 48m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7979a335cea04852a3ed97a54f24f889 : FAILURE in 29m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d75103befd9c45bea2d7da580a3c8a7d : SUCCESS in 24m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cb0dc89159c64f76b52d720ac5861a37 : SUCCESS in 31m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/970864fba6f34b8389c6568db03960ca : FAILURE in 3m 47s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":15},{"id":"b14d99dcc376f1f38f5dd431e10c30562583e13e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 12:59:10.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"b6989294970ada630e6c4c3e560fa787b1b147b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-05 13:13:06.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/16/check/DellEMC_PowerFlex_Manila_dhss_false/5148a9b/DellEMC_PowerFlex_Manila_dhss_false/312 : FAILURE in 13m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"5ffa464d0cc77b649e8ac44719691a348c4a4349","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 13:15:30.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/16/check/DellEMC_Unity_Manila_dhss_true/3702b09/DellEMC_Unity_Manila_dhss_true/525 : FAILURE in 16m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/16/check/DellEMC_Unity_Manila_dhss_false/6ea91fe/DellEMC_Unity_Manila_dhss_false/553 : FAILURE in 9m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"c794f591cbbe22383c2d15619d81383c6283a732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 13:24:34.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"165fd2d4af599af04a265819fe9b69f26fef45ab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 13:40:06.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/17/check/DellEMC_Unity_Manila_dhss_true/ff1580e/DellEMC_Unity_Manila_dhss_true/526 : FAILURE in 15m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/17/check/DellEMC_Unity_Manila_dhss_false/56cb3ee/DellEMC_Unity_Manila_dhss_false/554 : FAILURE in 8m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"1522adb2c0fbcca8938656e48b2d584d826e4ad5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-05 13:46:19.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/17/check/DellEMC_PowerFlex_Manila_dhss_false/41fee4f/DellEMC_PowerFlex_Manila_dhss_false/314 : FAILURE in 13m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"672eb0a30df522f7f31b397fe34a0cc97dc4065d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:00:10.000000000","message":"Patch Set 17: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3356319f7934cb2af2d00a0303ce86d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86aaa45d4c4a4928a5b034b9879b197c : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51c85b4dac38428fa224dc948879d547 : FAILURE in 40m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6c7cf7b6ffa247e2a699181fd70520a6 : FAILURE in 45m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/50ec9df58351460b8588af8a6c5aadd8 : FAILURE in 43m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ae319b6afa6340098eb364063d4d3a02 : SUCCESS in 1h 10m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8123609f88d2410a9b33b2f18c3ad5f9 : SUCCESS in 1h 15m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cfec1a1154f649c9a86c458cfe2b5e9a : SUCCESS in 1h 29m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3115fa8ddef04d608e38d683e527303b : SUCCESS in 43m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c77492840be8417a830c462dbc16c867 : SUCCESS in 48m 45s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e6d5a5633eb74f948024fa21363b28f4 : SUCCESS in 1h 06m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c18a6734ec324179b5ea2faffdfac5b7 : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/09baa073aea54944af9c05bbf00b73c4 : SUCCESS in 23m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/781a2a182bb44a97a0b9b87a971a8357 : FAILURE in 43m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/55481748ea3c468ebfe30a740fe70d5e : FAILURE in 21m 08s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":17},{"id":"9f5690bdb1bab4d3aeed94e0b398b7933f094b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 19:46:01.000000000","message":"Patch Set 18: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"32b82e45badb5bdc187c835eb13c0c5b72b1b788","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 20:17:14.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/18/check/DellEMC_Unity_Manila_dhss_true/117dd07/DellEMC_Unity_Manila_dhss_true/528 : FAILURE in 15m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/18/check/DellEMC_Unity_Manila_dhss_false/f404ecb/DellEMC_Unity_Manila_dhss_false/556 : FAILURE in 8m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"c80d128828cc133d6e66fb0fd0c7c8d6b2d5266b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:17:29.000000000","message":"Patch Set 18: Verified+1\n\n(7 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226a732cbfa94cbdacf91713cda009c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86012563c65e40598c67064244ef8f08 : SUCCESS in 3m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e11b1d46ad743fbb5196536d8f3f38f : SUCCESS in 50m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ee56055257e0461f9673a74d26bc4978 : FAILURE in 36m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/101d87eacf024d8bb7120b3fb7a6224e : FAILURE in 42m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1783d49869dc41cc83cbdfc887a27bb3 : SUCCESS in 1h 10m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/15b172666d3446f5879a5f7195c0a326 : SUCCESS in 1h 20m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2c73ed46bd544f0a8ef54b1553c16b3b : SUCCESS in 1h 23m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/085c4243f557473a9489118523a817cb : SUCCESS in 38m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1148ca2b04540ca87b714e213a66d4f : SUCCESS in 40m 01s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e4dbab468bee43958f409d5449fb6bfd : SUCCESS in 1h 18m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b5ffdc792e0645fcb98076821d4e17f6 : SUCCESS in 57m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/542f88a71fe14d78bdb9c9add17f35cc : SUCCESS in 34m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07e5cbf913fe48f089e893b72c92ff92 : SUCCESS in 38m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c946451203fb4e8d8a212c511b8122eb : FAILURE in 23m 31s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":18},{"id":"aa9f71008dab8af909f7212a27c3007a56bdb58e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 21:32:44.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/18/check/DellEMC_Unity_Manila_dhss_true/8e24dc2/DellEMC_Unity_Manila_dhss_true/529 : FAILURE in 15m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/18/check/DellEMC_Unity_Manila_dhss_false/0d15ae2/DellEMC_Unity_Manila_dhss_false/557 : FAILURE in 8m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"48b3cec0baf3ad0985b788d7eaa075c016093407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 21:53:33.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"cc763bd16fb71641662a58ca1bad513355942039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 21:55:44.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"eab66bae4ba3b1e7ee6a6c0d4eaa58cd5bbac3c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 22:10:15.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_Unity_Manila_dhss_true/7a094b7/DellEMC_Unity_Manila_dhss_true/531 : FAILURE in 14m 17s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_Unity_Manila_dhss_false/af0e8cc/DellEMC_Unity_Manila_dhss_false/559 : FAILURE in 8m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"01f40f8029f75dba931f4a17d102b1cbbff347b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 23:08:11.000000000","message":"Patch Set 20: Verified+1\n\n(7 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d8a5af257c4edd9ae98c2d95ef3e8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c32bc0076e746fe86934fec257a81ca : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cfb7b9c5285c4c558c1a006a66663b7e : SUCCESS in 47m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/265b597cd2fa4be9a0dbc3a136aea9c3 : FAILURE in 34m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fe046a8b9ffb40f3848e396e2841ce3d : FAILURE in 26m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e583835286944415ab0803b8469b5ee8 : SUCCESS in 57m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/758bef96e3f349598f05708a741a6d8d : SUCCESS in 1h 07m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/404ec29d0ec44b6c9722a99cadf2bc14 : SUCCESS in 57m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e08d4edb4f9413584ccc603d7ee455c : SUCCESS in 25m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2153361fa10c4a2dbf9b3c4ebde85cd7 : SUCCESS in 35m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1209001442d04f1f8f9a26909c0ada65 : SUCCESS in 27m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/376e96b3590a430380d10f526b04acbb : SUCCESS in 35m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/636acce0902d4ecaa843e825d956c55d : SUCCESS in 36m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/85836cc3b62948ed9e82147f0a721b8f : FAILURE in 54m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/865e15d042694258a010e0070f381fc9 : FAILURE in 22m 25s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":20},{"id":"c9e7621b6075fb4cbb247ae4eb4fcac77a9f03c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-05 23:23:10.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_Unity_Manila_dhss_true/bd1bcd5/DellEMC_Unity_Manila_dhss_true/534 : FAILURE in 14m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_Unity_Manila_dhss_false/df426fc/DellEMC_Unity_Manila_dhss_false/562 : FAILURE in 8m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"2b377380649f0b4817f96afa223648edbacaa703","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-05 23:44:31.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_PowerStore_Manila_dhss_false/4ac8cc2/DellEMC_PowerStore_Manila_dhss_false/134 : SUCCESS in 36m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":20},{"id":"2f95f32c1db24617a90285dc1eeaedf120239d8f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-06 02:23:56.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/893651/20/upstream-check/manila-tempest-plugin-ontap-dhss/535779b/ : FAILURE in 3h 14m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/893651/20/upstream-check/manila-tempest-plugin-ontap-no-dhss/6948208/ : TIMED_OUT in 3h 09m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"049767250c331f93bc95b1ce04d13c81b858f22f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-06 03:44:39.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/20/check/DellEMC_PowerFlex_Manila_dhss_false/f86043c/DellEMC_PowerFlex_Manila_dhss_false/315 : SUCCESS in 46m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"657b0ced5a4dc6dd304c2a3416d88e66718bb77c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-09-06 13:30:45.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"ca3135821f11646287a414f5d806bf033ed8e9cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-06 16:31:53.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"800bac516e4b8b741d8a487a40ab2d1aeb551749","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-06 16:44:20.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"6c4fce0bbd92c57d4b753c3a273c6b376ced91f1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-06 17:01:53.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/21/check/DellEMC_PowerFlex_Manila_dhss_false/dc9d816/DellEMC_PowerFlex_Manila_dhss_false/321 : SUCCESS in 29m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"a4b69e618902710d558e7cc39c48b7b616a2047a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 17:43:35.000000000","message":"Patch Set 21: Verified+1\n\n(7 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d05f1b1fd8b461aa70091b69907ff0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae9ce2d7cec47adb5c88f531dd605b0 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e446a77ccc2f419e8eb385d8bc3ea5b5 : SUCCESS in 44m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c388236a9f645f8be21255567b7e4dc : FAILURE in 40m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/daf4b82b4a89467890a261d7414afdc4 : FAILURE in 36m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f195fb21e7754cac82b7ef8ab1a2c318 : FAILURE in 58m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/63e9f605ae784c1ea1a7b7e114aba88e : SUCCESS in 1h 10m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ebb263ddb4464d8b8f697d2230f6cdae : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cdc89740218248e2893ff5f8f5d4caf0 : SUCCESS in 26m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1d6e9822f7884cd49c3a8238ad7d5e13 : SUCCESS in 33m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2e5361ae31c742a8b246e713c2c36814 : SUCCESS in 39m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1b5f0ef9389048708cddb37f78b9ca23 : SUCCESS in 35m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6a1335bf94c74620a271bf4e4d79b398 : SUCCESS in 25m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29054f6489d64c1c84d43caeca29fb55 : FAILURE in 40m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a2c9e54655174cb08339ee8bc7fbc67a : FAILURE in 13m 56s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":21},{"id":"7be9dafd04752c5f659fe84d65e42d726fe35c41","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-06 18:14:33.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/21/check/DellEMC_PowerFlex_Manila_dhss_false/f0d75e0/DellEMC_PowerFlex_Manila_dhss_false/323 : SUCCESS in 30m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"156731492da1279418ed2197363238f3e2845be3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-06 18:22:23.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/21/check/DellEMC_PowerStore_Manila_dhss_false/9369cd3/DellEMC_PowerStore_Manila_dhss_false/137 : SUCCESS in 38m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":21},{"id":"a562197aeec9394172d0f8c748d304be2babc667","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-09-06 19:24:09.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"74ee067d4dec27c471656dc9a523b488419caba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 19:24:45.000000000","message":"Patch Set 21: -Verified\n\n(7 comments)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":21},{"id":"544d8ad3b3b0c787886c70bcc5ff6acf409b7d3a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-06 19:46:04.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/893651/21/check/DellEMC_Unity_Manila_dhss_true/fd412d9/DellEMC_Unity_Manila_dhss_true/544 : FAILURE in 36m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/893651/21/check/DellEMC_Unity_Manila_dhss_false/9720581/DellEMC_Unity_Manila_dhss_false/571 : FAILURE in 2h 51m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"8440d907e77e75b9841447b65a71a0dfc34ce0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 20:11:53.000000000","message":"Patch Set 21: Verified+2\n\n(7 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35e6e646a244451ba95c6aec48d61b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36bdb8013dc4a81858ed3ff399fb84e : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/21d682a3503a47ce8faa9cf833298454 : SUCCESS in 33m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7561f9cd338d4cc7abc65808f59d0dfa : SUCCESS in 37m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aae971f2b0e34ba79996986b05fc5110 : SUCCESS in 41m 45s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":21},{"id":"8798b8bf68521b86da18a56dd422f15e00161bf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 20:11:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"ea0cfd9a2175d8ae55867e69e2f29cae7b6b2250","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-06 20:26:42.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/893651/21/upstream-check/manila-tempest-plugin-ontap-dhss/b38fd53/ : FAILURE in 2h 26m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/893651/21/upstream-check/manila-tempest-plugin-ontap-no-dhss/097a471/ : SUCCESS in 2h 41m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"91ca00a96f5a98c2282f4d896d1a45ebe013317a","revisions":{"9b27dd5bddee7813859fd293cbd0c640d6bf5598":{"kind":"REWORK","_number":1,"created":"2023-09-04 13:19:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/1"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe ubuntu focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9b27dd5bddee7813859fd293cbd0c640d6bf5598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9b27dd5bddee7813859fd293cbd0c640d6bf5598"}]},"branch":"refs/heads/master"},"a03272abb41e5ab6c6b6741c7c8dcb5d8e55479e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-04 13:55:03.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/2"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:55:03.000000000","tz":0},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a03272abb41e5ab6c6b6741c7c8dcb5d8e55479e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a03272abb41e5ab6c6b6741c7c8dcb5d8e55479e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f74a0092aac2b503655a5f849c7babf2db5b3e7f":{"kind":"REWORK","_number":3,"created":"2023-09-04 14:05:09.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/3"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-09-04 14:05:06.000000000","tz":0},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f74a0092aac2b503655a5f849c7babf2db5b3e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f74a0092aac2b503655a5f849c7babf2db5b3e7f"}]},"branch":"refs/heads/master"},"9c2e9523bc800b9185af974b8b29b4581504081e":{"kind":"REWORK","_number":4,"created":"2023-09-04 14:17:46.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/4"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 14:17:37.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c2e9523bc800b9185af974b8b29b4581504081e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c2e9523bc800b9185af974b8b29b4581504081e"}]},"branch":"refs/heads/master"},"dbd2ede3edd8e6b67cc8404897caf3c67fdef85e":{"kind":"REWORK","_number":5,"created":"2023-09-04 14:19:35.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/5"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 14:19:15.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbd2ede3edd8e6b67cc8404897caf3c67fdef85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbd2ede3edd8e6b67cc8404897caf3c67fdef85e"}]},"branch":"refs/heads/master"},"97c790dc1a82c7e94a9144cdc7a77019cc9b6693":{"kind":"REWORK","_number":6,"created":"2023-09-04 14:34:38.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/6"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 14:34:29.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97c790dc1a82c7e94a9144cdc7a77019cc9b6693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97c790dc1a82c7e94a9144cdc7a77019cc9b6693"}]},"branch":"refs/heads/master"},"45b536a1b79530428ae26d6341c47ccb861dbd3c":{"kind":"REWORK","_number":7,"created":"2023-09-04 14:50:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/7"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 14:50:28.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45b536a1b79530428ae26d6341c47ccb861dbd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45b536a1b79530428ae26d6341c47ccb861dbd3c"}]},"branch":"refs/heads/master"},"7b4cdbf0e60fd8ba022752308b7c418f8ef59d9f":{"kind":"REWORK","_number":8,"created":"2023-09-04 15:04:06.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/8"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 15:03:59.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b4cdbf0e60fd8ba022752308b7c418f8ef59d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b4cdbf0e60fd8ba022752308b7c418f8ef59d9f"}]},"branch":"refs/heads/master"},"d90b2faf86b1b0514452a47744dbb946d967d152":{"kind":"REWORK","_number":9,"created":"2023-09-04 17:04:16.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/9"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 17:04:07.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d90b2faf86b1b0514452a47744dbb946d967d152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d90b2faf86b1b0514452a47744dbb946d967d152"}]},"branch":"refs/heads/master"},"f960688da26199f29dc72e15c53c2a60970f354a":{"kind":"REWORK","_number":10,"created":"2023-09-04 17:10:02.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/10"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 17:09:53.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f960688da26199f29dc72e15c53c2a60970f354a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f960688da26199f29dc72e15c53c2a60970f354a"}]},"branch":"refs/heads/master"},"a5a9b8f479dd813b770ab5b04e4fd3c088805692":{"kind":"REWORK","_number":11,"created":"2023-09-04 17:25:24.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/11"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 17:25:14.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5a9b8f479dd813b770ab5b04e4fd3c088805692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5a9b8f479dd813b770ab5b04e4fd3c088805692"}]},"branch":"refs/heads/master"},"fa09ba0bba18f7795a968e9e0138fd83a2f9a86f":{"kind":"REWORK","_number":12,"created":"2023-09-04 17:50:12.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/12"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 17:50:04.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa09ba0bba18f7795a968e9e0138fd83a2f9a86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa09ba0bba18f7795a968e9e0138fd83a2f9a86f"}]},"branch":"refs/heads/master"},"c965fe2005a7f722006f4dd2db41b465f9c570c6":{"kind":"REWORK","_number":13,"created":"2023-09-04 18:11:16.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/13"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 18:11:08.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c965fe2005a7f722006f4dd2db41b465f9c570c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c965fe2005a7f722006f4dd2db41b465f9c570c6"}]},"branch":"refs/heads/master"},"78cef2aab50986abf6d2681ffa7c7ce6bbb0715c":{"kind":"REWORK","_number":14,"created":"2023-09-04 18:54:57.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/14"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 18:54:47.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78cef2aab50986abf6d2681ffa7c7ce6bbb0715c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78cef2aab50986abf6d2681ffa7c7ce6bbb0715c"}]},"branch":"refs/heads/master"},"cae7109d39bfb507f1c115f6e78ad5f44253807a":{"kind":"REWORK","_number":15,"created":"2023-09-04 22:34:23.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/15"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 22:34:15.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs","message":"Temporarily disable IPv6 jobs\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae7109d39bfb507f1c115f6e78ad5f44253807a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae7109d39bfb507f1c115f6e78ad5f44253807a"}]},"branch":"refs/heads/master"},"f849d45d07b62feace2ff9e3a55f4698b8889398":{"kind":"REWORK","_number":16,"created":"2023-09-05 12:59:10.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/16"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-05 12:58:32.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs starting from Bobcat","message":"Temporarily disable IPv6 jobs starting from Bobcat\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f849d45d07b62feace2ff9e3a55f4698b8889398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f849d45d07b62feace2ff9e3a55f4698b8889398"}]},"branch":"refs/heads/master"},"199c0eb3cf3d6b4c70c5c2bb4e4061d91d3ae01a":{"kind":"REWORK","_number":17,"created":"2023-09-05 13:24:34.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/17"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-05 13:23:38.000000000","tz":-180},"subject":"Temporarily disable IPv6 jobs starting from Bobcat","message":"Temporarily disable IPv6 jobs starting from Bobcat\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\nThe ubuntu focal image is now unsupported, so we need to use jammy\ninstead.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/199c0eb3cf3d6b4c70c5c2bb4e4061d91d3ae01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/199c0eb3cf3d6b4c70c5c2bb4e4061d91d3ae01a"}]},"branch":"refs/heads/master"},"021f905f29d8321c231760f21cb68e64162591e7":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2023-09-05 19:46:01.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/18"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-09-05 19:46:01.000000000","tz":0},"subject":"Temporarily disable IPv6 jobs starting from Bobcat","message":"Temporarily disable IPv6 jobs starting from Bobcat\n\nNova recently merged a change that bumps the libvirt version in\nthe Bobcat release. That broke the LVM job, which was still using\nthe Ubuntu Focal release, as ubuntu jammy doesn´t have a candidate\nfor installing quagga.\n\nIn order to have the LVM job passing again, we are disabling the\nIPv6 setup temporarily, until we are able to replace quagga.\n\nThe ubuntu focal image is now unsupported, so we need to use jammy\ninstead.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\nDepends-On: https://review.opendev.org/c/openstack/manila/+/893804\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/021f905f29d8321c231760f21cb68e64162591e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/021f905f29d8321c231760f21cb68e64162591e7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9a5dd277154a8e2b9cec91954cd3d59d4f87482b":{"kind":"REWORK","_number":19,"created":"2023-09-05 21:53:33.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/19"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-05 21:30:21.000000000","tz":-180},"subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","message":"Use Ubuntu Jammy and Temporarily disable scenario tests\n\nA couple of issues happening at the moment:\n\n- Nova recently merged a change that bumps the libvirt version in\nthe Bobcat release [1]. The new libvirt version is not available\nin Ubuntu Focal. In order to get the new libvirt version, our jobs\nthat were using Ubuntu Focal, are now forced to use Jammy, so we\nbump it to the new version in this patch.\n\n- The support for Ubuntu Focal has also been dropped in this\nrelease [2]. Making it one more reason to get the new jammy\nimages.\n\nWhile upgrading the image, some other issues surfaced:\n\n- Some CI jobs are doing IPv6 testing, which use quagga. Quagga\ndoes not have a candidate for installation in Jammy and must be\nreplaced. This issue is already documented [3].\n\n- The inability to set up IPv6, combined with the scenario tests\nbeing enabled in the LVM and CEPH drivers brought up another known\nissue: when doing IPv6 setup, we don\u0027t set up only IPv6 networking\nand we also end up setting crucial information for the scenario\ntests to run in the function that does IPv6 setup. So in case the\njob is only using IPv4 in scenario testing, we currently don\u0027t set\nsome key configuration to let the shares be created/mounted [4].\n\n- Due to the reasons above, the scenario tests will never pass, as\nwe can\u0027t use Focal images, and when using Jammy, we can\u0027t do\nscenario testing.\n\nThis is blocking the Manila gate and all other changes in the CI.\nAs we are forced to use the Jammy image and the scenario tests\nwon\u0027t pass, we disable them in this change, do some refactoring\nin Manila [5], and re-enable them again for IPv4, until we have\nfigured out the quagga replacement.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-September/034968.html\n[3] https://bugs.launchpad.net/manila/+bug/1998489\n[4] https://github.com/openstack/manila/blob/master/devstack/plugin.sh#L1104-L1116\n[5] https://review.opendev.org/c/openstack/manila/+/893804\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9a5dd277154a8e2b9cec91954cd3d59d4f87482b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9a5dd277154a8e2b9cec91954cd3d59d4f87482b"}]},"branch":"refs/heads/master"},"4a4e279dda027ff8462ea57b3b9772f32572bb8e":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2023-09-05 21:55:44.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/20"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-09-05 21:55:44.000000000","tz":0},"subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","message":"Use Ubuntu Jammy and Temporarily disable scenario tests\n\nA couple of issues happening at the moment:\n\n- Nova recently merged a change that bumps the libvirt version in\nthe Bobcat release [1]. The new libvirt version is not available\nin Ubuntu Focal. In order to get the new libvirt version, our jobs\nthat were using Ubuntu Focal, are now forced to use Jammy, so we\nbump it to the new version in this patch.\n\n- The support for Ubuntu Focal has also been dropped in this\nrelease [2]. Making it one more reason to get the new jammy\nimages.\n\nWhile upgrading the image, some other issues surfaced:\n\n- Some CI jobs are doing IPv6 testing, which use quagga. Quagga\ndoes not have a candidate for installation in Jammy and must be\nreplaced. This issue is already documented [3].\n\n- The inability to set up IPv6, combined with the scenario tests\nbeing enabled in the LVM and CEPH drivers brought up another known\nissue: when doing IPv6 setup, we don\u0027t set up only IPv6 networking\nand we also end up setting crucial information for the scenario\ntests to run in the function that does IPv6 setup. So in case the\njob is only using IPv4 in scenario testing, we currently don\u0027t set\nsome key configuration to let the shares be created/mounted [4].\n\n- Due to the reasons above, the scenario tests will never pass, as\nwe can\u0027t use Focal images, and when using Jammy, we can\u0027t do\nscenario testing.\n\nThis is blocking the Manila gate and all other changes in the CI.\nAs we are forced to use the Jammy image and the scenario tests\nwon\u0027t pass, we disable them in this change, do some refactoring\nin Manila [5], and re-enable them again for IPv4, until we have\nfigured out the quagga replacement.\n\nThis change also defines a new job to be used in the 2023.1 and\nZed branches, which can still do the IPv6 setup.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-September/034968.html\n[3] https://bugs.launchpad.net/manila/+bug/1998489\n[4] https://github.com/openstack/manila/blob/master/devstack/plugin.sh#L1104-L1116\n[5] https://review.opendev.org/c/openstack/manila/+/893804\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a4e279dda027ff8462ea57b3b9772f32572bb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a4e279dda027ff8462ea57b3b9772f32572bb8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"91ca00a96f5a98c2282f4d896d1a45ebe013317a":{"kind":"REWORK","_number":21,"created":"2023-09-06 16:31:53.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/893651/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/893651/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/893651/21"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-04 13:09:21.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-06 16:31:31.000000000","tz":-180},"subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","message":"Use Ubuntu Jammy and Temporarily disable scenario tests\n\nA couple of issues happening at the moment:\n\n- Nova recently merged a change that bumps the libvirt version in\nthe Bobcat release [1]. The new libvirt version is not available\nin Ubuntu Focal. In order to get the new libvirt version, our jobs\nthat were using Ubuntu Focal, are now forced to use Jammy, so we\nbump it to the new version in this patch.\n\n- The support for Ubuntu Focal has also been dropped in this\nrelease [2]. Making it one more reason to get the new jammy\nimages.\n\nWhile upgrading the image, some other issues surfaced:\n\n- Some CI jobs are doing IPv6 testing, which use quagga. Quagga\ndoes not have a candidate for installation in Jammy and must be\nreplaced. This issue is already documented [3].\n\n- The inability to set up IPv6, combined with the scenario tests\nbeing enabled in the LVM and CEPH drivers brought up another known\nissue: when doing IPv6 setup, we don\u0027t set up only IPv6 networking\nand we also end up setting crucial information for the scenario\ntests to run in the function that does IPv6 setup. So in case the\njob is only using IPv4 in scenario testing, we currently don\u0027t set\nsome key configuration to let the shares be created/mounted [4].\n\n- Due to the reasons above, the scenario tests will never pass, as\nwe can\u0027t use Focal images, and when using Jammy, we can\u0027t do\nscenario testing.\n\nThis is blocking the Manila gate and all other changes in the CI.\nAs we are forced to use the Jammy image and the scenario tests\nwon\u0027t pass, we disable them in this change, do some refactoring\nin Manila [5], and re-enable them again for IPv4, until we have\nfigured out the quagga replacement.\n\n[1] https://review.opendev.org/c/openstack/nova/+/887255\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-September/034968.html\n[3] https://bugs.launchpad.net/manila/+bug/1998489\n[4] https://github.com/openstack/manila/blob/master/devstack/plugin.sh#L1104-L1116\n[5] https://review.opendev.org/c/openstack/manila/+/893804\n\nChange-Id: Ia45a8e7f7c9cf9959bcc9b9a7243cccbbd1f4916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
