)]}'
{"id":"openstack%2Fmanila-tempest-plugin~894168","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id845c60c39072a03171fa5906461eb254c2736a7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"dd","attention_set":{},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2023-09-13 14:43:48.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-31 13:01:58.000000000","reason":"Change was submitted"},"34373":{"account":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"last_update":"2023-10-31 13:01:58.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-10-31 13:01:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id845c60c39072a03171fa5906461eb254c2736a7","subject":"Optimize the dd command to improve performance","status":"MERGED","created":"2023-09-07 10:47:20.000000000","updated":"2023-10-31 13:01:58.000000000","submitted":"2023-10-31 13:01:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894168-dd","meta_rev_id":"2d93f27f10e491322f31db45bf1cb2a5add1677d","_number":894168,"virtual_id_number":894168,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 13:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"recommended":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"all":[{"value":1,"date":"2023-10-18 14:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":2,"date":"2023-10-31 12:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-18 14:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2023-10-26 20:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":1,"date":"2023-10-30 01:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-10-31 11:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-10-31 11:58:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:21:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 11:54:52.000000000","updated_by":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"reviewer":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"state":"REVIEWER"},{"updated":"2023-09-07 12:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 12:34:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-07 12:59:36.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-07 15:04:25.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2023-09-07 18:31:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-09-08 02:28:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-09-14 00:29:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-10-17 15:15:34.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-10-17 18:27:55.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2023-10-30 01:25:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-10-31 12:34:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"5f0a92554f59ae99324eeca701b457d800e7edbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-07 10:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c31b3e7801dc3a5fac5140de32d8e0a63500fdf7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 11:21:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/1/check/DellEMC_PowerFlex_Manila_dhss_false/3275135/DellEMC_PowerFlex_Manila_dhss_false/328 : SUCCESS in 31m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"60e633a5e276b6c079091f920bf2f3b6678e364e","author":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"date":"2023-09-07 11:54:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6040b55fcfd41a6bbefb7aea22ce7479df7d8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 12:24:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23be0cc77e024488860a6974fcfb016f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ba1d3ff91645428bbdf01df57f9d48 : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/45c06e5b19f3463592052ce3d91d343a : SUCCESS in 42m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0f3c55e6b1b9495e8d2c76b06f9a2b91 : FAILURE in 46m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4701befa9b18446a8abb93d06cf71ef2 : FAILURE in 43m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c42ad7130eab4477bc169ca3870c7f40 : SUCCESS in 1h 08m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b29a7a43e794ad29be2a70b1aa37647 : SUCCESS in 1h 03m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/297ea202cad1449ca212873042161cd2 : FAILURE in 1h 29m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6fa317993fe4787a8357b6161b97e24 : SUCCESS in 37m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/59a548f81878481b84fa5bd7b64aef11 : SUCCESS in 39m 05s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2b1bbed1017f4a319de26f154329a536 : SUCCESS in 34m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/50c15104939241eeb2bea9542ea3156b : SUCCESS in 41m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/91ff1478a16d44aeaf26369f8aa2a9af : SUCCESS in 24m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e5c38fbd076b4ef88a8558c17f133377 : SUCCESS in 52m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d04ac3c02d5344baaf337b5b05dbbe4a : FAILURE in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a2f416b1b42b3776d09f34236accafc931ac8d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-07 12:34:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/1/upstream-check/manila-tempest-plugin-ontap-dhss/3c88749/ : RETRY_LIMIT in 7m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/48ca195/ : RETRY_LIMIT in 8m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ff6513beb6d4dac3af5081c93db4fdfb6326ec4c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 12:54:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/1/check/DellEMC_PowerFlex_Manila_dhss_false/d251827/DellEMC_PowerFlex_Manila_dhss_false/330 : SUCCESS in 30m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9af86584008242eb671b573f715aed60a1f908ca","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-07 12:59:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/1/check/DellEMC_PowerStore_Manila_dhss_false/437527a/DellEMC_PowerStore_Manila_dhss_false/141 : SUCCESS in 35m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6916b473a6e443ba3e7ccea7eab352300b4a14b6","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2023-09-07 15:04:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df8970b0c0d0b0c8da7815ccf5dcd90a35549749","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-07 18:31:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/68/894168/1/check/DellEMC_Unity_Manila_dhss_true/75167a5/DellEMC_Unity_Manila_dhss_true/553 : FAILURE in 51m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/1/check/DellEMC_Unity_Manila_dhss_false/b7aadfd/DellEMC_Unity_Manila_dhss_false/577 : FAILURE in 2h 51m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d3165a9e3dba5306f73f01ecb30345a396404ac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-08 02:28:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcba1d8ce3ead4da89df59f1be82852221f8a8b4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-10 09:26:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ba0a357ece219add4da117fb595b67e2e825727","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 22:19:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8242ca5afe31a9d54920419d48238d30719f486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-13 08:57:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a567ffc7a2cc89cad354b98279b810ce5f9bcb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-13 09:08:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb0df467747e844e40c2552a64c96935553a52cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-13 09:24:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fa0931a8f23fa8393ba4176179e776566842590","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-13 09:53:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/4/check/DellEMC_PowerFlex_Manila_dhss_false/20ce44c/DellEMC_PowerFlex_Manila_dhss_false/364 : SUCCESS in 28m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"8aa3a80aee4a1c579a1db33d921b7a6b4c02c869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 10:55:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/682813bdb21a4abc8a789a19083d3f1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2f24703bb745449677321deb12aff7 : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/579adc9e47244ad18a5cd59204df63f8 : SUCCESS in 35m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4d3bf720126149e99b6cf608f264dc9b : FAILURE in 48m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c3dc702fe9445409cce85bdfbe6aceb : FAILURE in 35m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f8bda3447bb54727aac02071e30a25ec : SUCCESS in 1h 16m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9093b09ee9804a8f9fbb1859bf343536 : SUCCESS in 1h 03m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/54f1699b283743d0a8673668502552b5 : SUCCESS in 1h 23m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4cfc411dd0904f27862e3cec7f000c40 : SUCCESS in 32m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eeab31b9a11b4bb7a18d916d285c40f2 : SUCCESS in 38m 28s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/334b256dbfd849f2ba8e7f24536315e4 : SUCCESS in 56m 08s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1c240efe53b24f35be24cd1032c142b2 : SUCCESS in 31m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c95bc70493ed4661a2459695ef88d60e : SUCCESS in 31m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/211dfbdfb3b5477abcf0ad7838683ed3 : FAILURE in 37m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e2386090c7bb471c8cf719bd708f68d8 : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31ddff79abbf7878a4e45158da5ac8b776a1888e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-13 11:27:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/4/check/DellEMC_PowerFlex_Manila_dhss_false/bc0d0fa/DellEMC_PowerFlex_Manila_dhss_false/365 : SUCCESS in 31m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"00b7fcf161455d27d3b05688d57b1e170f51bfae","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-13 11:33:32.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/4/check/DellEMC_PowerStore_Manila_dhss_false/e4d30a5/DellEMC_PowerStore_Manila_dhss_false/155 : SUCCESS in 37m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"9ef65234a2ec8bb85a7dbf8e2c60d3b1e58873cb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-13 12:55:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/4/upstream-check/manila-tempest-plugin-ontap-dhss/5eff904/ : FAILURE in 1h 52m 27s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/047b2df/ : SUCCESS in 1h 57m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3e345986c2cf5447baecc936034d54ed933e3dfc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-13 14:30:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/68/894168/4/check/DellEMC_Unity_Manila_dhss_true/8bd07e0/DellEMC_Unity_Manila_dhss_true/580 : SUCCESS in 3h 42m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/4/check/DellEMC_Unity_Manila_dhss_false/cc8ad04/DellEMC_Unity_Manila_dhss_false/606 : SUCCESS in 1h 22m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"66ff12033eee4cd9f24fa7e953905795899a405f","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2023-09-13 14:43:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"33535b900d3e62c722cfbba5adb1dc97780787c2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-14 00:29:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79848eb96f4acabef0e6bd25aa80bfc06269f05f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-10-17 15:15:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"988761a5ec9adbdb7ec8fe1a24092c0e5cc72e30","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-10-17 18:27:55.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef144b00ce7db51b4938d9a9941aa5cb57e3ac06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-18 14:20:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cb45e95bb29b3c217c0531a1e20cc02a2424a5d0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 14:52:07.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_PowerFlex_Manila_dhss_false/ecaefb6/DellEMC_PowerFlex_Manila_dhss_false/577 : SUCCESS in 31m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"3fee57fd84206f39cfcfece27cd002b567da2c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 17:24:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d29888312a21460588f7c6c748101f0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0631ffaf9bc24b9da9c60f7b13a4fb1b : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/61c4f51a64564bf79b0c1232e970b623 : FAILURE in 39m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/82830d164d0b4a56ba4923dd1affe4f5 : SUCCESS in 2h 20m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd98a42259c846be95340b388057cfcd : TIMED_OUT in 3h 02m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/de291270171e4cbabb28019d0f4a3b43 : SUCCESS in 57m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6cd40db065604a668f648651c86d1713 : SUCCESS in 1h 45m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/582fa93ef381499cb49cf95a6c47c480 : SUCCESS in 1h 07m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ce8553b2cc034f79af1f56311bd5f7b0 : SUCCESS in 29m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd3047dd39a545e7bb56959b4cb8daec : SUCCESS in 46m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/09078bafd96746038759860e1955621f : SUCCESS in 1h 17m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1f394b86f348446ea2a6339b09e9c6fb : SUCCESS in 52m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d473866be6bd43abb290ead49bf2e99f : FAILURE in 22m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/90a3662f86824892a3b8367398531010 : SUCCESS in 49m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25d2d15c31564a8693e78b0f4d2bc262 : FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb0580c7f44dc2368c4f65ec7b70ec5bdc58c109","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-18 18:34:56.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_Unity_Manila_dhss_true/8cd0488/DellEMC_Unity_Manila_dhss_true/680 : SUCCESS in 4h 08m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_Unity_Manila_dhss_false/b75b83d/DellEMC_Unity_Manila_dhss_false/717 : SUCCESS in 1h 03m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a8e93eeec1998a41617844ba81c6339a44c5af5d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-26 02:57:16.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8dc539cd2e56a8b6517ec776aebe3cc1ea26f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 04:49:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65e70a0089584a9bb06c09a0a6edc941\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82b2bb1d77343739fde30e35a2c89cf : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0814204d516d4c18a034ab9154ca8b08 : SUCCESS in 1h 16m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5ff424891ea94bd8923bcc9a4b8f6b72 : SUCCESS in 1h 45m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6001075962b45b8a33296be5f4fa1b6 : SUCCESS in 1h 14m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b676a3b75880484197e6096a96b63376 : FAILURE in 54m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ea73c4cc44d84d5a8045f19529190447 : SUCCESS in 1h 42m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7c87cce76434481a13212571b59f5c0 : SUCCESS in 1h 07m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/41bf49326af8496d8315ca8854380f46 : SUCCESS in 27m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/22a8629fa55542f8a66f45e9bfaca282 : SUCCESS in 30m 02s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/dc0eb784ccb74545aa7e4ffacfc2f752 : SUCCESS in 1h 19m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/00ea1fcb3693403b90e0c3b3ff16dc04 : SUCCESS in 1h 19m 52s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9c49f7d01488440d8c0427c224c1d375 : SUCCESS in 37m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f8a127e5480d4d99ae08639883153aa0 : SUCCESS in 52m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0d915dd60654abf9fe73041caf06a3c : FAILURE in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0831f5a59e90f665b1064fa5c87714255dbf7e41","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-26 05:06:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/5/upstream-check/manila-tempest-plugin-ontap-dhss/43b95ed/ : RETRY_LIMIT in 8m 52s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/894168/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/d7d58ff/ : RETRY_LIMIT in 10m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9d81f8f421b8181fcb3922b936c09998b0ca4899","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-26 05:20:03.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_PowerFlex_Manila_dhss_false/b0d582f/DellEMC_PowerFlex_Manila_dhss_false/599 : SUCCESS in 30m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"2cb2d507d6a3ba55e0ce810b7a2d3a6021d6eeab","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-26 05:27:07.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_PowerStore_Manila_dhss_false/02e15c9/DellEMC_PowerStore_Manila_dhss_false/257 : SUCCESS in 37m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"d2df8eebc0f6bc9212751ef5b5dc5089d471c945","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-26 08:08:13.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_Unity_Manila_dhss_true/4b57b65/DellEMC_Unity_Manila_dhss_true/700 : SUCCESS in 2h 43m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/68/894168/5/check/DellEMC_Unity_Manila_dhss_false/51f23c6/DellEMC_Unity_Manila_dhss_false/739 : SUCCESS in 1h 18m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e56cf3f59d007a8677ec39351daef8f2a0acdf51","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-26 20:28:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aecf8a4b540d44a570f82db522c90afc6d0e729f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:25:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7af4275b0ad4cdcf19a82a36cf9a2b323855317f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-10-31 11:57:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8cff094cfd8edc845cc44bdb45587bb08336e6f0","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-10-31 11:58:05.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a4624c9298e7521277cf467ba4a5014ff373d656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:58:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"30921333c5be0e34ce113064727f30bd42ec6cf6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-31 12:34:38.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"668c0f14fff3c64ef5fd63a5d1e670a70496955f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:01:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5797cf3af4aa4d74bf5930ad702de3f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c385ee56987f406aa381e43412365c45 : SUCCESS in 3m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a419f24051264bbbb93c5c5e982fc287 : SUCCESS in 1h 01m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ba1fb7f2254149a685b2623f50c2706b : SUCCESS in 24m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed368dcd7b874c639331fd7596de90f6 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":5},{"id":"2d93f27f10e491322f31db45bf1cb2a5add1677d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:01:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"192d408d700d39df3a74edb99da853ede01b5b82","revisions":{"f73cc69522a4c98752d715b224ebb3e04179d5ac":{"kind":"REWORK","_number":1,"created":"2023-09-07 10:47:20.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/68/894168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/894168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/1"}}},"commit":{"parents":[{"commit":"91ca00a96f5a98c2282f4d896d1a45ebe013317a","subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"subject":"Optimize the dd command to improve performance","message":"Optimize the dd command to improve performance\n\nUse if\u003d/dev/zero: Instead of reading from /dev/urandom,\nwhich is slow for generating random data, use /dev/zero\nto write zeros to the disk. Zeros are generated much faster.\nOmit conv\u003dfsync: The conv\u003dfsync option forces dd to\nsynchronize the data after each write, which can be slow for\nlarge writes.\n\nChange-Id: Id845c60c39072a03171fa5906461eb254c2736a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f73cc69522a4c98752d715b224ebb3e04179d5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f73cc69522a4c98752d715b224ebb3e04179d5ac"}]},"branch":"refs/heads/master"},"f9d4995d977d2187bf6fadba7f5a355368d6a522":{"kind":"REWORK","_number":2,"created":"2023-09-13 08:57:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/68/894168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/894168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/2"}}},"commit":{"parents":[{"commit":"91ca00a96f5a98c2282f4d896d1a45ebe013317a","subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-13 08:57:13.000000000","tz":180},"subject":"Optimize the dd command to improve performance","message":"Optimize the dd command to improve performance\n\nUse if\u003d/dev/zero: Instead of reading from /dev/urandom,\nwhich is slow for generating random data, use /dev/zero\nto write zeros to the disk. Zeros are generated much faster.\nOmit conv\u003dfsync: The conv\u003dfsync option forces dd to\nsynchronize the data after each write, which can be slow for\nlarge writes.\n\nChange-Id: Id845c60c39072a03171fa5906461eb254c2736a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9d4995d977d2187bf6fadba7f5a355368d6a522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9d4995d977d2187bf6fadba7f5a355368d6a522"}]},"branch":"refs/heads/master"},"87373b4d9e5a6aa1a9175835858b09bcdbf6e42b":{"kind":"REWORK","_number":3,"created":"2023-09-13 09:08:40.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/68/894168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/894168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/3"}}},"commit":{"parents":[{"commit":"91ca00a96f5a98c2282f4d896d1a45ebe013317a","subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-13 09:08:29.000000000","tz":180},"subject":"Optimize the dd command to improve performance","message":"Optimize the dd command to improve performance\n\nUse if\u003d/dev/zero: Instead of reading from /dev/urandom,\nwhich is slow for generating random data, use /dev/zero\nto write zeros to the disk. Zeros are generated much faster.\nOmit conv\u003dfsync: The conv\u003dfsync option forces dd to\nsynchronize the data after each write, which can be slow for\nlarge writes.\n\nChange-Id: Id845c60c39072a03171fa5906461eb254c2736a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/87373b4d9e5a6aa1a9175835858b09bcdbf6e42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/87373b4d9e5a6aa1a9175835858b09bcdbf6e42b"}]},"branch":"refs/heads/master"},"7100d78cdf8d833bc12e11cf93b90affe32b8bf8":{"kind":"REWORK","_number":4,"created":"2023-09-13 09:24:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/68/894168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/894168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/4"}}},"commit":{"parents":[{"commit":"91ca00a96f5a98c2282f4d896d1a45ebe013317a","subject":"Use Ubuntu Jammy and Temporarily disable scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91ca00a96f5a98c2282f4d896d1a45ebe013317a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-13 09:24:40.000000000","tz":180},"subject":"Optimize the dd command to improve performance","message":"Optimize the dd command to improve performance\n\nUse if\u003d/dev/zero: Instead of reading from /dev/urandom,\nwhich is slow for generating random data, use /dev/zero\nto write zeros to the disk. Zeros are generated much faster.\nOmit conv\u003dfsync: The conv\u003dfsync option forces dd to\nsynchronize the data after each write, which can be slow for\nlarge writes.\n\nChange-Id: Id845c60c39072a03171fa5906461eb254c2736a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7100d78cdf8d833bc12e11cf93b90affe32b8bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7100d78cdf8d833bc12e11cf93b90affe32b8bf8"}]},"branch":"refs/heads/master"},"192d408d700d39df3a74edb99da853ede01b5b82":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-10-18 14:20:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/68/894168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/894168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/894168/5"}}},"commit":{"parents":[{"commit":"1ba0a8d94cd860abf261d2b9c89c39f768bf580d","subject":"Re-enable IPv4 scenario tests in LVM driver job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba0a8d94cd860abf261d2b9c89c39f768bf580d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-09-07 10:44:31.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-18 14:20:08.000000000","tz":0},"subject":"Optimize the dd command to improve performance","message":"Optimize the dd command to improve performance\n\nUse if\u003d/dev/zero: Instead of reading from /dev/urandom,\nwhich is slow for generating random data, use /dev/zero\nto write zeros to the disk. Zeros are generated much faster.\nOmit conv\u003dfsync: The conv\u003dfsync option forces dd to\nsynchronize the data after each write, which can be slow for\nlarge writes.\n\nChange-Id: Id845c60c39072a03171fa5906461eb254c2736a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/192d408d700d39df3a74edb99da853ede01b5b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/192d408d700d39df3a74edb99da853ede01b5b82"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
