)]}'
{"id":"openstack%2Fmanila-tempest-plugin~895211","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If341dda351a9ffe6d133bc2207d5e7d5342b8547","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1998489","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-07-28 12:27:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-07-28 16:33:13.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-07-28 13:51:24.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-07-28 16:33:13.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-07-28 16:33:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If341dda351a9ffe6d133bc2207d5e7d5342b8547","subject":"Re-enable IPv6 tests","status":"MERGED","created":"2023-09-14 19:44:42.000000000","updated":"2026-07-28 16:33:13.000000000","submitted":"2026-07-28 16:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895211-bug/1998489","meta_rev_id":"3e0958116b81c5391816076303b3892121da8a47","_number":895211,"virtual_id_number":895211,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-28 16:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-07-28 12:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-07-28 13:51:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-07-28 15:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 20:18:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-15 11:21:52.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2026-04-16 17:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-19 16:39:15.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-27 16:46:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-27 16:46:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-07-27 16:46:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-07-27 16:47:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"60c26984932dc2eb7e839987896249f34b1747de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-14 19:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecaf9f837567d3f9a01400c7e9f140bd3b085be2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-14 20:18:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/11/895211/1/check/DellEMC_PowerFlex_Manila_dhss_false/4256274/DellEMC_PowerFlex_Manila_dhss_false/383 : SUCCESS in 33m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca6c438c599e524c807c78443914bf0af2cc2f91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-15 00:06:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"210f838f4463d82af40129cd73b9c3fb397fe400","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-15 00:41:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/11/895211/2/check/DellEMC_PowerFlex_Manila_dhss_false/f16b169/DellEMC_PowerFlex_Manila_dhss_false/387 : SUCCESS in 34m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a594648a06a9b9515800864720f608d60e3180e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-15 11:21:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/11/895211/2/check/DellEMC_Unity_Manila_dhss_true/d188135/DellEMC_Unity_Manila_dhss_true/593 : SUCCESS in 3h 14m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/11/895211/2/check/DellEMC_Unity_Manila_dhss_false/1a3d19c/DellEMC_Unity_Manila_dhss_false/623 : SUCCESS in 1h 13m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"539ab1194d84c872f1459aca3346692ff70fdac6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-19 18:48:22.000000000","message":"Topic bgp removed","accounts_in_message":[],"_revision_number":2},{"id":"fe30365479ca50f321eee5e1bfb806f1440e6983","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-19 18:48:26.000000000","message":"Topic set to bug/1998489","accounts_in_message":[],"_revision_number":2},{"id":"d058c362ce6a312ca49e4b656a732d119f862a72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-20 22:37:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c749ecf2cee6941bc5b1431799a6aab05de48c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 06:47:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"842438005cf1f94e65215002de9f0bdac0306521","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 06:47:27.000000000","message":"Topic bug/1998489-ipv6-test-jobs removed","accounts_in_message":[],"_revision_number":4},{"id":"2b05a53f727445bcab52991593a63e86dbd33eaf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 06:47:31.000000000","message":"Topic set to bug/1998489","accounts_in_message":[],"_revision_number":4},{"id":"9eb95784771cf739ab54150803a467d26009f52d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 17:14:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d8bfe9d4de8e2c1aa5aacb8ae431d457eb9453c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 17:17:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"336f37c6f49b2ca87cf7a4092a2e0cbfcebbe43a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 17:18:42.000000000","message":"Topic bug/1998489-ipv6-test-jobs removed","accounts_in_message":[],"_revision_number":5},{"id":"223d78afba51525f04f51e6c82b5ecb57484ae8c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 17:18:47.000000000","message":"Topic set to bug/1998489","accounts_in_message":[],"_revision_number":5},{"id":"5281e1bfef30831b44e96dec53cc6e7903844e77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:50:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e85cc2c396524c279b810beb3d137242\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66460ad895294910b7ad8d33e50f2843 : FAILURE in 28m 43s","accounts_in_message":[],"_revision_number":5},{"id":"27eeaf0f85385c52b4d7a0d1e3a030790b8b0af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-16 21:06:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"68b59eea74d6f52d5ecf0aa8fc9acc53e8f7ee58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 21:58:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9e16a5f494a49d2a392d084346ca144\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b4086ad505bb4513b3d50222f0303865 : FAILURE in 50m 25s","accounts_in_message":[],"_revision_number":6},{"id":"22eb70301dac9a3c8dc0c3b717ddd296492d819c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 05:57:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bf07ce22e1709e479be841638469d0c46b2003f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 07:20:40.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a532c2b492e24cba82eb509d626b98b2\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82b2195b40d8485cb95921bf1c297430 : FAILURE in 1h 17m 17s","accounts_in_message":[],"_revision_number":6},{"id":"8e83ddd0ebdec88ff258b333d7310becff34b415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 17:10:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"fe6a2fc5f6c7fa0009c66a5c8e2d85a71a56a373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:16:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f1e7ae0dd9b4dafb353f7e91e14eb72\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb14f0c12fb84fbda59242a5d3c11390 : FAILURE in 1h 02m 32s","accounts_in_message":[],"_revision_number":7},{"id":"46dd98507b190224a3e2befbd1d253b89a5cae11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 18:51:09.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"00c4dfcf49be57d279b57687ad6312053cad1817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 19:59:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11381bfbe1684c6ca609b9f45028be4b\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/532c0166d5344530a7b92d794855bb55 : FAILURE in 1h 01m 28s","accounts_in_message":[],"_revision_number":8},{"id":"772ed0847ea8202568157118f849282ac71a3b8c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 20:35:05.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3056c3f22217fca00e45b19cd38c749d2e73d9c5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 20:35:16.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"d0e8d0e9c1bdfa4f0d43bb1628662e577f778b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 21:54:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bff501043e544d71b0fe523e564b3a62\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fe2de88dbbc94582875760bb66630c7a : FAILURE in 1h 12m 29s","accounts_in_message":[],"_revision_number":8},{"id":"88aaf80460760e4b95d45b677aa35dfee4062511","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-17 22:45:14.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"202024550e40a4589611c66c56a4eb1b2d514452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 23:47:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ba3390def814a02bfbcb19703a27795\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f42bf262f24147e19a64385b869507bc : FAILURE in 1h 00m 10s","accounts_in_message":[],"_revision_number":8},{"id":"6000513ffd665ab14f55960e570e8e61532d6a0d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-18 02:21:05.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"0d602715e5ada2e29a65238bb6faa066838f4938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 03:55:38.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e6f35756e184b2eaa6b5e9d0311306a\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1d5d30f56af847f38bf4344f4207ae0e : FAILURE in 1h 03m 11s","accounts_in_message":[],"_revision_number":8},{"id":"ccb20ae83d220e3450dc60910bb8f531df9e2163","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-18 06:00:38.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"d78cc128e830b7a8ca4554b96ab0b794ba9b3db0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-18 06:34:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"06ed0468ce9dacfacc017bbce5a659fe83c69a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 07:17:22.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfb780330efb4792b302a5e830f79c6f\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b116995d319242ec98ce7ed6426ca48d : POST_FAILURE in 41m 33s","accounts_in_message":[],"_revision_number":8},{"id":"5eaa3337d950f1741497f7762eba66e75db019ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-18 19:49:59.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"09cf47b36d3dc9500ef21b5357b25ee29f253cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 21:04:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c8c7510e06e467d93a4123c4c20fad4\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/92fcf5885a1a495bbc3b9285e9955fa2 : POST_FAILURE in 1h 09m 13s","accounts_in_message":[],"_revision_number":9},{"id":"7cce3eabb358ed8abef2a83d199d2967ce9eac5e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-19 15:27:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"866dbedc8af82179fc5436e6873c94c6e69b655e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-19 16:36:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997ea6104f6b43d585f638fc32ce5d26\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cd660dfc5ff3496bb39a5beaccdf8faf : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":9},{"id":"079c167492ee1f3f0c69028f4173e8a703de493b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-19 16:39:15.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/8f88e17e41b549028185f8ecba8ed7d8 : RETRY_LIMIT in 1m 28s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/f19c9b41bfb44ff1a4cc0878fa138a88 : RETRY_LIMIT in 1m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"1cbd4bbe0823cca140f5393ddec72a05df8ead7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-19 20:50:27.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"2208cde16199d5597f28669c5a1d4bac9ef7befe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-19 21:36:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/081e43a37d0144b88bb275686ca1af11\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e6ac8d95c34447f792e6634fcc9a3784 : FAILURE in 44m 41s","accounts_in_message":[],"_revision_number":10},{"id":"8f54393582dddac21e10bf37156245c9735b0a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-21 05:08:49.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"40b9ebd6bea6a900a9e7d572c4d0fd4099e39703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 06:02:11.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30e2df0ed4d04666820326ffd537b003\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2ebd696b81de4748a24097dcad2fb9d4 : FAILURE in 51m 06s","accounts_in_message":[],"_revision_number":11},{"id":"5285fe778ac918109d2d6e2a078ef78963452c94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-13 17:07:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"6ca9efb13628dcebd360c0799ce1c7ab7478696b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 19:34:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065c2e42db7a4a97b35b033f6b17d8d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfac11ec0c3496592afffe7977d1387 : SUCCESS in 3m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aec4942713d147d18adb39ef18f0e26a : SUCCESS in 57m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/26afdd274d7d420ca0e8f76a906255fe : SUCCESS in 2h 19m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2080e32340e54d959e2b238a503a6056 : SUCCESS in 41m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4bfa42ec0d5a4af6852b4e0e1281204d : SUCCESS in 50m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/768e9dddb1e5402d9bff89b73f893eb2 : SUCCESS in 45m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/25e99347924f45258e0269b94d9ea8dc : FAILURE in 1h 27m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/228b88a69bbb446ca299c12d76c69f00 : SUCCESS in 32m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2c1a5ecbdca74612add4e11cbda575ad : SUCCESS in 52m 55s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/a4e189bf88d74431b31540378e8f225a : SUCCESS in 1h 14m 03s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/a120984daf8140cd884e6d00fb46c3ec : SUCCESS in 1h 14m 50s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/08fc5c68478e4998bce91decba1875b8 : SUCCESS in 1h 18m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ab5c8d01624342c6bc27ab5a5fca8c57 : SUCCESS in 40m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/04635c9b9d9d456184980663d694cff5 : FAILURE in 18m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9de8f05c42f1605b52c7452580f13e8469f8bde2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-13 20:53:22.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/12/upstream-check-manila/manila-tempest-plugin-ontap-dhss/506aba6/ : FAILURE in 1h 16m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"31b679275270df884f335ba563b19dfbbdd61247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-13 21:40:38.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"e55a3cffb5b486fbfd55a95810f25aaebd640f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 00:11:54.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8363c1624d8f4891bae9b21b88fb15ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9cfd452ec84c2cb52dccec0acb342c : SUCCESS in 1m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49e8810002ab475f9108dbc45523162f : SUCCESS in 1h 47m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f0dbd017ec0a4e709f7e909c24999bd3 : FAILURE in 2h 22m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33efee1ce10a4d808171699b9303b74f : SUCCESS in 1h 08m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/20e8410f2160404ea83a4c328e2bb277 : SUCCESS in 55m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f9d8fccf36244f62b67825e57339dc23 : FAILURE in 34m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c45a2305b1534e79bcbd23eaf9056ac1 : FAILURE in 1h 17m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/887ef08e5a574bd98aac763d7235059b : SUCCESS in 34m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4f692d598d3d422db5d4837a6fa89fec : SUCCESS in 43m 39s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/54bd7ede1ca449ce9191e25320c3d274 : SUCCESS in 1h 17m 49s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/3cbf4d32d4f34db7b95fa4b9cc2ebd7b : SUCCESS in 36m 20s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0c704f9b767b49548df1c51036113099 : SUCCESS in 1h 16m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/34510e55d97d462e9b8e50aecbdcc2d9 : SUCCESS in 22m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7704c827e7947f1bb2a8ed1d57b4236 : FAILURE in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1672652d67ca06be024926f51fe3e48fd5bc6382","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-14 01:38:50.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/13/upstream-check-manila/manila-tempest-plugin-ontap-dhss/69f3337/ : POST_FAILURE in 1h 23m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"90d596bf91f70bc37bf48947e546887e403be2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-15 14:04:23.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"1ce99682be406077cfc7a0ff6315fb207a5fea72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:02:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab66edcfcb564a85b1619bbc935cc311\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680aecf35f2d4bb79f74f292b8fc0851 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/58f26ea218424924ab3d0a12a4e49f64 : SUCCESS in 1h 00m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b99c71be0f924403a9dc9cba24d91a5f : SUCCESS in 1h 37m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/76c5a60a39d04de19fd703314f383c31 : SUCCESS in 1h 12m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/35f80e0a97774de98f119d26b438f7b9 : SUCCESS in 52m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/fd7941a4bbbb4e568b882ef9604ddda8 : POST_FAILURE in 1h 50m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f90db31a38ee47dabbc4532112e3d38e : SUCCESS in 1h 11m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6992b11dd44c44d49781516c0bf85c37 : SUCCESS in 35m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af15f9145bb843d3b447517e616828cb : SUCCESS in 38m 35s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/d788f1459e8e4ee49ccc08a1bbbcfa22 : SUCCESS in 57m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/465a24ab03d14be4aa5e043faea1f3d7 : SUCCESS in 1h 13m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/18cea145ed7441e081a9c872c093d4ad : SUCCESS in 1h 11m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d6b1018550b54a6693eb85a8f7b50085 : SUCCESS in 35m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1203774297e745d1b5ca967b8c5973e1 : FAILURE in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ef10526a7895eeae2fdbcdcb31ddd3f40abff1db","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-15 17:19:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"523a183b6d1152f209f8b71e7667d9fb49ee7cf9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-15 17:20:33.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss/95c4cc1/ : FAILURE in 1h 15m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"3f49f084dff872b88c0ec22429d0a805a20289c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 19:19:51.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efcc1fdce7c4485692f4da156fd8d575\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f02a1ad5cb140b58dd9f89ff31a677f : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b51ff68e20d4d06a60be4136a6281b4 : SUCCESS in 1h 02m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e2e001ef0c024cfa884d3c71460b8f97 : SUCCESS in 1h 34m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ddf7b9d631148ae91f0cc03d01932a5 : SUCCESS in 1h 14m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9e33d8cac684400388d14965615cd8da : SUCCESS in 1h 07m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1269a0d017ab4767a4a7e032d9caa89c : FAILURE in 1h 53m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b830930817aa4a5cb95093cbe29b1ae3 : FAILURE in 1h 01m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/87aa1ee1c6fe4e44b6e21686f442a72b : SUCCESS in 23m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/df9a3ccd9476424a87cdb28f3a2a1df3 : SUCCESS in 35m 13s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/47077c46e2aa4ff388b06f98cd791b83 : SUCCESS in 1h 06m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/f261f67fa90a4ebb9d04c17d945353ec : SUCCESS in 1h 10m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2fc5877af6fb4b1382622d41f9bf49a0 : SUCCESS in 55m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/88f3564893bf418f92b073455f5cc85f : SUCCESS in 37m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5cc069a3cb4347eaa0bc0ecd47edc873 : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1ea2b2ae31e15a20d127b0a51380eaf59910cf7e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-15 20:45:25.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss/e1e73ab/ : FAILURE in 1h 22m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"3b95b036c97ec4cef71a2f4a9a875d2285fbd53f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-15 21:23:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"17458e9029f8f07d3d122dceb2f1b484a93eb223","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-15 22:05:08.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"4a631038e12aea9dd9636c0546012f9527401974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-15 22:05:14.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.","accounts_in_message":[],"_revision_number":16},{"id":"c8b9f04bccc7016a00a296a68613c190fe8f23b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 00:30:52.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c458e2de744a219dc89bcd517a4685\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c9ebf14ad44ea09f4f70d219c35c21 : SUCCESS in 4m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/65c06b1e06ae4e2794a225977a3d9f1c : SUCCESS in 36m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0645a0554a3343aaaacb1c083d72f1ec : SUCCESS in 1h 22m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/98d1bda3cfd04e34a983552644a4ed78 : SUCCESS in 47m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/19a4e87e0cbb4b65b108e2a9376a2bd0 : SUCCESS in 1h 07m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c0d98ff66a3c42888197434f5ab20b09 : FAILURE in 2h 15m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49fb0c4ae72b4f38aa4c2370fb502ff7 : SUCCESS in 49m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f184b24931104571b224bc5e902cb469 : SUCCESS in 36m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/34fd07ee8d8043298dde1e2da72e8092 : SUCCESS in 38m 33s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/f0186ebedc874e2982f7c23a08f091c0 : SUCCESS in 50m 52s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/3de033142b0147e2b278e4845f9ae789 : SUCCESS in 1h 20m 57s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/7c7e4e4239554ae9b0708991f04cf71e : SUCCESS in 1h 49m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7f332ee478344b5d9dfb07ed7eac1529 : SUCCESS in 30m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b2e1c2c03f9d4ee6947913738392f4c7 : FAILURE in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"2eca119599337ae729add95a58d67edbc792d88d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-16 01:56:38.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/16/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ad51f12/ : FAILURE in 1h 23m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"5860af5979adfe749dc802325d8a0f9040012dd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-27 00:09:13.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"2a134a181c9d37b809fdc49324119f63132dc37b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 02:24:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29670fbcae0e4ad685dc661680832dc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a37d024a784b1cb9163ae739a8f6da : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/987eca455a5a47ff975387521e7c3821 : SUCCESS in 32m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/686fe0b38b2546f8a2fd9f31e437ea4f : SUCCESS in 1h 22m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0d70ed5d4423414782134e27b3bbd437 : SUCCESS in 1h 26m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5ec1a6dfda774343b57dacf993ed9e9f : SUCCESS in 1h 07m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d07550c6fe1a47e3aa47bb8e2208a8d3 : FAILURE in 2h 08m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f70cf185bbcb415bb33cb710c13b8932 : SUCCESS in 1h 06m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1b4d754b1782435b90d926290e9dfa15 : SUCCESS in 34m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d007f3a02314ac8a99ca2fe2eb6a66f : SUCCESS in 35m 35s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/4dc7bbd1a70046c3b8dc7fd926205911 : SUCCESS in 1h 00m 10s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/f955fff809cc42bd9da7505992a54496 : SUCCESS in 1h 09m 44s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/de092ea42bb9481d81e327348ae8dfb8 : SUCCESS in 46m 10s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/3dc7079d810f49929624bc7c372b2cfe : SUCCESS in 30m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/eedeb4c597c04bb091ef6cf0014b0761 : FAILURE in 54m 44s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b6e5dd7bc0f32afd435107968ebcee52e2140515","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-27 02:32:30.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/11/895211/17/upstream-check-manila/manila-tempest-plugin-ontap-dhss/6b790b2/ : RETRY_LIMIT in 7m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"38835e07507f59c925ac3f2558b071997875b595","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-27 16:46:41.000000000","message":"Topic bug/2148529-prove-bug removed","accounts_in_message":[],"_revision_number":17},{"id":"12795274d6fc7136406847e8650d1a806d61a3fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-27 16:46:47.000000000","message":"Topic set to bug/1998489","accounts_in_message":[],"_revision_number":17},{"id":"2655c67fb2d2bf31588901cfeda7129e92f375d9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-28 12:27:36.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"683762cea1291f7ee42ec4be42326d85b467074c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-28 13:51:24.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"ca7c28b1a960a148756333b42eac593d8271f2b9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-28 15:29:53.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"0719a54b069b9ce79d6a5ef831c8a8df87c47699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 15:31:10.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"476a7b6ca22416f65fa76832a2c6eaab72484c3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 16:33:13.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b021897b3aca4d679904d3a1c47a467d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3098406a3844d4b0d817e9b1cbdc0b : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b444c4c7f9a84434b0bb9b9d47d6977c : SUCCESS in 51m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0296a65ee4bf4246a6818713a5ac4be9 : SUCCESS in 41m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d1267ee24ded441a8fc80c3b072dd69a : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":17},{"id":"3e0958116b81c5391816076303b3892121da8a47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 16:33:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"cc5da846cebbcb9f6d8598fd5b7899ad0cfc16d6","revisions":{"53dde6cc72e50c25c9359d81ec3da3f161a3d3dd":{"kind":"REWORK","_number":1,"created":"2023-09-14 19:44:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/1"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-14 19:44:10.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-14 19:44:10.000000000","tz":-420},"subject":"[DNM] Ipv6 test jobs","message":"[DNM] Ipv6 test jobs\n\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53dde6cc72e50c25c9359d81ec3da3f161a3d3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53dde6cc72e50c25c9359d81ec3da3f161a3d3dd"}]},"branch":"refs/heads/master"},"4b5061682087f9e5ed1a816798b1c89c36272a1b":{"kind":"REWORK","_number":2,"created":"2023-09-15 00:06:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/2"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-14 19:44:10.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-15 00:06:57.000000000","tz":0},"subject":"[DNM] Ipv6 test jobs","message":"[DNM] Ipv6 test jobs\n\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b5061682087f9e5ed1a816798b1c89c36272a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b5061682087f9e5ed1a816798b1c89c36272a1b"}]},"branch":"refs/heads/master"},"1cf100c2b513e69d95c98d624f3946ae24351695":{"kind":"REWORK","_number":3,"created":"2025-11-20 22:37:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/3"}}},"commit":{"parents":[{"commit":"283b664c8be0655bcf370bce10c851cf8dacd2c6","subject":"Merge \"Stop attempting share type deletions that\u0027d fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/283b664c8be0655bcf370bce10c851cf8dacd2c6"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-14 19:44:10.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-20 22:37:11.000000000","tz":-480},"subject":"[DNM] Ipv6 test jobs","message":"[DNM] Ipv6 test jobs\n\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1cf100c2b513e69d95c98d624f3946ae24351695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1cf100c2b513e69d95c98d624f3946ae24351695"}]},"branch":"refs/heads/master"},"51e3ee0fc70ffa7cf6da1363525fe1f74b37e2d2":{"kind":"REWORK","_number":4,"created":"2026-04-16 06:47:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/4"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51e3ee0fc70ffa7cf6da1363525fe1f74b37e2d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51e3ee0fc70ffa7cf6da1363525fe1f74b37e2d2"}]},"branch":"refs/heads/master"},"16ea1c7071685387ce6904c97ae3be9e57720718":{"kind":"REWORK","_number":5,"created":"2026-04-16 17:17:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/5"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 17:16:53.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16ea1c7071685387ce6904c97ae3be9e57720718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16ea1c7071685387ce6904c97ae3be9e57720718"}]},"branch":"refs/heads/master"},"5e12ce437df50d1652bae4822076b456c014c465":{"kind":"REWORK","_number":6,"created":"2026-04-16 21:06:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/6"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 21:06:13.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e12ce437df50d1652bae4822076b456c014c465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e12ce437df50d1652bae4822076b456c014c465"}]},"branch":"refs/heads/master"},"86573a924165257eec70a536693aef3e01401353":{"kind":"REWORK","_number":7,"created":"2026-04-17 17:10:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/7"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-17 17:10:49.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86573a924165257eec70a536693aef3e01401353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86573a924165257eec70a536693aef3e01401353"}]},"branch":"refs/heads/master"},"23e61cf21ddab8163cd83dc20c4afb5fd6177e83":{"kind":"REWORK","_number":8,"created":"2026-04-17 18:51:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/8"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-17 18:50:54.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/23e61cf21ddab8163cd83dc20c4afb5fd6177e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/23e61cf21ddab8163cd83dc20c4afb5fd6177e83"}]},"branch":"refs/heads/master"},"ccac5ab6759b335b84b29478a7f7388bb171cdb3":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-04-18 19:49:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/9"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-18 19:49:51.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/985233\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccac5ab6759b335b84b29478a7f7388bb171cdb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccac5ab6759b335b84b29478a7f7388bb171cdb3"}]},"branch":"refs/heads/master"},"47cfca0c389f556c264375c7739fc2f9c5c7ed6b":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-04-19 20:50:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/10"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-19 20:50:19.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47cfca0c389f556c264375c7739fc2f9c5c7ed6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47cfca0c389f556c264375c7739fc2f9c5c7ed6b"}]},"branch":"refs/heads/master"},"e7fa3b2d607dbf8b1bda42f047ec98c09329a6d8":{"kind":"REWORK","_number":11,"created":"2026-04-21 05:08:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/11"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-21 05:08:17.000000000","tz":-420},"subject":"[DNM] Re-enable IPv6 test jobs","message":"[DNM] Re-enable IPv6 test jobs\n\nRe-enable IPv6 testing in the LVM job now that\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7fa3b2d607dbf8b1bda42f047ec98c09329a6d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7fa3b2d607dbf8b1bda42f047ec98c09329a6d8"}]},"branch":"refs/heads/master"},"ba933ef49b908091bc8d18d7589cd1446686d1af":{"kind":"REWORK","_number":12,"created":"2026-05-13 17:07:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/12"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-13 17:06:53.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba933ef49b908091bc8d18d7589cd1446686d1af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba933ef49b908091bc8d18d7589cd1446686d1af"}]},"branch":"refs/heads/master"},"207058705937f3b58ccc9109d1d17a4ee198f8f2":{"kind":"REWORK","_number":13,"created":"2026-05-13 21:40:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/13"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-13 21:40:29.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/207058705937f3b58ccc9109d1d17a4ee198f8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/207058705937f3b58ccc9109d1d17a4ee198f8f2"}]},"branch":"refs/heads/master"},"cf306083f5f23798bb0522867ad224df3c636599":{"kind":"REWORK","_number":14,"created":"2026-05-15 14:04:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/14"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-15 14:04:16.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf306083f5f23798bb0522867ad224df3c636599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf306083f5f23798bb0522867ad224df3c636599"}]},"branch":"refs/heads/master"},"23bea61447f109772f6d0afe49c7307648807615":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-05-15 22:05:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/15"}}},"commit":{"parents":[{"commit":"74c619c35751024cd377253e4656b9bde37577c8","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74c619c35751024cd377253e4656b9bde37577c8"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-15 22:04:13.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/23bea61447f109772f6d0afe49c7307648807615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/23bea61447f109772f6d0afe49c7307648807615"}]},"branch":"refs/heads/master"},"336fd9504f419974c5ea4777fff0a0acbfb3a86c":{"kind":"NO_CHANGE","_number":16,"created":"2026-05-15 22:05:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/16"}}},"commit":{"parents":[{"commit":"74c619c35751024cd377253e4656b9bde37577c8","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74c619c35751024cd377253e4656b9bde37577c8"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-15 22:04:36.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/336fd9504f419974c5ea4777fff0a0acbfb3a86c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/336fd9504f419974c5ea4777fff0a0acbfb3a86c"}]},"branch":"refs/heads/master"},"cc5da846cebbcb9f6d8598fd5b7899ad0cfc16d6":{"kind":"REWORK","_number":17,"created":"2026-07-27 00:09:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/895211/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/895211/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/895211/17"}}},"commit":{"parents":[{"commit":"86500cdf2beeefd36cea93ea8a1b8f784d8d387e","subject":"Enable SRBAC for glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86500cdf2beeefd36cea93ea8a1b8f784d8d387e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-16 06:36:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-26 08:40:48.000000000","tz":-420},"subject":"Re-enable IPv6 tests","message":"Re-enable IPv6 tests\n\nLP#1998489 is addressed by replacing Quagga with\nFRR in Manila\u0027s devstack plugin. Also, os-ken\n4.2.0 fixes the broken spawn_after in the native\nthreading hub (LP#2148744), so BGP hold timers\nwork correctly without needing the eventlet hub.\n\nRe-enable IPv6 scenario tests in the LVM, CephFS\nNFS and multinode CephFS NFS jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/895212\nCloses-Bug: #1998489\nChange-Id: If341dda351a9ffe6d133bc2207d5e7d5342b8547\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc5da846cebbcb9f6d8598fd5b7899ad0cfc16d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc5da846cebbcb9f6d8598fd5b7899ad0cfc16d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
