)]}'
{"id":"openstack%2Fmanila-tempest-plugin~897574","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib3c35f15cb1440be4b3eb0f0f6b9612b9ad373b6","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-17 19:57:20.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-17 19:57:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib3c35f15cb1440be4b3eb0f0f6b9612b9ad373b6","subject":"Run stable jobs on 2023.2","status":"ABANDONED","created":"2023-10-06 16:39:55.000000000","updated":"2023-10-17 19:57:20.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f08cd0c838986d02751def273e82d623a7789ef2","_number":897574,"virtual_id_number":897574,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-06 18:28:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 16:41:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-10-06 17:27:34.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-06 18:28:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c414ea31176ef39757061c33f1fd0da16c15f11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-06 16:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a0aa77c99dcc83ef907ecf568bda557405b18bd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-06 16:41:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82404d03877e3fc00d3b69c85f64205963dd5f77","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-06 17:25:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"336bf2a86de19b649eeaddab16877785fb5fd264","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-06 17:27:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/897574/1/check/DellEMC_PowerFlex_Manila_dhss_false/3f151f0/DellEMC_PowerFlex_Manila_dhss_false/552 : SUCCESS in 31m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"55f8cb232e60c20c55c774f0c9b7887249f63bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 18:28:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5cb70e8b007497e8be138b3c1d99653\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c531803a0b3b41e2ae117b65a061d226 : SUCCESS in 3m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c57fb635f23b43bfb2eaa8b22d7b26a0 : SUCCESS in 35m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/723d285647584c89a06dd3809bd1b38b : SUCCESS in 1h 46m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/50881ab47dac491886712488f1cabb96 : SUCCESS in 1h 10m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cb42fd7bdfe54ad696957e1754e3b9d7 : SUCCESS in 1h 15m 01s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ff7f59b8837d4f73a84339f61c370f36 : SUCCESS in 1h 39m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/103e234d533f421b89681b0269454bd4 : FAILURE in 1h 13m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc610a0cd2624877aa42b19fb81fb676 : SUCCESS in 38m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ddbc3d89ef6479ebabb64daf2ce0e86 : SUCCESS in 36m 56s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d452e3c868fb4924a683b5e5213da9ed : SUCCESS in 48m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6762f9d474d54dc5a647368a5a6b42db : SUCCESS in 42m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fac4f65c7df24c1f8e1285f916bce3b2 : FAILURE in 29m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aee8512d538543449d54d9c92117880a : FAILURE in 51m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2777bac1b4db41bdb07a8e2fbb353acd : FAILURE in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f08cd0c838986d02751def273e82d623a7789ef2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-17 19:57:20.000000000","message":"Abandoned\n\nChange not needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4759cf63acf8160c70280c2f4760249b9d242bfa","revisions":{"4759cf63acf8160c70280c2f4760249b9d242bfa":{"kind":"REWORK","_number":1,"created":"2023-10-06 16:39:55.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/74/897574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/897574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/897574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/897574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/897574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/897574/1"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-06 16:39:17.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-06 16:39:17.000000000","tz":-180},"subject":"Run stable jobs on 2023.2","message":"Run stable jobs on 2023.2\n\nAdd the 2023.2 Bobcat branch to the list of branches we should\nrun our tempest stable jobs.\n\nChange-Id: Ib3c35f15cb1440be4b3eb0f0f6b9612b9ad373b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4759cf63acf8160c70280c2f4760249b9d242bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4759cf63acf8160c70280c2f4760249b9d242bfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
