)]}'
{"id":"openstack%2Fmanila-tempest-plugin~898257","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5bdcfc1c975f6e2cbaf0601d344495b697176961","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2020816","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-14 01:31:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5bdcfc1c975f6e2cbaf0601d344495b697176961","subject":"[WIP] Clone snapshot to a share of larger size","status":"NEW","created":"2023-10-13 23:42:39.000000000","updated":"2023-10-19 05:44:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27dff97c7b4ccb00547b07c68609425f92265e63","_number":898257,"virtual_id_number":898257,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-19 01:28:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-14 01:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-14 03:32:17.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-16 18:28:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"40f0ddbb5cf02fd5e7f4cc75212b8ae1a4d9115c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-13 23:42:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c5ba1d7cdc4fbddb159b08bf237d2e95a0ff7f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 01:31:36.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f85deec3aa924e36a773b831439734f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29a5daeb8874a378e8b1d0beb529357 : FAILURE in 4m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee56d51d85d541909c04830bc0014a4f : SUCCESS in 26m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d069d02e469341c0be4d7385431061f0 : FAILURE in 1h 47m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6329da7563f7470388b1ff1395dba289 : SUCCESS in 1h 05m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4c0577945cd444679e5a71aaeaafba7c : FAILURE in 52m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/fb948ed82d5449d19dad561c830e6e85 : SUCCESS in 1h 12m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7736bf02205541b49591bd507570adec : FAILURE in 53m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aa2d7c7efae14a32a5cb82082f90406f : SUCCESS in 38m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f5c7d11e86274f78901eb70a41faed92 : SUCCESS in 40m 56s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/bc1aa304598446c98be78302b70fdc37 : SUCCESS in 45m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/470570afb66d41e4aaa4b3ff187b4b6b : SUCCESS in 41m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6c38b0728d174f528650decd8e4c1c8b : FAILURE in 35m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bb508fc9208645dd8776fb856eb87b9a : FAILURE in 57m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e86e6014da684670ab1f79bbf5131f11 : FAILURE in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33296e7c45173342bcf818474de8652bc29f6f5f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-14 03:32:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/57/898257/1/check/DellEMC_Unity_Manila_dhss_true/b26700f/DellEMC_Unity_Manila_dhss_true/655 : SUCCESS in 3h 49m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/1/check/DellEMC_Unity_Manila_dhss_false/24127ff/DellEMC_Unity_Manila_dhss_false/692 : SUCCESS in 1h 11m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"3cb7ab0ef960678195186b4bcc7537c5c9cb8f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 17:58:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dfe8358306338bc7c6d4af431e7a59778e97355e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-16 18:28:39.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/2/check/DellEMC_PowerFlex_Manila_dhss_false/c68c57a/DellEMC_PowerFlex_Manila_dhss_false/555 : SUCCESS in 29m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"d81157d4a1d06a832e885b927038fb03cabbcf92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:01:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f06cd55d309d48b99964381fc86833bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82d151393354882a090c6c5cc1bce57 : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ffff8208b3144102b8e601c0d074766b : SUCCESS in 37m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1113283b2ac84461b959768818afcfb0 : FAILURE in 2h 55m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/57750acb0271431e9fb5aac2cf36727a : SUCCESS in 1h 27m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3a48dc0ffdba45f3b7fc7a2aebcfe6e5 : FAILURE in 1h 05m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/17872b8a2a324b2aa891bd46046b63e8 : SUCCESS in 1h 37m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9cdf9997f70c4df4a6a5be35a0f52106 : FAILURE in 1h 21m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/71d05617087f41209749e9141077a7c8 : SUCCESS in 37m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec838685b5544dc298f2bc47167b0b30 : SUCCESS in 48m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3474ff808e82473294d6a52381d3a79f : SUCCESS in 49m 45s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3b2f3c790a5c4f92892e2a71f1337884 : SUCCESS in 37m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5f3607d64cb14b12b5d1fe7b5beb2317 : FAILURE in 26m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f7ce6248c6b4bee923107ef79a533a3 : FAILURE in 40m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8e706db3e7634a26b81f9f1b8f8db440 : FAILURE in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b032ddac849a0f0306737efcb16d012b71415e94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 22:14:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"18152f6220579c427ad0670f4cc22998d3dd068f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-16 22:46:13.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/3/check/DellEMC_PowerFlex_Manila_dhss_false/318d50f/DellEMC_PowerFlex_Manila_dhss_false/561 : SUCCESS in 31m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"f8082a87ad7a40dc279680dee14e847192d14496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 00:32:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038b6e2ba6274c12912f2fbebaef9341\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0194b01b148c4884a1c7f9efe923b69a : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c06a3145ec4e4ff3bfaa72484259608c : FAILURE in 36m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ec1709f11e904da29862c2f7a8325c2f : FAILURE in 1h 48m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb0fb16b3ed649e48c17ca7426d21477 : SUCCESS in 1h 14m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/db469f19aa544185af1e0927e98d2fd7 : FAILURE in 59m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/786424fccebd4fa9872a497377059098 : SUCCESS in 1h 28m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdfb78fcab874bf49f30e3510c4a8aea : FAILURE in 1h 07m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d55b723164694c47b09270003d34f541 : SUCCESS in 27m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a0b659575c84fdb95febfbb67db41db : SUCCESS in 38m 38s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d94bb12c1ce14a8c80881b084e03d823 : FAILURE in 1h 09m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9c4daaf8ad8a4ca297408eb15b397a4f : FAILURE in 1h 02m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d85a824ee50741c4a3f50dc52ec6780f : FAILURE in 26m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/94ef81013f834afa98d5b16fe2bab66a : FAILURE in 44m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7259763a9e44a4fba1c503072b1c31f : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3933a154486d278dce7fdca805b6070dd110f9d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-17 10:00:57.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/57/898257/3/check/DellEMC_Unity_Manila_dhss_true/989d8a4/DellEMC_Unity_Manila_dhss_true/665 : SUCCESS in 4h 04m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/3/check/DellEMC_Unity_Manila_dhss_false/bc36574/DellEMC_Unity_Manila_dhss_false/702 : SUCCESS in 1h 03m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"28bc559b8d8331e0ab0252c9bc976dc62f02bd67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-17 22:57:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b34fa8e36d1a2877355ff879d9e0a1f8c36e3530","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-17 23:54:02.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/4/check/DellEMC_PowerFlex_Manila_dhss_false/c5422ef/DellEMC_PowerFlex_Manila_dhss_false/567 : SUCCESS in 29m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b48a0663990881064bda6a9b0dfb4360ce91a06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 00:49:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd59363cbd504a33a98db7d4912aa229\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c0e1d734f14f9bb2d66a8c1925c636 : SUCCESS in 5m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f54ca2994a6b49e99245e9cc585158d5 : FAILURE in 1h 01m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f4a46feefe604d7584da092e5176cd0c : FAILURE in 1h 50m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af81f3d3677a43bea397774e5e98419a : SUCCESS in 1h 17m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2a8cb6ce32cc477c848d248b390ecacd : FAILURE in 1h 06m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0a613ca04b204704a05fb2ef2d141801 : SUCCESS in 1h 32m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d3ff7dc1a15042c1839df4b197f68bfa : FAILURE in 1h 22m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f4c5198a16b240038e5dd79fe4083a91 : SUCCESS in 37m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cbdb45811d8645b88feab3ad3450ada7 : SUCCESS in 30m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/51bb71d1f607420d8ba727526ac01a87 : FAILURE in 1h 26m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d562bb533f76422ca6c74ecd605ee88e : FAILURE in 1h 15m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/746c2473c49d4349aa7e1d3217e5020a : FAILURE in 21m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6ee15eea45484625baa3336198c32fe6 : SUCCESS in 47m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3c45ba3118dd4620bbd6b6eb60885703 : FAILURE in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d125b6d1baf2d7f54dfdc6abb7e0bceebab16c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-18 02:26:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4c790a0dbb999c11321fedb4326db37e0176cd1e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 02:58:10.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/5/check/DellEMC_PowerFlex_Manila_dhss_false/c0e35ed/DellEMC_PowerFlex_Manila_dhss_false/568 : SUCCESS in 31m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a238d69c60c1ccbb9bfd2a4875ceb30b2f908492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 04:45:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4879712c08c441239f6d8e5ed5efda73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e7b22d86ce4513849fb8d2b1055051 : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3a031231d6814055b3ba1053784488ce : FAILURE in 59m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9cf0a3b91c624423a4f9854a54b8430c : FAILURE in 1h 57m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d711ca553f88451f91b855758c229944 : SUCCESS in 1h 24m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/612bc29194884e6488297ae83c56fa8c : FAILURE in 57m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6504fd8a14d74652a3dc57d270a8cdb8 : SUCCESS in 1h 42m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f407cd066bc844c99044f73a35a72084 : SUCCESS in 1h 05m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b95834fc11e844f49e3d6a5a521debee : SUCCESS in 38m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f361c8e09e634136934c223d4ae7e876 : SUCCESS in 45m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e5e60586a3c145fe8da71f73b5832061 : FAILURE in 1h 22m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/24a636e1add64c2b97cbb267510179df : FAILURE in 1h 15m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b8d602f3fd244319abd94f1d1110ab23 : FAILURE in 35m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d54d4b9acf6d4352a6767b9fb5a70a28 : FAILURE in 48m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/eb9aa447a7c145289a58fe5f49082582 : FAILURE in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c37669d525587321c7d539037ea9c60e0fa422e1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-18 05:25:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/57/898257/5/check/DellEMC_Unity_Manila_dhss_true/58d355a/DellEMC_Unity_Manila_dhss_true/671 : SUCCESS in 2h 58m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/5/check/DellEMC_Unity_Manila_dhss_false/b44f769/DellEMC_Unity_Manila_dhss_false/708 : SUCCESS in 1h 19m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"4cbf63cbeb090d8d2b68cd737a97cd312aae66f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-18 22:25:40.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cbb0c66d41da0c5f5478637f224cbb0c0d2f7987","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-18 22:25:58.000000000","message":"Topic bug/2004230 removed","accounts_in_message":[],"_revision_number":6},{"id":"7a414f39e70df6d8d1369ea39c843b7d574b2e8f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-18 22:26:02.000000000","message":"Topic set to bug/2020816","accounts_in_message":[],"_revision_number":6},{"id":"7b1d122dcc059d96b898b913f4bddf13a32e7159","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 23:24:22.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/6/check/DellEMC_PowerFlex_Manila_dhss_false/365eeb4/DellEMC_PowerFlex_Manila_dhss_false/579 : SUCCESS in 31m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"1aaa6407706555041109bb0222f1ea8d317bb8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 01:28:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/498d0b423ad34c228428e6163ecd7cfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a49faa444c74b1abbff225f4c530aee : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/59616af260104e5cb02a5850412930b9 : FAILURE in 1h 33m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e34bfbe39ad24aa28533ca2ef8f8bd32 : FAILURE in 2h 54m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4170bd04d0d04ecaaff8a57af6087de9 : SUCCESS in 1h 24m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0aa2c291b30a4a05948d6ad2f712bc17 : FAILURE in 1h 10m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cb19cc1707ea4ffa9ca8300bb6a42b9c : FAILURE in 1h 04m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/96a4b26cc63f4955a0bbbc31b40e1104 : SUCCESS in 1h 14m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6bd08f4718848288bc70c5a02f6602c : SUCCESS in 36m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec3f87db65e440178e2f34b045aea1fa : SUCCESS in 47m 41s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/46811babbf604a1bbb466d913dd0abc7 : FAILURE in 1h 35m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ea72ee16a3d5411893e988107bfe5169 : FAILURE in 52m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/190913df36d940fa92c0a11bf4bc188e : FAILURE in 21m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a5ee35d07034f2b989badb7cc7f837a : FAILURE in 41m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5b3488ca73cf4662b591b3f9629523e8 : FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"27dff97c7b4ccb00547b07c68609425f92265e63","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-19 05:44:46.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/57/898257/6/check/DellEMC_Unity_Manila_dhss_true/a0ec8be/DellEMC_Unity_Manila_dhss_true/682 : FAILURE in 4h 15m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/57/898257/6/check/DellEMC_Unity_Manila_dhss_false/fd43b4b/DellEMC_Unity_Manila_dhss_false/719 : SUCCESS in 1h 15m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ba1ea92e235dd520811c4f6892fc8f9116032cd2","revisions":{"fd2aee96debaad6a2bf8e4c6521fa0b97bcf3166":{"kind":"REWORK","_number":1,"created":"2023-10-13 23:42:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/1"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:42:26.000000000","tz":-420},"subject":"Clone snapshot to a share of larger size","message":"Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd2aee96debaad6a2bf8e4c6521fa0b97bcf3166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd2aee96debaad6a2bf8e4c6521fa0b97bcf3166"}]},"branch":"refs/heads/master"},"18f587bd82c58d9730be4283f0a6e2b77e824cec":{"kind":"REWORK","_number":2,"created":"2023-10-16 17:58:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/2"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-16 17:58:34.000000000","tz":-420},"subject":"Clone snapshot to a share of larger size","message":"Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/18f587bd82c58d9730be4283f0a6e2b77e824cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/18f587bd82c58d9730be4283f0a6e2b77e824cec"}]},"branch":"refs/heads/master"},"5b7a2db5cb2f053cdeddee1de11393592ad8d1ac":{"kind":"REWORK","_number":3,"created":"2023-10-16 22:14:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/3"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-16 22:13:30.000000000","tz":-420},"subject":"Clone snapshot to a share of larger size","message":"Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nAlso re-enable scenario tests on the LVM driver job\nso we can run this test case.\n\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b7a2db5cb2f053cdeddee1de11393592ad8d1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b7a2db5cb2f053cdeddee1de11393592ad8d1ac"}]},"branch":"refs/heads/master"},"9e6f755af12f375b525750deb27eb2fb5a1be10e":{"kind":"REWORK","_number":4,"created":"2023-10-17 22:57:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/4"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-17 22:56:49.000000000","tz":-420},"subject":"Clone snapshot to a share of larger size","message":"Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nAlso re-enable scenario tests on the LVM driver job\nso we can run this test case.\n\nDepends-On: Ie6599bc3533a002fa118ebe8b9243d677bdb410e\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e6f755af12f375b525750deb27eb2fb5a1be10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e6f755af12f375b525750deb27eb2fb5a1be10e"}]},"branch":"refs/heads/master"},"618e61713c29d2015106056c7695ef3275a88254":{"kind":"REWORK","_number":5,"created":"2023-10-18 02:26:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/5"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-18 02:25:46.000000000","tz":-420},"subject":"Clone snapshot to a share of larger size","message":"Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nAlso re-enable scenario tests on the LVM driver job\nso we can run this test case.\n\nDepends-On: Ie6599bc3533a002fa118ebe8b9243d677bdb410e\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/618e61713c29d2015106056c7695ef3275a88254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/618e61713c29d2015106056c7695ef3275a88254"}]},"branch":"refs/heads/master"},"ba1ea92e235dd520811c4f6892fc8f9116032cd2":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-10-18 22:25:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/57/898257/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/898257/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/898257/6"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-13 23:34:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-18 22:25:40.000000000","tz":0},"subject":"[WIP] Clone snapshot to a share of larger size","message":"[WIP] Clone snapshot to a share of larger size\n\nChanging this scenario test will help us establish\nyet another requirement from our share backend driver\nexpectations:\n\n- Users should be able to create a share of size X\n- They can snapshot the share\n- They can then create a new share from snapshot of\n  size Y (\u003e X)\n- Backend share drivers are expected to ensure that\n  the resulting share is extended past the size of the\n  parent share.\n\nAlso re-enable scenario tests on the LVM driver job\nso we can run this test case.\n\nDepends-On: I7842681c44aa5a6176eaa0124fa354d4a0587b27\nDepends-On: Ie6599bc3533a002fa118ebe8b9243d677bdb410e\nChange-Id: I5bdcfc1c975f6e2cbaf0601d344495b697176961\nRelated-Bug: #2020816\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba1ea92e235dd520811c4f6892fc8f9116032cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba1ea92e235dd520811c4f6892fc8f9116032cd2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
