)]}'
{"id":"openstack%2Fmanila-tempest-plugin~898712","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I754e8f802c13011762be2f27fb45df849f95a25e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"check_srbac_gate","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-03-14 07:39:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I754e8f802c13011762be2f27fb45df849f95a25e","subject":"Check SRBAC gate","status":"ABANDONED","created":"2023-10-18 07:06:33.000000000","updated":"2024-03-14 07:39:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e258a5d9c4a0319a0aac164f70843af5d44db2a","_number":898712,"virtual_id_number":898712,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-18 09:30:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-18 07:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-18 07:43:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-18 10:05:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-18 12:14:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-18 14:26:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"762455bc3aabf0e51b794fc72d0c1fa344fbd16d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 07:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58328bbce34623e4088985e75fa9be3b3c8d6cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:06:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n  #      - check-requirements\n  #      - tempest-plugin-jobs\n  #      - manila-tempest-plugin-jobs-using-service-image\n      queue: manila\n      check:\n        jobs:\n  #        - manila-tempest-plugin-dummy-no-dhss\n  #        - manila-tempest-plugin-dummy-dhss\n  #        - manila-tempest-plugin-lvm-yoga\n  #        - manila-tempest-plugin-lvm-xena\n          - manila-tempest-plugin-dummy-no-dhss-rbac\n  #        - manila-tempest-plugin-container:\n  #            voting: false\n  #        - manila-tempest-plugin-glusterfs-nfs:\n  #            voting: false\n      gate:\n        jobs:\n  #        - manila-tempest-plugin-dummy-no-dhss\n  #        - manila-tempest-plugin-dummy-dhss\n      experimental:\n        jobs:\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n\n  in \"openstack/manila-tempest-plugin/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"cc9ad2fa678c73ae8e6ef9a434dd1f6ea9269f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 07:09:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ba67b548a85c80da6ec8a5426474e4b24c6aa47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:09:38.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila on branch master.  The error was:\n\n  The project template \"manila-tempest-plugin-jobs-using-service-image\"\n  was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n        - manila-tempest-plugin-jobs-using-service-image\n      queue: manila\n      check:\n        jobs:\n          - manila-tox-genconfig\n          - openstack-tox-pylint:\n              voting: false\n              timeout: 5400\n          - openstack-tox-cover:\n              voting: false\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-container:\n              voting: false\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs:\n              voting: false\n          - manila-tempest-plugin-lvm-fips:\n              voting: false\n          - manila-grenade:\n              voting: false\n          # NOTE: this job needs to be voting in SLURP release and option to\n          # voting in non SLURP. But keeping it voting in non SLURP also will\n          # help us to know any upgrade break in advance.\n          - manila-grenade-skip-level-always:\n              voting: false\n          - manila-rally-no-ss:\n              voting: false\n          - manila-rally-ss:\n              voting: false\n      gate:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n\n  in \"openstack/manila/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"c5b2643cc173fc1ebef4db04d7b3da1027258544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 07:10:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d5025ce4d57825aa7b932410c2d28ef8f49bb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:11:12.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila on branch master.  The error was:\n\n  The project template \"manila-tempest-plugin-jobs-using-service-image\"\n  was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n        - manila-tempest-plugin-jobs-using-service-image\n      queue: manila\n      check:\n        jobs:\n          - manila-tox-genconfig\n          - openstack-tox-pylint:\n              voting: false\n              timeout: 5400\n          - openstack-tox-cover:\n              voting: false\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-container:\n              voting: false\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs:\n              voting: false\n          - manila-tempest-plugin-lvm-fips:\n              voting: false\n          - manila-grenade:\n              voting: false\n          # NOTE: this job needs to be voting in SLURP release and option to\n          # voting in non SLURP. But keeping it voting in non SLURP also will\n          # help us to know any upgrade break in advance.\n          - manila-grenade-skip-level-always:\n              voting: false\n          - manila-rally-no-ss:\n              voting: false\n          - manila-rally-ss:\n              voting: false\n      gate:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n\n  in \"openstack/manila/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":3},{"id":"e64654f0c50a0fbc1f17a8c7161d2b527415c4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 07:15:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ec67542c8f9b194e71f8bf093bc35d4ccbe484a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:15:29.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila on branch stable/wallaby.  The error was:\n\n  The project template \"manila-tempest-plugin-jobs-using-service-image-\n  wallaby\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-python3-wallaby-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n        - manila-tempest-plugin-jobs-using-service-image-wallaby\n      queue: manila\n      check:\n        jobs:\n          - manila-tox-genconfig\n          - openstack-tox-pylint:\n              voting: false\n              timeout: 5400\n          - openstack-tox-cover:\n              voting: false\n          - manila-tempest-plugin-dummy-no-dhss-wallaby\n          - manila-tempest-plugin-dummy-dhss-wallaby\n          - manila-tempest-plugin-container-wallaby:\n              voting: false\n          - manila-tempest-plugin-glusterfs-native-wallaby:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs-wallaby:\n              voting: false\n          - manila-tempest-plugin-lvm-fips-wallaby:\n              voting: false\n          - manila-grenade:\n              voting: false\n          - manila-rally-no-ss:\n              voting: false\n          - manila-rally-ss:\n              voting: false\n      gate:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss-wallaby\n          - manila-tempest-plugin-dummy-dhss-wallaby\n      experimental:\n        jobs:\n          - tripleo-ci-centos-8-scenario004-standalone\n\n  in \"openstack/manila/zuul.d/project.yaml@stable/wallaby\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"3dfff52056b0769e824cc4379ab92143cd8384bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 07:16:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"67bd6c877acf87576b63fff1418fdce8a7f90508","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 07:43:21.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/898712/5/check/DellEMC_PowerFlex_Manila_dhss_false/89b27b3/DellEMC_PowerFlex_Manila_dhss_false/573 : SUCCESS in 26m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a26c37a5df5c9baf6d6e9cbff41ae973d9aa37d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:58:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/602dc2c02d4c468bb1848eefd4069a50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06042f59a3854b25a8a200af64e93ddc : SUCCESS in 2m 52s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/11b1f2c46cbe45518a7ea05725d06fcc : FAILURE in 34m 28s","accounts_in_message":[],"_revision_number":5},{"id":"34594c9fab95f85ea4de4ff0e5ef4c4cc344cbcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 08:02:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"047f4edcd3db06e6d6947e33c31f9e06bd55e499","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 08:36:16.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/898712/6/check/DellEMC_PowerFlex_Manila_dhss_false/8e54551/DellEMC_PowerFlex_Manila_dhss_false/574 : SUCCESS in 34m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"31b30622b6eccb886e87a12d6174ace6c956f02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:41:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bebc8d82007e47f1a58ae19e75046966\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b762b84ce34a82a375bc261768e273 : SUCCESS in 2m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f8134c1763fa4b259d41773a59009a2f : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":6},{"id":"0934e341a209bccdf7604f8d86ee59083b8826d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-10-18 08:49:30.000000000","message":"Uploaded patch set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9b64e95b19c23443dcbe7add4bb71e825eaa3b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 09:30:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42cdc4377e31449b98b7f23bd4f82f2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6145bda9094e32a7cd3334a107c51f : SUCCESS in 2m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/79db630bb19744d4bb41f41b99b1d153 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":7},{"id":"59288f657ae5e490d1de0caefb0394921cb7261f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-18 09:35:05.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/898712/7/check/DellEMC_PowerFlex_Manila_dhss_false/b1d22de/DellEMC_PowerFlex_Manila_dhss_false/576 : SUCCESS in 29m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"f4edcac51fc615109389152f9d169ddf9c93a6b1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-18 10:05:30.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/898712/7/check/DellEMC_PowerStore_Manila_dhss_false/1761027/DellEMC_PowerStore_Manila_dhss_false/248 : SUCCESS in 34m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"b37d560ea32dad754d8ab0e2a8e597f2dcbfec59","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-18 12:14:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/898712/7/upstream-check/manila-tempest-plugin-ontap-dhss/a673393/ : RETRY_LIMIT in 9m 32s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/898712/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/48b91f3/ : FAILURE in 2h 42m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"5122ae8af609e3d3de300b6c220e60d9ac54dc95","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-18 14:26:21.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/12/898712/7/check/DellEMC_Unity_Manila_dhss_true/2e70b54/DellEMC_Unity_Manila_dhss_true/679 : SUCCESS in 2h 45m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/898712/7/check/DellEMC_Unity_Manila_dhss_false/028158f/DellEMC_Unity_Manila_dhss_false/716 : SUCCESS in 1h 00m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"0e258a5d9c4a0319a0aac164f70843af5d44db2a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-03-14 07:39:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"21a5df489eb62670578e08a015d956a415b933c2","revisions":{"700b85fcdbfc7e921b3de180a1b3f764390d14ef":{"kind":"REWORK","_number":1,"created":"2023-10-18 07:06:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/1"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/700b85fcdbfc7e921b3de180a1b3f764390d14ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/700b85fcdbfc7e921b3de180a1b3f764390d14ef"}]},"branch":"refs/heads/master"},"cc40c64ab4a9787fe46a686d5836a8d6cb1ddc78":{"kind":"REWORK","_number":2,"created":"2023-10-18 07:09:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/2"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:08:58.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc40c64ab4a9787fe46a686d5836a8d6cb1ddc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc40c64ab4a9787fe46a686d5836a8d6cb1ddc78"}]},"branch":"refs/heads/master"},"5bdcdfe113510628cfa6370c6fb35ea556b17c16":{"kind":"REWORK","_number":3,"created":"2023-10-18 07:10:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/3"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:10:42.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bdcdfe113510628cfa6370c6fb35ea556b17c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5bdcdfe113510628cfa6370c6fb35ea556b17c16"}]},"branch":"refs/heads/master"},"9ddd0109635805d0d67c3a9b779e0808ca5ece37":{"kind":"REWORK","_number":4,"created":"2023-10-18 07:15:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/4"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:14:57.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ddd0109635805d0d67c3a9b779e0808ca5ece37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ddd0109635805d0d67c3a9b779e0808ca5ece37"}]},"branch":"refs/heads/master"},"6bdd944af0926fb76ac28b59680e10d1d17e895f":{"kind":"REWORK","_number":5,"created":"2023-10-18 07:16:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/5"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:16:23.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bdd944af0926fb76ac28b59680e10d1d17e895f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bdd944af0926fb76ac28b59680e10d1d17e895f"}]},"branch":"refs/heads/master"},"d194e735851716f8648eacf63b8318c1a762084e":{"kind":"REWORK","_number":6,"created":"2023-10-18 08:02:01.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/6"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 08:01:51.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d194e735851716f8648eacf63b8318c1a762084e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d194e735851716f8648eacf63b8318c1a762084e"}]},"branch":"refs/heads/master"},"21a5df489eb62670578e08a015d956a415b933c2":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-10-18 08:49:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/12/898712/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/898712/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/898712/7"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 07:06:22.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-10-18 08:49:07.000000000","tz":180},"subject":"Check SRBAC gate","message":"Check SRBAC gate\n\nDepends-on: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/872260\n\nChange-Id: I754e8f802c13011762be2f27fb45df849f95a25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21a5df489eb62670578e08a015d956a415b933c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21a5df489eb62670578e08a015d956a415b933c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
