)]}'
{"id":"openstack%2Fmanila-tempest-plugin~908701","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib0e984a4acdc2f82f5e3a75e34c7ec922674b0a5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-jobs","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-02-13 12:39:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-02-13 17:15:15.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-02-13 18:37:47.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-02-13 14:36:05.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"Ib0e984a4acdc2f82f5e3a75e34c7ec922674b0a5","subject":"[CI] Stable branch CI adjustments","status":"MERGED","created":"2024-02-09 22:37:05.000000000","updated":"2024-02-13 18:37:47.000000000","submitted":"2024-02-13 18:37:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908701-stable-jobs","meta_rev_id":"5903ff868360020be4b5b8d0a3d3b61fd9be348a","_number":908701,"virtual_id_number":908701,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 18:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-02-13 12:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-02-13 14:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2024-02-13 12:13:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-02-13 17:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-09 23:11:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-10 00:14:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-10 07:23:03.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-12 22:39:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-02-12 22:39:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-02-13 07:19:15.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-13 12:13:06.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2024-02-13 12:39:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"058f9108726ac722c68d8dad18464c6204c301ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-09 22:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b24a6890944e925eb01e9019da33391204155e3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-09 23:11:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/1/check/DellEMC_PowerFlex_Manila_dhss_false/e981575/DellEMC_PowerFlex_Manila_dhss_false/790 : SUCCESS in 31m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"58cf88af4164b7d230098cac08626ba0b16009c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 00:14:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f2ee160178348fa8c6b1f46d32a7c16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0d0f4067b849c5960d4cadfc9d1643 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f133db9e8f4e426f89375872ca33eac8 : SUCCESS in 52m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/39c5779475734f63bd682d3bab4c05df : SUCCESS in 1h 34m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/89de939428b74d29b514adda33acfd55 : SUCCESS in 1h 14m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/63939ef541e147b2a310903771036a8b : SUCCESS in 49m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e0570cf436c44d35a05567040b368e90 : SUCCESS in 1h 28m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a638569300434f019d03fbf873227d64 : FAILURE in 44m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2050a37a7bba49c6b5867cbaa4389437 : SUCCESS in 26m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f4dec382ce784077bc633d85ba71b477 : SUCCESS in 32m 24s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/dfadf45019c94101bfac1b9a40ee3e8d : FAILURE in 25m 09s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/79e8bd604206422b947f01101185f5e1 : FAILURE in 33m 56s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/26c6e2683aaa40758f3d938e7bf60a86 : SUCCESS in 1h 20m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/27e0753fa93244ee990563a0138969af : SUCCESS in 31m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9383b93f30024a78a6850b73ee9cb8f9 : FAILURE in 57m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/660e1bf6e40540868dc342a2a8624c5d : FAILURE in 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c4cefdff2e5a2303d360487353ca880fe920f39","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-10 07:23:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/01/908701/1/check/DellEMC_Unity_Manila_dhss_true/560a01a/DellEMC_Unity_Manila_dhss_true/796 : SUCCESS in 3h 42m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/1/check/DellEMC_Unity_Manila_dhss_false/e72f24a/DellEMC_Unity_Manila_dhss_false/849 : SUCCESS in 46m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"87351b5d4179e31c47aea487d93a17b5684a0754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-12 22:39:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6383c7e00373d20bada39ef03111bad2f575093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 22:40:10.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-stable-jobs.yaml@master\", line 85, column 5\n  did not find expected key\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-stable-jobs.yaml@master\", line 91, column 7","accounts_in_message":[],"_revision_number":2},{"id":"3ba9b13bcf8b1b1d92e5e04fc04825935988b763","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-12 23:11:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/2/check/DellEMC_PowerFlex_Manila_dhss_false/9943483/DellEMC_PowerFlex_Manila_dhss_false/797 : SUCCESS in 31m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c010fafb0253905314773e584027bd0d4737b440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-13 01:09:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e136826f91131d9b85847f6f83b0e5a5fbf2447","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-13 01:10:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84b69e28b5dd0c96c390d703d61ba1dd9eaf2510","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-13 02:20:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/3/check/DellEMC_PowerFlex_Manila_dhss_false/045c6c9/DellEMC_PowerFlex_Manila_dhss_false/800 : SUCCESS in 29m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"c03c6662d15aa7ee8171045beb53c941d5ea58e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 04:12:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14cf324dd9db42d9a9adfe7582362bab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b07fd4be80482bad5be4c07801a419 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/df5273d40e494bba8730788c4057990a : FAILURE in 1h 42m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e4217184bafb45a7a21ed846f4c11fc5 : SUCCESS in 2h 11m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4e37dbcaf17b4a66bedb49e9a0b2a2f1 : FAILURE in 2h 49m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/37e473174b154fd89f4fc0589301bc09 : SUCCESS in 1h 22m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/015e88d207ac4bcdbe7b7b9cebfc5ea9 : FAILURE in 1h 28m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/143498331f6840a19420f72e1464b628 : SUCCESS in 1h 04m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/33cb9f592e55457095e3dad614b7eb96 : SUCCESS in 35m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cebf200200464b13a1e20ab7e2ae9d69 : SUCCESS in 32m 30s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/8757c8e5a681497aaa3fa9cd4a8763f5 : SUCCESS in 1h 09m 06s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/c0e20f71cde3409e90a4292731e55f1e : SUCCESS in 58m 04s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/447b394dd68f49c28dd093ea3d711d05 : SUCCESS in 1h 10m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2f9902e379d24f3a9ecdff4e01ade9ff : SUCCESS in 29m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6498a689b41f471c94c02b3a8338eb96 : FAILURE in 39m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/56705af9961f40228e92744a5f57623f : FAILURE in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7f41bd73d272f9eb335c99601c9429c7dcdf1e9e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-13 05:22:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffd1eca4baf3f7d3fd811540623710d89982088b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 07:06:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8814b5a1404abba7615fbf5e433655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb470e088bb49d8b998a76be4468598 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49080bd472b0457096b0330d5539e9e3 : SUCCESS in 52m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9673705a1d0f48a58710a86fc67d5abe : SUCCESS in 1h 42m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8024140548634ee29aae300aeb0da4ae : SUCCESS in 56m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/caa0c36775fc42288e2dff16a9ba1755 : SUCCESS in 1h 00m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/264b0c92013e42bca454459d103ea2bb : FAILURE in 1h 19m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/08ef84d9e59549a5bf2c4696befd453d : SUCCESS in 56m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ae693d4b43a42db9f61d37506c1fca2 : SUCCESS in 22m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/615d62fbc0c24da7a2a1fc7abe47ce32 : SUCCESS in 32m 33s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/2f85b146222a4fb5bc988759439ee920 : SUCCESS in 1h 03m 48s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/11a3d451100047a2b982c546a9e48046 : SUCCESS in 54m 26s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/678cd058a9f5437face3d67a6c64181f : SUCCESS in 59m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ccf5a31d5ece48499debc629744f50fe : SUCCESS in 24m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/14e7efd926c44d5290005fb9ac80dfc7 : SUCCESS in 54m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/409f9fae2cf5408695d4e65a762e8ebb : FAILURE in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7276fff374f86e9656b0ef67b01a641dc41151c6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-13 07:19:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/908701/3/upstream-check/manila-tempest-plugin-ontap-dhss/4d0b381/ : RETRY_LIMIT in 9m 12s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/908701/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/57984d7/ : RETRY_LIMIT in 10m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"df3808582fd8e475c3adabcd308264b9b4d7dd0e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-13 07:39:46.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/3/check/DellEMC_PowerFlex_Manila_dhss_false/90ae0ec/DellEMC_PowerFlex_Manila_dhss_false/805 : SUCCESS in 32m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"db055ad9983e5f4aa0fc15052d9c96e06c1e3f56","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-13 07:47:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/01/908701/3/check/DellEMC_Unity_Manila_dhss_true/5bd5503/DellEMC_Unity_Manila_dhss_true/805 : SUCCESS in 2h 01m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/01/908701/3/check/DellEMC_Unity_Manila_dhss_false/7ccf74e/DellEMC_Unity_Manila_dhss_false/858 : SUCCESS in 51m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2b5c9194be75365de0ae720ef89c7776f8eef2e2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2024-02-13 12:13:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f196fcfc6904c28b5bd4da8595d204a903486e94","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2024-02-13 12:13:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04774c2228c6e6e7cf11c7b49d0b4b36bef66a3d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-02-13 12:39:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53abdee2ac70812794ec66d43e43099b93a9c778","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-02-13 14:36:05.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"858a884e72114b04ea4417dd8a3a6dce01c45ca0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-13 17:15:15.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"22b947ae2e7998fc6c2866dec7b089af5b57c8b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 17:15:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dc05d1dc3b27e72545bd798ef9fb9bb1e2c86580","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:37:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cfda45803644140b5dc37fa14c035c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a459049c39342119ef11c20a18d4a08 : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/24c8a89522374030a6c5ffe2356cba26 : SUCCESS in 1h 16m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ac4a7277021b49c78ee581f4f663ad5e : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d518ed6cb1374afa8d932e656382c7a3 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":3},{"id":"5903ff868360020be4b5b8d0a3d3b61fd9be348a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:37:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d6a517351d10a61aafc97bc70a086e552b1a3d64","revisions":{"79bae58c1d70ccd8f1ca56fdc1da119de444b5a9":{"kind":"REWORK","_number":1,"created":"2024-02-09 22:37:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/01/908701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/01/908701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/1"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-09 22:36:28.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-09 22:36:28.000000000","tz":-480},"subject":"[CI] Stable branch CI adjustments","message":"[CI] Stable branch CI adjustments\n\nXena and Yoga are past the stable maintenance\ncycles. We\u0027re soon going to tag these release\nbranches as \"unmaintained\" branches [1]. We don\u0027t\nexpect the trunk version of manila-tempest-plugin\nto work against these older branches any more.\nInstead, we have tagged a \"xena-last\" and will\nsoon [2] tag a \"yoga-last\" branch that will be more\nappropriate.\n\nGoing forward we must be testing the \"maintained\"\nstable branches, which are zed, antelope, bobcat\nas of this commit.\n\nChange-Id: Ib0e984a4acdc2f82f5e3a75e34c7ec922674b0a5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79bae58c1d70ccd8f1ca56fdc1da119de444b5a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79bae58c1d70ccd8f1ca56fdc1da119de444b5a9"}]},"branch":"refs/heads/master"},"3099df1cb58fb09baa2a0c8beca57a517f4efbd4":{"kind":"REWORK","_number":2,"created":"2024-02-12 22:39:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/01/908701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/01/908701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/2"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-09 22:36:28.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-12 22:39:10.000000000","tz":-480},"subject":"[CI] Stable branch CI adjustments","message":"[CI] Stable branch CI adjustments\n\nXena and Yoga are past the stable maintenance\ncycles. We\u0027re soon going to tag these release\nbranches as \"unmaintained\" branches [1]. We don\u0027t\nexpect the trunk version of manila-tempest-plugin\nto work against these older branches any more.\nInstead, we have tagged a \"xena-last\" and will\nsoon [2] tag a \"yoga-last\" branch that will be more\nappropriate.\n\nGoing forward we must be testing the \"maintained\"\nstable branches, which are zed, antelope, bobcat\nas of this commit.\n\nChange-Id: Ib0e984a4acdc2f82f5e3a75e34c7ec922674b0a5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3099df1cb58fb09baa2a0c8beca57a517f4efbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3099df1cb58fb09baa2a0c8beca57a517f4efbd4"}]},"branch":"refs/heads/master"},"d6a517351d10a61aafc97bc70a086e552b1a3d64":{"kind":"REWORK","_number":3,"created":"2024-02-13 01:09:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/01/908701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/01/908701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/01/908701/3"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-09 22:36:28.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-13 01:08:38.000000000","tz":-480},"subject":"[CI] Stable branch CI adjustments","message":"[CI] Stable branch CI adjustments\n\nXena and Yoga are past the stable maintenance\ncycles. We\u0027re soon going to tag these release\nbranches as \"unmaintained\" branches [1]. We don\u0027t\nexpect the trunk version of manila-tempest-plugin\nto work against these older branches any more.\nInstead, we have tagged a \"xena-last\" and will\nsoon [2] tag a \"yoga-last\" branch that will be more\nappropriate.\n\nGoing forward we must be testing the \"maintained\"\nstable branches, which are zed, antelope, bobcat\nas of this commit.\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#unmaintained\n[2] https://review.opendev.org/c/openstack/releases/+/908381\nChange-Id: Ib0e984a4acdc2f82f5e3a75e34c7ec922674b0a5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a517351d10a61aafc97bc70a086e552b1a3d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6a517351d10a61aafc97bc70a086e552b1a3d64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
