)]}'
{"id":"openstack%2Fmanila-tempest-plugin~912466","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iaaf3c3914089ad0d27b61a524f14353cae35c4d5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-backup","attention_set":{},"removed_from_attention_set":{"24993":{"account":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"last_update":"2024-04-15 18:24:08.000000000","reason":"Change was submitted"},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2024-04-15 18:24:08.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-04-15 18:24:08.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-04-15 18:24:08.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-03-11 15:21:27.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}}},"hashtags":[],"change_id":"Iaaf3c3914089ad0d27b61a524f14353cae35c4d5","subject":"Bump test microversion and fix backup test","status":"MERGED","created":"2024-03-11 15:17:35.000000000","updated":"2024-04-15 18:24:08.000000000","submitted":"2024-04-15 18:24:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"912466-bp/share-backup","meta_rev_id":"9d4797a443f4c244a45b3e6cd21082c7794716e5","_number":912466,"virtual_id_number":912466,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 18:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-04-15 09:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-04-15 17:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-04-15 17:18:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-11 15:19:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"state":"REVIEWER"},{"updated":"2024-03-11 15:19:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-03-11 15:19:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-03-11 15:19:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2024-03-11 15:50:24.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-11 17:14:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-11 17:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-12 01:30:38.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-15 09:57:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-04-15 17:19:07.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"60bf178586d0b374c0d32779d7fe55ae6b4a9857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-03-11 15:17:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e61fa2506f3bc17c61da57ded0a48f21d95e8bce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-03-11 15:20:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"77eabdf10f159a7623e38e2b549ea6ad3eefe2b7","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-03-11 15:21:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee335007ceffcb674d5f75f3306314c5ea6a278f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-11 15:50:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/66/912466/2/check/DellEMC_PowerFlex_Manila_dhss_false/4b8b8f5/DellEMC_PowerFlex_Manila_dhss_false/960 : SUCCESS in 29m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8d096d8f41f765106735f59656627fcd4bdd25ff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-11 17:14:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/66/912466/2/check/DellEMC_Unity_Manila_dhss_true/9b6d281/DellEMC_Unity_Manila_dhss_true/921 : FAILURE in 1h 53m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/66/912466/2/check/DellEMC_Unity_Manila_dhss_false/37d0de6/DellEMC_Unity_Manila_dhss_false/982 : SUCCESS in 54m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4261b210637865e75783ab02850c71924bfab7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 17:34:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85aca2db581542fcbf75454ead45949a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f892e8799e824c19aeca5a11656f6e36 : SUCCESS in 3m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7446c9bbf6a4485eb362dcb27927640c : SUCCESS in 1h 08m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/75ebc64a590a4495a5edbdd59228f4ff : SUCCESS in 2h 12m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aaeacf0273514079953e4857b152e91a : SUCCESS in 1h 19m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/49d469fe48d44a52947d9292c3e1e04e : SUCCESS in 56m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/14ca15128b344f2e86a0025c2e007de5 : FAILURE in 1h 31m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/97451bbf13b14ad8b62af351eb8e4754 : SUCCESS in 57m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2df28087d27b4d16a59f2b1bff9f09bc : FAILURE in 27m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f35dcaf3d23044ad8f8160d5ad554880 : SUCCESS in 41m 44s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/a9f21fe568bf4773af8130968429bf8f : SUCCESS in 55m 27s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/e875181bc59743ef89d4341af42d13d2 : SUCCESS in 59m 25s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/0820717c3f074715a18d6f7b46b409fd : SUCCESS in 1h 16m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c8b92258cc6c4c6c95758d07af75e70f : SUCCESS in 34m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8a9abcdd9ab24cad973c2735b53d6fc7 : SUCCESS in 50m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/80e20b966ca64790b232b7958e4041a7 : FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f69c83a180e83cbb7f7da88117cde809e756c6e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-11 18:47:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62c23c6c9a3f32de54794a6ada43b969eaa43e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-11 23:38:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cead2821a91662c5a3e8812b85bfd930dc4aee43","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-12 00:01:36.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8d50b208269f79297b14bb901a3ddaaca3cf956e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 01:20:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914889740e034cdd8fbf99b56eb221c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7069f5a710424a23a5dfee9b3a27dde7 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2af7c9846d084df5a82de385ad1f53a2 : SUCCESS in 46m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/85882c21c3894eceac5ce6983d83bedc : SUCCESS in 1h 35m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/88476ec06d3248acae2b9aa20f72290e : SUCCESS in 48m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ecd2b8a388e544f3b0043fd8ffeb0b2e : SUCCESS in 1h 04m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/415ae8d46e2a4d86acf62cfe7f4aa02e : SUCCESS in 1h 01m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6d8009375ff4696b1ae69a285e6bf27 : SUCCESS in 1h 29m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9325ac17b8e4489aa02cd02fb44982fd : SUCCESS in 30m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/42d1dc0f5e2d454a8bc3f3c10124ba6b : SUCCESS in 40m 41s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/535fd8b77de44e909c34e4dc74916d47 : SUCCESS in 57m 34s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/ff6b74df33414e4d9496ab5e20b55e48 : SUCCESS in 56m 43s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/74bf029290114590945f1ffc0096b525 : SUCCESS in 1h 03m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c103178105834374a58f2aefbdbcbc96 : SUCCESS in 36m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f49e156ae7224380b2fca17304ce6d73 : FAILURE in 32m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6b9fc9c8e98843718b16e0ae377120b7 : FAILURE in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63f0c9d6ee696816d0823207b059579ae71d5180","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-12 01:30:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/66/912466/3/upstream-check/manila-tempest-plugin-ontap-dhss/0e72a3a/ : RETRY_LIMIT in 3m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/66/912466/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/98d130d/ : RETRY_LIMIT in 3m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4e79c66121a53008e1fc1f5b8e5500afe8b93a51","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-12 11:54:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/66/912466/3/check/DellEMC_Unity_Manila_dhss_true/bd45bd1/DellEMC_Unity_Manila_dhss_true/1070 : SUCCESS in 2h 42m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/66/912466/3/check/DellEMC_Unity_Manila_dhss_false/5c78a98/DellEMC_Unity_Manila_dhss_false/1160 : SUCCESS in 1h 04m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"192751da798650734cadb3c3257c4757f69f3f7b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-04-15 09:57:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d7ae745a9ecac306322eda073efa0df79c6a6d2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-04-15 17:18:20.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a58eee149da97653ebc73d4866d9e15fa8e130a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:18:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ba8b39b827bc58e9b7f94b576c9ccd555f2fc0c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-04-15 17:19:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fd4ea34e382ec6c7ac6416872ba9c5ae05322171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 18:24:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd67f3c89c9e4f4e91558e4fba25520c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b42e01e2e54d46d48f77ff0512ad7d3d : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/524ea687e9904040921a0847046598c6 : SUCCESS in 1h 04m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2e9f866c37e47518f6fce17bb27b728 : SUCCESS in 27m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/09427d0614414ded8983809c5c516e15 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":3},{"id":"9d4797a443f4c244a45b3e6cd21082c7794716e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 18:24:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"077fd3d19864fd945554ff7a019658d618f7fb27","revisions":{"ae57f0a58b9ae04ac43b8fc0bd92cebe490af550":{"kind":"REWORK","_number":1,"created":"2024-03-11 15:17:35.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/66/912466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/912466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/1"}}},"commit":{"parents":[{"commit":"cbab9fe0c72337704e1a364ce1db1d78a3c13826","subject":"Fix backup build erro exception class name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cbab9fe0c72337704e1a364ce1db1d78a3c13826"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-03-11 15:13:00.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-03-11 15:15:26.000000000","tz":330},"subject":"Fixed backup test after adding \"backup_type\" field.","message":"Fixed backup test after adding \"backup_type\" field.\n\nWe added new field \"backup_type\" in share-backups API due to that one test\nis failing which validate the backup response fields. I added \"backup_type\"\nin expected key list to fix the issue.\n\nChange-Id: Iaaf3c3914089ad0d27b61a524f14353cae35c4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae57f0a58b9ae04ac43b8fc0bd92cebe490af550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae57f0a58b9ae04ac43b8fc0bd92cebe490af550"}]},"branch":"refs/heads/master"},"d101f57614aa85d0b3b94c604c37ad41ac377212":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-11 15:20:45.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/66/912466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/912466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/2"}}},"commit":{"parents":[{"commit":"cbab9fe0c72337704e1a364ce1db1d78a3c13826","subject":"Fix backup build erro exception class name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cbab9fe0c72337704e1a364ce1db1d78a3c13826"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-03-11 15:13:00.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-03-11 15:19:27.000000000","tz":330},"subject":"Fixed backup test after adding \"backup_type\" field.","message":"Fixed backup test after adding \"backup_type\" field.\n\nWe added new field \"backup_type\" in share-backups API due to that one\ntest is failing which validate the backup response fields. I added\n\"backup_type\" in expected key list to fix the issue.\n\nChange-Id: Iaaf3c3914089ad0d27b61a524f14353cae35c4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d101f57614aa85d0b3b94c604c37ad41ac377212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d101f57614aa85d0b3b94c604c37ad41ac377212"}]},"branch":"refs/heads/master"},"077fd3d19864fd945554ff7a019658d618f7fb27":{"kind":"REWORK","_number":3,"created":"2024-04-11 23:38:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/912466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/912466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/912466/3"}}},"commit":{"parents":[{"commit":"1687dd6ef680d13205c9cfb99396752f864987c6","subject":"Merge \"Add access rules restriction tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1687dd6ef680d13205c9cfb99396752f864987c6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-03-11 15:13:00.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-04-11 23:38:06.000000000","tz":-420},"subject":"Bump test microversion and fix backup test","message":"Bump test microversion and fix backup test\n\nWe added new field \"backup_type\" in share-backups API\nin 2.85.\n\nChange-Id: Iaaf3c3914089ad0d27b61a524f14353cae35c4d5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/077fd3d19864fd945554ff7a019658d618f7fb27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/077fd3d19864fd945554ff7a019658d618f7fb27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
