)]}'
{"id":"openstack%2Fmanila-tempest-plugin~914823","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I34266790f86162dc04bd055c6156bf47b577a7e6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ceph-nfs","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-04-10 20:26:44.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-04-10 20:26:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I34266790f86162dc04bd055c6156bf47b577a7e6","subject":"Change nodeset for cephfs-nfs job","status":"ABANDONED","created":"2024-04-01 17:58:04.000000000","updated":"2024-04-10 20:26:44.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4e795d6c1a393ee976e5dfd4d5530b94166be971","_number":914823,"virtual_id_number":914823,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-03 19:44:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-01 18:53:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-01 19:09:08.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-04-01 19:40:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-01 21:50:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-02 01:21:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"10cdaf79291d5d15738a24866867b88c048eb3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-01 17:58:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"697f2bbff2009e89df6248f521e913adacb45e84","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 18:53:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/1/check/DellEMC_PowerFlex_Manila_dhss_false/e906bb8/DellEMC_PowerFlex_Manila_dhss_false/736 : SUCCESS in 31m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"80c2043b655fd08637f831f6d0886c0a7f45874b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-01 19:09:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2f6308a778a4f56a2734f70e10f405ee4af24b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 19:40:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b338d14bdd34ce98c7f989a404e6a40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/855bcea717c54bd6b2f0d37294cb1a4a : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6746546b25d74466a125e0160f7a0408 : FAILURE in 1h 16m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/22c56b0fc32d4158a9899a6a5d1eeebc : SUCCESS in 1h 25m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3e74f14231324fceb33b871eadb6e505 : SUCCESS in 56m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4cb2c0f3087f4c048ce4da7c14bdb1d4 : SUCCESS in 47m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f6825ed0404542928da82433328d206e : SUCCESS in 1h 29m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d5392d0546a8417190d42a29ffa08775 : SUCCESS in 1h 33m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b27a22229a34376852304dd6f3b2ab0 : SUCCESS in 26m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e22866c92ab343f4961172af5c924ce9 : SUCCESS in 31m 42s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/18185dda4cd94b93b345f0a0b07f768d : SUCCESS in 57m 30s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/93af4cad18b744338d646a88f901a09c : SUCCESS in 1h 27m 23s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/ab27beab81ad4fa4abaea0be55d2ca0b : SUCCESS in 1h 22m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/962f407833ed487bb4bc347ed3b3d852 : SUCCESS in 25m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/545001c8f0ca4670be70d21b605fe1ad : FAILURE in 42m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7be1f9d7d9d34fbc9220260e581eb83b : FAILURE in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e1f79cad8b1d3abc5c01ce4a8b676cf110aebd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-01 20:07:00.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3456e5d8d135813e39f71ff530ea7436235c744f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 21:03:15.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/2/check/DellEMC_PowerFlex_Manila_dhss_false/82a1731/DellEMC_PowerFlex_Manila_dhss_false/739 : SUCCESS in 33m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"bee2fb957ec58f0dace53a873b6490d94c56b992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:45:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab31aa7cc854977a1231330304ef8b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5342afabbf42a2a807cdb77c72ffae : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/041018710824480d83fb5b13bc63690d : SUCCESS in 57m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/db4729bff3f546deb39220c3af310c6d : SUCCESS in 1h 36m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aee045615e5a45978ef65fc3be6bf5f2 : SUCCESS in 1h 01m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bc0acd62b3904b23aa8c339c3c221226 : SUCCESS in 55m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ea25c6be7c5c45f7801326e1ce93f7eb : FAILURE in 1h 16m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e23acac7ef3e4f1686de4862e887d299 : SUCCESS in 50m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/26d7a3ae8d524f44abf298ce26675a34 : SUCCESS in 24m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/494d12f543f84f0ca272d446ae024820 : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/9bb23abaafcf4831a12ba7c3c04c9fd2 : SUCCESS in 56m 36s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/0c18c7281dda4ccda9b9b65138d12f64 : SUCCESS in 1h 18m 10s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/d8b62793e7cb4b60a32cc80a5c20d6a5 : SUCCESS in 57m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/32a1cd4f796f46a6bd0770af054d9651 : SUCCESS in 33m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/90bb749aec494ce1b5f6de92f2d04ec9 : FAILURE in 48m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/054c69e8745d4cae98d039ca7c19be26 : FAILURE in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e081aa2e9e7e9035090a269e8d9d50eee803ba65","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-01 21:50:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/914823/2/upstream-check/manila-tempest-plugin-ontap-dhss/d6a548a/ : RETRY_LIMIT in 3m 49s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/914823/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/24eb78b/ : RETRY_LIMIT in 3m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e3a04265475b717a32e89e3b3ed729e0250df619","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 22:16:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/2/check/DellEMC_PowerFlex_Manila_dhss_false/a12852b/DellEMC_PowerFlex_Manila_dhss_false/740 : SUCCESS in 30m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8570bbfbbcd785ab43ba8ec01300a1cf34bc1941","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-02 01:21:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/23/914823/2/check/DellEMC_Unity_Manila_dhss_true/145230d/DellEMC_Unity_Manila_dhss_true/1021 : SUCCESS in 3h 06m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/2/check/DellEMC_Unity_Manila_dhss_false/174b3ba/DellEMC_Unity_Manila_dhss_false/1109 : SUCCESS in 1h 17m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e16ec6ba6604cbb4937e0ac7f03f54f8efcfbfe0","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-02 15:04:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccf43560083bef1543680d81ace2250501de5e3c","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-02 15:05:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce09583c891fa265957c800283586c0d34ac914d","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-02 15:13:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ae0b6f84e2617c7092f4f395fea56456e2e73f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:52:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3337e314aaa44a34b54b794bb0e1e9d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b608912e77344218451a12fc7fa5c69 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/20a72de1ad7744f4952cbdd0d096b95c : SUCCESS in 1h 18m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9fb7a165515b44a9a851b103c743ec2c : SUCCESS in 1h 45m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5acb00d9f454443da42b7d6ed8e0905e : SUCCESS in 1h 39m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3668eefb483a465bb89ea2f4159622a6 : SUCCESS in 50m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aca8c6b384b34044b88f85a65c720767 : SUCCESS in 1h 38m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e26bf15511e431fa1725ce3fab5632c : SUCCESS in 1h 23m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ef0c5ac84fa44f06a31e249b1dbf2064 : SUCCESS in 32m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aa701fffaeed4978ba6368fa51407453 : SUCCESS in 41m 42s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/3b8fc5594f4b4fed8c2d6c8d7e662a43 : SUCCESS in 1h 09m 11s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/8e62cf461e85484daaf9d5d27c0a07df : SUCCESS in 1h 06m 52s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/c3bd872ddc0c48ab908ab9296d2720d7 : SUCCESS in 1h 13m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e2ac63eb7a744ef3bff146f90426436b : FAILURE in 13m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f868885fb3124dfeb633cc4401cfcc71 : FAILURE in 52m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/84415890b3df47f79b14bf0d6222615d : FAILURE in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c1237c3f47b90efcb183e6038a09cfa74b3b17a","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-03 16:03:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5738c85a9e33dfc28a868e564ee92255b5914607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-03 16:04:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"af5184507ea6a1e9933fdf1914f631c25f361918","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-03 18:53:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/3/check/DellEMC_PowerFlex_Manila_dhss_false/e1128d9/DellEMC_PowerFlex_Manila_dhss_false/779 : SUCCESS in 31m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"a80c39f99b39dd7158a429ba62b79f9c8abf6090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 19:44:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f1d466d4ec455f8aca50069db47454\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c270afc255b4b0f993f905006d83942 : SUCCESS in 3m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9572886523d1495aa1972877ff8e4f2a : SUCCESS in 1h 12m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a1817fe7832c41abae098a2314d3130d : SUCCESS in 1h 57m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a74e463993148d2b14c6595b2fac7f8 : SUCCESS in 1h 10m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3068d025d5ff47c992ac3e04e86c353e : SUCCESS in 1h 08m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2724cc5ce6e94fe297a5067d5dcdb2a3 : SUCCESS in 1h 36m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff04d16b93cb4432a0ab727bd2405fac : SUCCESS in 1h 13m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d88a43ad2edd4e19b52f6fa105ca46f3 : SUCCESS in 40m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78e2c3fe03854870acdd75fafe7e8612 : SUCCESS in 41m 00s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/e480a713e1994eeeb01620ca96ecdb8f : SUCCESS in 1h 08m 11s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/733d5467899642dabadf678f6ace03f1 : SUCCESS in 1h 10m 50s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/a2a66cd24f8a422ab3855bfb04be6382 : SUCCESS in 1h 18m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ad4ac86ea68d4c4f921271ff8d7c34be : SUCCESS in 37m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f2977266f257419397e7f104c4475215 : FAILURE in 51m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6ec50491ebdc45368fffe6fe48c27b3b : FAILURE in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"abfe456b1cf79428f41b3cfaa5520b772732d34a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-03 20:33:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/914823/3/upstream-check/manila-tempest-plugin-ontap-dhss/1d5509e/ : RETRY_LIMIT in 3m 31s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"35d4fe0e3c613c82702edf1e7e91c57cb2fc335b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-03 21:00:27.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/3/check/DellEMC_PowerFlex_Manila_dhss_false/f2cd1ad/DellEMC_PowerFlex_Manila_dhss_false/783 : SUCCESS in 30m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc2b8066ebc8dab43e70fb9118604dce74514379","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-07 07:47:24.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/23/914823/3/check/DellEMC_Unity_Manila_dhss_true/6928ae4/DellEMC_Unity_Manila_dhss_true/1042 : SUCCESS in 2h 55m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/23/914823/3/check/DellEMC_Unity_Manila_dhss_false/318e9fa/DellEMC_Unity_Manila_dhss_false/1134 : SUCCESS in 1h 16m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4e795d6c1a393ee976e5dfd4d5530b94166be971","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-04-10 20:26:44.000000000","message":"Abandoned\n\npersuing an alternative fix","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"281a042f7e5e8c6362c398a82fc5d0fba803086d","revisions":{"86d32208bc1c92b57a44ba93be9c1bc36c608129":{"kind":"REWORK","_number":1,"created":"2024-04-01 17:58:04.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/23/914823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/23/914823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/1"}}},"commit":{"parents":[{"commit":"1687dd6ef680d13205c9cfb99396752f864987c6","subject":"Merge \"Add access rules restriction tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1687dd6ef680d13205c9cfb99396752f864987c6"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-04-01 17:56:52.000000000","tz":-240},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-04-01 17:58:04.000000000","tz":-240},"subject":"Change nodeset for cephfs-nfs job","message":"Change nodeset for cephfs-nfs job\n\nChange from centos9 to ubuntu focal\n\nChange-Id: I34266790f86162dc04bd055c6156bf47b577a7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86d32208bc1c92b57a44ba93be9c1bc36c608129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86d32208bc1c92b57a44ba93be9c1bc36c608129"}]},"branch":"refs/heads/master"},"687f7ab5cb3d3c578819dd2afba3a4e8751244aa":{"kind":"REWORK","_number":2,"created":"2024-04-01 20:07:00.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/23/914823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/23/914823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/2"}}},"commit":{"parents":[{"commit":"1687dd6ef680d13205c9cfb99396752f864987c6","subject":"Merge \"Add access rules restriction tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1687dd6ef680d13205c9cfb99396752f864987c6"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-04-01 17:56:52.000000000","tz":-240},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2024-04-01 20:07:00.000000000","tz":0},"subject":"Change nodeset for cephfs-nfs job","message":"Change nodeset for cephfs-nfs job\n\nChange from centos9 to ubuntu focal\n\nChange-Id: I34266790f86162dc04bd055c6156bf47b577a7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/687f7ab5cb3d3c578819dd2afba3a4e8751244aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/687f7ab5cb3d3c578819dd2afba3a4e8751244aa"}]},"branch":"refs/heads/master"},"281a042f7e5e8c6362c398a82fc5d0fba803086d":{"kind":"REWORK","_number":3,"created":"2024-04-03 16:04:32.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/23/914823/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/23/914823/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/914823/3"}}},"commit":{"parents":[{"commit":"1687dd6ef680d13205c9cfb99396752f864987c6","subject":"Merge \"Add access rules restriction tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1687dd6ef680d13205c9cfb99396752f864987c6"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-04-01 17:56:52.000000000","tz":-240},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2024-04-03 16:04:32.000000000","tz":0},"subject":"Change nodeset for cephfs-nfs job","message":"Change nodeset for cephfs-nfs job\n\nChange from centos9 to ubuntu focal\n\nChange-Id: I34266790f86162dc04bd055c6156bf47b577a7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/281a042f7e5e8c6362c398a82fc5d0fba803086d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/281a042f7e5e8c6362c398a82fc5d0fba803086d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
