)]}'
{"id":"openstack%2Fmanila-tempest-plugin~918702","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I532ff3f0b2613340e0cb46c3b7e57a22cfa50c61","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-05-14 04:14:58.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-05-10 08:07:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-05-14 04:14:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I532ff3f0b2613340e0cb46c3b7e57a22cfa50c61","subject":"Fix racy DHSS\u003dTrue share groups test","status":"MERGED","created":"2024-05-08 23:18:15.000000000","updated":"2024-05-14 04:14:58.000000000","submitted":"2024-05-14 04:14:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918702","meta_rev_id":"d2515f9ce2700459ecff34e2fd73efd599ea85bf","_number":918702,"virtual_id_number":918702,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 04:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"recommended":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"all":[{"value":1,"date":"2024-05-09 19:14:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2024-05-14 03:20:49.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2024-05-14 03:20:49.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 23:45:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 00:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 03:59:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-09 16:16:31.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2024-05-09 22:11:10.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-10 06:10:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-05-13 19:43:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"0e94580df5ebb509447153abc8c24ed7d2d87b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-08 23:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14d9a915fd4156ae796e8b7730abf56700332c36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-08 23:45:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/02/918702/1/check/DellEMC_PowerFlex_Manila_dhss_false/b33e68e/DellEMC_PowerFlex_Manila_dhss_false/1012 : SUCCESS in 27m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ccbed290e5305a74113e8079c01e52dad9845ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 00:59:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae076f28809342c19e5e8f15da2d6af0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d215c79c5b2c431fad92e82660596781 : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ddf173f0208c4f4ab3e9c08c0296c0b2 : SUCCESS in 50m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a1e4d7face5a49d4bb4f58ad821be2fa : SUCCESS in 1h 32m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/29ad332bcc7244dd880ff9a0c9104775 : SUCCESS in 50m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/72089c93cf2e46b7ba994d25cb1c72a5 : FAILURE in 1h 15m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d9a61c8189e24cc79c267b7bd43f0b09 : SUCCESS in 1h 29m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7dfaab37afc2429ebff6a9122a477cb5 : SUCCESS in 1h 24m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/219ceee152ee4ce9bb7243972384514b : SUCCESS in 29m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cafb4ec6721a48b787db1c876d992a46 : SUCCESS in 27m 49s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/8c13717c2aee4701a6c5a0b35632ddad : SUCCESS in 56m 09s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/53fdba4545144672b8363340d9fbb98e : SUCCESS in 1h 14m 54s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/2b6357a3d5144276b3b6bf3d2c986225 : FAILURE in 22m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d97cd114d06c4036aab059b9583c8bde : SUCCESS in 30m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/de0b4f4396064f5bb84674456fb9c77f : SUCCESS in 41m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9ed2a37382b74b51bf6b08841ab69029 : FAILURE in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"478f14c6dfe1823f1ba689eae8212418c6ecbeb0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-09 03:59:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/02/918702/1/check/DellEMC_Unity_Manila_dhss_true/090fa40/DellEMC_Unity_Manila_dhss_true/1215 : SUCCESS in 3h 15m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/02/918702/1/check/DellEMC_Unity_Manila_dhss_false/aad96d5/DellEMC_Unity_Manila_dhss_false/1313 : SUCCESS in 1h 51m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"161a877a4e283ab90bcf18dcafa8f3eef9238523","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-05-09 16:16:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dfa0b78fc52c31bf2b90c8dd410940b50bafc38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-09 19:14:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"69980be80f8de32b78401afa3d2c57441266ae9b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 20:05:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/02/918702/2/check/DellEMC_PowerFlex_Manila_dhss_false/5ca4361/DellEMC_PowerFlex_Manila_dhss_false/1018 : SUCCESS in 30m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"422d7edd2bc26c204d2657dfa251d2700b1c4907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 21:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/002b2535eb634301abf50860ee0e1e32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8ef564f4184ffe9b06b7e86141338f : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8f37945cf7264101b4f2b60d49977c88 : SUCCESS in 57m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/325a5f6591ff4f7380c24235f537fbf4 : SUCCESS in 2h 24m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f845ccf493144d986f0e8a1ac860e5b : SUCCESS in 1h 19m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e0d7d80e5a11477bb252d1a78fcdeafe : SUCCESS in 1h 16m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/17971bb3a02f4a5b9fb70437851f1db6 : SUCCESS in 1h 22m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3ba51e38262444282f161bc6bb94e61 : SUCCESS in 1h 25m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4d3f5c2e5a924d229c9ee4e1828cbf77 : SUCCESS in 25m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/257d5675dc5c4b61bd7672d6fedb6265 : SUCCESS in 46m 03s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/fb37e6a04b264ba0915af16d06881f9d : SUCCESS in 56m 29s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/a965e8e796a94571be4b6eda0a36700f : SUCCESS in 1h 12m 21s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/c5d62024aa024ff6a4cf17da5cf5a762 : SUCCESS in 1h 19m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f067c9b964ea4e6ea39ed292b8b8e56d : SUCCESS in 31m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9c965ffd7608478f91f834f5eae528d1 : SUCCESS in 47m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/90d1298dee8b4317a7c78be1e982d7b3 : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"437589ef3dae44bb2c33b2fbba89bf42e852f614","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-09 22:11:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/918702/2/upstream-check/manila-tempest-plugin-ontap-dhss/1991d30/ : RETRY_LIMIT in 3m 22s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/918702/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/be5af5e/ : RETRY_LIMIT in 3m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"17adf17dfa0710d1bf3467170c16fead0a0345e7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 22:16:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/02/918702/2/check/DellEMC_PowerFlex_Manila_dhss_false/520ce8a/DellEMC_PowerFlex_Manila_dhss_false/1021 : SUCCESS in 31m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c6bc6023777c5dcc39031fa30b6b8d1df1ee982b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-05-10 08:07:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78cfac0b534f6511303cf0278b8ca96676968042","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-10 08:35:40.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/02/918702/2/check/DellEMC_Unity_Manila_dhss_true/79a6ac2/DellEMC_Unity_Manila_dhss_true/1220 : SUCCESS in 3h 09m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/02/918702/2/check/DellEMC_Unity_Manila_dhss_false/2f75046/DellEMC_Unity_Manila_dhss_false/1318 : SUCCESS in 1h 14m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"269abc6660608e5d83f82d55f26e1353c93c66a9","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-05-14 03:20:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aca3afe2d2a75cf7dac8890a76a415cbae2bb8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 03:21:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfdb860acd22c2dca3a0b318e823fd4f5f0b9839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 04:14:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58fa98df8bf6496ab21edd00aaccfc4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd7b0ad256c4ace84dc62dc82f5c10d : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d932e0d645614da1ac3cbd278631fe0e : SUCCESS in 53m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b026f97fc98848f7a6eb1156d2ab2044 : SUCCESS in 36m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd1cb25864a145b48a3c5614758a31c3 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d2515f9ce2700459ecff34e2fd73efd599ea85bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 04:14:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"986c7bd09d4645b1aefb4ed96b83ecce49091369","revisions":{"290b09517940f27cfc21cadb6cb09f8cd6cd8420":{"kind":"REWORK","_number":1,"created":"2024-05-08 23:18:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/02/918702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/02/918702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/1"}}},"commit":{"parents":[{"commit":"1227d689e42d94028c533d232e7ed3bc34d86867","subject":"Merge \"Remove old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1227d689e42d94028c533d232e7ed3bc34d86867"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-08 23:18:08.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-08 23:18:08.000000000","tz":-420},"subject":"Fix racy DHSS\u003dTrue share groups test","message":"Fix racy DHSS\u003dTrue share groups test\n\nWe recently addressed a bug in Manila where\nshare networks couldn\u0027t be deleted when they were\nstill associated with any share groups [1]. After this\nfix, we had a couple of tests failing sporadically\nbecause they weren\u0027t waiting for the deletion of\nthe share groups prior to attempting deletion\nof the share network subnets.\n\nChange-Id: I532ff3f0b2613340e0cb46c3b7e57a22cfa50c61\nRelated-Bug: #2004212\nCloses-Bug: #2064673\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/290b09517940f27cfc21cadb6cb09f8cd6cd8420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/290b09517940f27cfc21cadb6cb09f8cd6cd8420"}]},"branch":"refs/heads/master"},"986c7bd09d4645b1aefb4ed96b83ecce49091369":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-09 19:14:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/02/918702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/02/918702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/02/918702/2"}}},"commit":{"parents":[{"commit":"6d00585991832165d7a2a1c2ff3701a75d584afa","subject":"Add stable/2024.1 job on gate and remove zed job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6d00585991832165d7a2a1c2ff3701a75d584afa"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-08 23:18:08.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-09 19:14:35.000000000","tz":0},"subject":"Fix racy DHSS\u003dTrue share groups test","message":"Fix racy DHSS\u003dTrue share groups test\n\nWe recently addressed a bug in Manila where\nshare networks couldn\u0027t be deleted when they were\nstill associated with any share groups [1]. After this\nfix, we had a couple of tests failing sporadically\nbecause they weren\u0027t waiting for the deletion of\nthe share groups prior to attempting deletion\nof the share network subnets.\n\nChange-Id: I532ff3f0b2613340e0cb46c3b7e57a22cfa50c61\nRelated-Bug: #2004212\nCloses-Bug: #2064673\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/986c7bd09d4645b1aefb4ed96b83ecce49091369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/986c7bd09d4645b1aefb4ed96b83ecce49091369"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
