)]}'
{"id":"openstack%2Fmanila-tempest-plugin~919619","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4a1c8d75faa8151fb94cd227e24967a06d02ff4b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ocaas","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-06-03 17:21:41.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-06-03 18:16:32.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-06-03 18:16:32.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-06-03 17:15:48.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"I4a1c8d75faa8151fb94cd227e24967a06d02ff4b","subject":"Use devstack\u0027s OSCaaS in test jobs","status":"MERGED","created":"2024-05-14 20:21:25.000000000","updated":"2024-06-03 18:16:32.000000000","submitted":"2024-06-03 18:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919619-ocaas","meta_rev_id":"4bb9577ff5ad192d2c87d25ebd4a4d5c3b7c8755","_number":919619,"virtual_id_number":919619,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-03 18:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-06-03 17:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-06-03 17:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-06-03 17:21:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 20:55:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-14 22:03:33.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-05-14 22:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-14 22:36:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-15 05:06:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-03 15:18:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-06-03 15:18:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-06-03 15:18:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"}],"messages":[{"id":"8823b920c5b25950173213bbbf121eac4371fb53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-14 20:21:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9906c7202fbb00488a44d6dbd503bc2a4b06a6bb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-14 20:21:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0c8f64c555ad0ef653b566b49d1df4be734aa3fc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 20:55:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/1/check/DellEMC_PowerFlex_Manila_dhss_false/52c92be/DellEMC_PowerFlex_Manila_dhss_false/1048 : SUCCESS in 30m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a599d329ad60b0aa03ae47fe0954e101c48af33","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-05-14 22:03:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daaa7504494c2b72915cba681c7387089dd57177","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-14 22:10:12.000000000","message":"Topic oscaas removed","accounts_in_message":[],"_revision_number":1},{"id":"add4f5bef0f9b51dbe0a248b349710aa51cbf9d9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-14 22:10:16.000000000","message":"Topic set to ocaas","accounts_in_message":[],"_revision_number":1},{"id":"7b0021ef4274783ae693a994ba69e86da1498965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 22:31:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c5041870bd14388b3a5453b4a614369\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768810b131ce4f8dbe3f567d2e588083 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7ae168cb988344e0a0f8959826878809 : SUCCESS in 1h 04m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/855c204b41b2477a972b3e272eda393f : SUCCESS in 2h 04m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9b32cef2487643cfab521c9886c209e1 : SUCCESS in 1h 46m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/310f1eabc5f74b62b7a5e55788980242 : SUCCESS in 55m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/981fc4e28c6d46c88740e11c7e57c0d1 : FAILURE in 1h 19m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/36806d80c85f458cb6851b2bc14ebad0 : FAILURE in 1h 06m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/55cdaf52c299438495142650d2527f34 : SUCCESS in 30m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6f4a69526bea436da20a96e3333fc51a : SUCCESS in 33m 38s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/4a6edfdb7bfd47feac895c0ac564ec43 : SUCCESS in 55m 34s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/96b2428bea314bb1b0ded71cf94ccad0 : SUCCESS in 55m 05s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/3973be9cd8bc4ff58227ad9b27c4482f : SUCCESS in 1h 14m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e277420719f34cfc9bb9b1b04f4a3f38 : SUCCESS in 38m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c2089b00d41406cbc1e7effce378b5b : FAILURE in 52m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2f945a1857b7402abfff929be608ad1b : FAILURE in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d80f515d95dcd7bcd92e6b1d6337de1cd1de2d8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-14 22:36:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/919619/1/upstream-check/manila-tempest-plugin-ontap-dhss/4fec3fd/ : RETRY_LIMIT in 3m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/919619/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/3812387/ : RETRY_LIMIT in 3m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7fbaabda1bc6aa15a89e0ad06eed4c533151460a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 23:01:23.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/1/check/DellEMC_PowerFlex_Manila_dhss_false/f56cd10/DellEMC_PowerFlex_Manila_dhss_false/1049 : SUCCESS in 29m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"500fda93f7bd6de6734f1caedd78263ae5a784fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-15 05:06:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/919619/1/check/DellEMC_Unity_Manila_dhss_true/3b4113c/DellEMC_Unity_Manila_dhss_true/1238 : SUCCESS in 4h 16m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/1/check/DellEMC_Unity_Manila_dhss_false/22cf250/DellEMC_Unity_Manila_dhss_false/1337 : SUCCESS in 2h 04m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6658975d5235f02a205d36da59f1a3b813f9a929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-22 22:39:15.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"43484a572b98102665b0c9b553030354a1662735","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-22 23:09:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_Unity_Manila_dhss_true/037e2c7/DellEMC_Unity_Manila_dhss_true/1253 : FAILURE in 29m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_Unity_Manila_dhss_false/8e59a7f/DellEMC_Unity_Manila_dhss_false/1352 : FAILURE in 4m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"bca313e72e06a28a78d2b7db656c4ae7623043e5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-22 23:11:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_PowerFlex_Manila_dhss_false/552edee/DellEMC_PowerFlex_Manila_dhss_false/1077 : SUCCESS in 32m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c01b15db574ce6cafac3a92e1d6298f1ebb29a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 00:16:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540fcbdcf18a42d68e6157fbf0babb4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff0497ddde84ff28ff8f53e462dfdf5 : SUCCESS in 2m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32ddc0bebf334fcd97569f5f75d6a44f : SUCCESS in 1h 27m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fc4cfee0fd794158b0c8d66fb92bdc2b : SUCCESS in 1h 35m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/887ef826c9014fda8c42cebe00fd8191 : SUCCESS in 1h 20m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/959b8bd195ee47d682ec495163ec41eb : SUCCESS in 1h 21m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f7c7378e22bd48b4bfd2e5d2e0b718a3 : FAILURE in 1h 26m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/613b2b1035ba4b4dada94bc034eef693 : SUCCESS in 1h 18m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/76e33bca4ad34e6b888723da7cd778db : SUCCESS in 35m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/40c4ac11990945178b53d78697ed1e40 : SUCCESS in 35m 23s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/26097bf87198473999a64679246df7b3 : SUCCESS in 59m 51s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/ca081ecc014b437b8b8ea0d29dd13266 : SUCCESS in 1h 09m 02s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/132f70eec8604daab11543e830b8d64c : SUCCESS in 1h 14m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9fbb940e3f9a49dd9f28aaa1d9ff0202 : SUCCESS in 21m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6359ee55351400cb2bd8ee49f4d1375 : FAILURE in 48m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b4f0aefd8f5a4bb48457c71137674baa : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f19ff85066dd9c4a6e9aeeb5886f369001a83617","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-23 00:22:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/919619/2/upstream-check/manila-tempest-plugin-ontap-dhss/c3b23ac/ : RETRY_LIMIT in 3m 34s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/919619/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/f7a4bbd/ : RETRY_LIMIT in 3m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cd99ea54a0647ee0b7942e9547a1b30f99f5ca50","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-23 00:49:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_PowerFlex_Manila_dhss_false/f5e72be/DellEMC_PowerFlex_Manila_dhss_false/1079 : SUCCESS in 31m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"37845de833e6a1715ac8300c8c96479eb377812e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 02:13:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_Unity_Manila_dhss_true/cc97170/DellEMC_Unity_Manila_dhss_true/1255 : FAILURE in 30m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/919619/2/check/DellEMC_Unity_Manila_dhss_false/9a5ec99/DellEMC_Unity_Manila_dhss_false/1354 : SUCCESS in 1h 56m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"71a2fd1b09c743aeb3dd1f6d264820c4c9109af8","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-06-03 17:15:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71848da25033440588e0b6f3bc545befcc134d5c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-03 17:21:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14bbe739e16f10aee9988fab070e682fb7eb74cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 17:22:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f343efa41613732a50af8eb39e6d4ff7747ff735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 18:16:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054d6af0bca448aea885cb4a9b04e947\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78bd1745ce1b4bccab133a9dd82d7d45 : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d401f9bb0d534e4586244ba7f6b6ae70 : SUCCESS in 53m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3aa4ee7010bf43a7aad9d0d7fa7abd7d : SUCCESS in 39m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4f03b9b9061743b9af72a39968e2b14c : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4bb9577ff5ad192d2c87d25ebd4a4d5c3b7c8755","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 18:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26fe97cd3b80c55bb5581ad0c5f36ac941b26388","revisions":{"c77447d410566ee8da9044ba61dafaa4a79cea53":{"kind":"REWORK","_number":1,"created":"2024-05-14 20:21:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/919619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/919619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/1"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-14 20:21:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-14 20:21:12.000000000","tz":-420},"subject":"Test devstack\u0027s OSCaas","message":"Test devstack\u0027s OSCaas\n\nDepends-On: I28e6159944746abe2d320369249b87f1c4b9e24e\nChange-Id: I4a1c8d75faa8151fb94cd227e24967a06d02ff4b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c77447d410566ee8da9044ba61dafaa4a79cea53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c77447d410566ee8da9044ba61dafaa4a79cea53"}]},"branch":"refs/heads/master"},"26fe97cd3b80c55bb5581ad0c5f36ac941b26388":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-22 22:39:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/919619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/919619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/919619/2"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-14 20:21:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-22 22:39:14.000000000","tz":0},"subject":"Use devstack\u0027s OSCaaS in test jobs","message":"Use devstack\u0027s OSCaaS in test jobs\n\nI28e6159944746abe2d320369249b87f1c4b9e24e introduced\nOSCaaS and enabling it offers a good amount of speedup\nin our test jobs. So lets enable it in the base jobs.\n\nChange-Id: I4a1c8d75faa8151fb94cd227e24967a06d02ff4b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26fe97cd3b80c55bb5581ad0c5f36ac941b26388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26fe97cd3b80c55bb5581ad0c5f36ac941b26388"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
