)]}'
{"id":"openstack%2Fmanila-tempest-plugin~921419","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0f3e88d4322fecc83721e1e192e14d617281a3ef","project":"openstack/manila-tempest-plugin","branch":"master","topic":"additional_configuration_options","attention_set":{},"removed_from_attention_set":{"36962":{"account":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"last_update":"2024-07-18 23:43:56.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-07-16 23:48:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36951":{"account":{"_account_id":36951,"deleted":true},"last_update":"2024-07-18 23:43:56.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-07-18 22:03:04.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-07-18 23:43:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f3e88d4322fecc83721e1e192e14d617281a3ef","subject":"additional configuration options for scenario tests","status":"MERGED","created":"2024-06-06 05:29:41.000000000","updated":"2024-07-22 07:19:38.000000000","submitted":"2024-07-18 23:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"921419-additional_configuration_options","meta_rev_id":"fc332d5ad9a9a4f09783438c4c77fd70e87fb51b","_number":921419,"virtual_id_number":921419,"owner":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36951},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-18 23:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-07-16 23:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36951},{"value":2,"date":"2024-07-18 22:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2024-07-18 22:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36951},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":36951,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 08:29:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-06 08:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-06 23:10:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-11 00:15:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-06-11 02:41:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-06-11 07:22:31.000000000","updated_by":{"_account_id":36951,"deleted":true},"reviewer":{"_account_id":36951,"deleted":true},"state":"REVIEWER"},{"updated":"2024-07-15 21:55:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-07-15 21:55:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-07-15 21:55:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"8742445039e31a6f6f78ef6513a8dcb34bb26e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-06 05:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45969bd4b05b57d39ab2fb59c1a9e30ba62f0ffb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-06 08:29:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/1/check/DellEMC_PowerFlex_Manila_dhss_false/f8473f2/DellEMC_PowerFlex_Manila_dhss_false/1130 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6be82f4f8b883fc50551b37b568d138d63cd51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 08:35:56.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/357620da71da45a0b07a8be54cbc2f1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da64083e63b47de8475f1ac30c27ce8 : FAILURE in 3m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c2b28148ad2249adb5cebf8d0f607736 : FAILURE in 1h 15m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b2f92449424841d68775d96c0d869f61 : SUCCESS in 2h 29m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af3f2613b4a64f048b434d43d1803d3a : FAILURE in 57m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e604f7d521df47c4933e255c8de2081a : SUCCESS in 46m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e6610d17f70f4a4892852060369ba091 : FAILURE in 1h 28m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a0075bc209b499b8d2eb3388dfe85b0 : FAILURE in 50m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04d47e2ce6294670b18450753ea6acf2 : SUCCESS in 29m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9e8f63d67f75471b942ec88d87a3589f : SUCCESS in 37m 39s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/6a51175693914e7aa8d80229af9612b1 : SUCCESS in 1h 16m 35s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/9396125433b24e4d9432f215d0d62cda : SUCCESS in 1h 17m 24s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/e3c98c8dacff460da22981799910d209 : SUCCESS in 1h 22m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8c12d5ff04b844b7800527a73f8c51c9 : SUCCESS in 32m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/31a1bb69207e4e9a8fd5f3fc5b65867d : FAILURE in 33m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f2e105870cf84d8ebf8032de763966fd : FAILURE in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bf86720a7bd22f6edc24b814420fc7624bb85af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-06 20:18:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee9459fa88028cc94a06495022db122cc8b74e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 22:44:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b321cdb2b04c4536b38176af1d05f59d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7e977137574986a804bf6f16104ed4 : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5df52abfdb2e4d8a88c56b9b0bad4c95 : SUCCESS in 1h 09m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/01d941fef3444a12946f8606f694f543 : SUCCESS in 2h 17m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8df6a8ba87ed492ea33931b21d6b99bc : SUCCESS in 1h 44m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9299eb74cd9f479794591f3bd7ed14a3 : SUCCESS in 1h 05m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d133763bd8ae45fd9008faa2ef8d078e : SUCCESS in 1h 33m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b405dc1546a440e84ac447ab8a08faf : SUCCESS in 1h 13m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7526a4ee32f7494c85f8a8c68f8812c3 : SUCCESS in 37m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ab4b6d430cd4216a4e1bc151c2f2416 : SUCCESS in 40m 27s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/7c917198ae684349b1b274e77e14e3b7 : SUCCESS in 1h 18m 12s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/0496b0b8af6746aeb4f3d3c8933fd164 : SUCCESS in 1h 12m 14s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/c65f92af63134a8b83b6217bfc134e52 : SUCCESS in 1h 07m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a0d929b487ca48f491d5b6aa93b12905 : SUCCESS in 38m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1b28702b49d74f96aa7b55ffb05fdb28 : FAILURE in 40m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0310f26eff04b7cb674380a7b15a384 : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2b0a980172a44080b8565f5b99384c1efd61044","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-06 23:10:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/2/check/DellEMC_Unity_Manila_dhss_true/b498d87/DellEMC_Unity_Manila_dhss_true/1292 : SUCCESS in 2h 51m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/2/check/DellEMC_Unity_Manila_dhss_false/777b944/DellEMC_Unity_Manila_dhss_false/1390 : SUCCESS in 54m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6717dfdce15426507c39dcc0a38143d5b41f9bcd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-06 23:19:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/2/check/DellEMC_PowerFlex_Manila_dhss_false/fbec1cf/DellEMC_PowerFlex_Manila_dhss_false/1132 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"65ffcf411de6bdcca572fd458c51b32546d4933b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-06 23:25:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"38a4a2ba3e80605aa1aa8e844f21c3f7b4eb818a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 01:56:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a599d20abae144198a9fda7ffd96fff2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6348a1ce7d524fa9b2462227bd041ef5 : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9103914cd12d44abaa2cba993e8355a2 : FAILURE in 1h 27m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/73323648f48a4a3fabe86970d4e1c2ef : FAILURE in 2h 25m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/462391c7c0434a0eb179281250ef4edc : SUCCESS in 2h 05m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1f944242f6794d5b8526df5bb01bb177 : FAILURE in 1h 04m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c62a2b47cfc24daf966e5826a47bcc20 : FAILURE in 1h 39m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1b986cc57cd4fc585feb474dddcb97e : FAILURE in 1h 12m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d32cd21f8ef64c13b39967a21d10a654 : SUCCESS in 29m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4142c3f270774b2bb507144a5ff60a72 : SUCCESS in 40m 29s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/d8765d4d570940f2ad7cc9ba5c7f84c3 : FAILURE in 1h 09m 08s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/727d8652182f4351805627f8a04907ae : FAILURE in 51m 50s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/a7d3326733db458bb1eb53ee1fbf2f67 : FAILURE in 46m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6479c6336d744e25b4c1e9308bd8e642 : SUCCESS in 20m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/617e7aac5c924c3fa8456bcee7cdf4de : FAILURE in 31m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/21cf9f645026400ea34deb0949ad6913 : FAILURE in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3360bfa318ed5c52d0d43eebf142905e7cf98999","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-07 02:25:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/3/check/DellEMC_PowerFlex_Manila_dhss_false/7dfa054/DellEMC_PowerFlex_Manila_dhss_false/1133 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"f75cda80ca90dd5afdc9f2005db50a28fffc6489","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-07 02:33:26.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/3/check/DellEMC_Unity_Manila_dhss_true/475843d/DellEMC_Unity_Manila_dhss_true/1293 : SUCCESS in 3h 07m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/3/check/DellEMC_Unity_Manila_dhss_false/0e609b1/DellEMC_Unity_Manila_dhss_false/1391 : SUCCESS in 56m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b0a0a50454fc92364bd3b082da9e411eed3dabf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-08 13:21:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a8c28262a87b7f325bff1d722cd35c95d7b106db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 15:03:54.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b72453f34f462ba0deddb2adaddbb2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f26b7e14843465eb225f978828bca29 : FAILURE in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2bfd41cb03564b348303925d17bfc5c8 : SUCCESS in 1h 05m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/31bf2ff3d7584859a0015bef8892e003 : POST_FAILURE in 26m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99f9fc2bc6de424eb47add33ffe34dd1 : FAILURE in 1h 17m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a4d2cdb9dee4457da75debc641eaa545 : FAILURE in 1h 02m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1381c59431544bf99db34bec42421776 : FAILURE in 1h 36m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b5f8fddd8f584ed0ba47a5d47fc152d1 : FAILURE in 1h 20m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07b41a824a92404095b36de46d6a6f9c : SUCCESS in 36m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7778c11fbe30462ea71fef2b6b4191ce : SUCCESS in 41m 47s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/7046a09579d8466a8430e7ebfc691f9e : SUCCESS in 1h 10m 28s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/52f1d090e3a54d04b9ee85be1a53dda4 : SUCCESS in 1h 17m 01s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/adfb8966f2454610a62a1ef56a7f088c : SUCCESS in 1h 23m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d19d53bbbf65402c8c95ea143bc3c543 : SUCCESS in 38m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/61f365a3611346b3989fd3c593a072fb : FAILURE in 46m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1fecee86c4c44347bfae08239824f8f7 : FAILURE in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1471945721e0d2f5bc98bc77d31ed27ab1f7e896","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-08 16:21:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/4/check/DellEMC_PowerFlex_Manila_dhss_false/190c932/DellEMC_PowerFlex_Manila_dhss_false/1139 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"72f47ef73e50b26bf8bb3391daa607eabb7220fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-08 16:21:47.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/4/check/DellEMC_Unity_Manila_dhss_true/ed6bac0/DellEMC_Unity_Manila_dhss_true/1299 : SUCCESS in 3h 00m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/4/check/DellEMC_Unity_Manila_dhss_false/8c93d24/DellEMC_Unity_Manila_dhss_false/1397 : SUCCESS in 55m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6accc9906d826ce59c45461b84e9be5e369e4344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-10 04:08:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fefe35a747f0d14c818021048569d5c745beb4a5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-10 07:08:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/5/check/DellEMC_PowerFlex_Manila_dhss_false/72a7db4/DellEMC_PowerFlex_Manila_dhss_false/1140 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"aed5645abf1b85433b133f298ee886dc327b77da","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-10 07:14:05.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/5/check/DellEMC_Unity_Manila_dhss_true/9fa747d/DellEMC_Unity_Manila_dhss_true/1300 : SUCCESS in 3h 05m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/5/check/DellEMC_Unity_Manila_dhss_false/1b4a4b0/DellEMC_Unity_Manila_dhss_false/1398 : SUCCESS in 48m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b3fa75663dc52b76c8cb830e907f8f7674bf47d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 07:23:41.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6221ff9e76824a98a8335c4a6fb43584\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ac86789c30414ba373929c5c62f52e : FAILURE in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/80728681b01a47459c3ce0511690217c : SUCCESS in 1h 14m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/36c68ff11fba4a228270b3c79d8ab414 : FAILURE in 3h 02m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bdbd8991f5c94f2ea3d161bfca15ccdf : FAILURE in 1h 15m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b372e3d584724d37b9db00758f9c70ca : FAILURE in 1h 06m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/be64d6a5de094ef99d626dfa950d884a : FAILURE in 1h 19m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e78191574ee940b48d8dda726bfef2d4 : FAILURE in 1h 05m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/13be58fe6ee7457f8a23ee136f98f0ea : SUCCESS in 30m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3119488785654092b3271699a5de53ba : SUCCESS in 33m 44s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/1f00090d9fe7468b912aa7c068f9ef2a : SUCCESS in 56m 37s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/b8d7ea2dc41247208be54bdfbeb477d7 : SUCCESS in 1h 05m 28s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/0e67dd39edf744719cb4b821093fbd40 : SUCCESS in 1h 22m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/53232d7eea994bb3b28012223147d1e8 : SUCCESS in 33m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7e2bf1c92f0d471d8d85fa6424b52e4f : FAILURE in 55m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3fbb4f8acf894fd4a0c0ece129f917c4 : FAILURE in 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1450f13738d668b70b959a6259093edd9583d430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-06-10 22:57:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"908e7165d5545ccd923e81a4871a386f8a9636ab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-06-11 00:15:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5fb8b0905e5a5301a5857bf114da2c6648ab7420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 00:52:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa97b15d35e64e3ba935361baf60304c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5bd8f67ff6943d781f8fbab627a2b62 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e51b2d17bbb44f00a1d9c7b78781129d : SUCCESS in 1h 05m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/74de07761b9547b3811b6f81fbffc97b : SUCCESS in 1h 41m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9198371d75044522b2512258749a4ff7 : FAILURE in 1h 13m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b63c4db5ff65407ba309fe60599b8be4 : FAILURE in 1h 11m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c8387807264b4a23a4249987b53bb541 : FAILURE in 1h 39m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cbbcd3153d014d0db5ea5809e9780e44 : FAILURE in 1h 25m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22819d56549146fa8923699cff9ebd70 : SUCCESS in 36m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7b87b64a6f12453a8e68c7356d4e0348 : SUCCESS in 37m 58s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/9ae86bc625504ecfa2ffa40d3c0a2f2e : SUCCESS in 1h 05m 51s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/92f0602e4cf4436ba3337ca0083a77ea : SUCCESS in 1h 16m 16s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/ea2c24752dd247dd8e5a90274abd1f03 : SUCCESS in 1h 06m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/564a0c3b2f914483b543b7a4b381522b : SUCCESS in 34m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2945b28bcace46038a66706033538346 : FAILURE in 44m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/70634b3b7a404c3d894c73b600abb51a : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dddb8a14d76964324502cdfb044fde486a759220","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-11 01:28:07.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/6/check/DellEMC_Unity_Manila_dhss_true/811359c/DellEMC_Unity_Manila_dhss_true/1301 : SUCCESS in 2h 30m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/6/check/DellEMC_Unity_Manila_dhss_false/bad5103/DellEMC_Unity_Manila_dhss_false/1399 : SUCCESS in 1h 11m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"dc42950cc1bd198e8df436e3cb41db163b4d8b6c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-11 01:58:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/6/check/DellEMC_PowerFlex_Manila_dhss_false/6f3d4a1/DellEMC_PowerFlex_Manila_dhss_false/1141 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"da575a4ec130d568f052403779ab5e3ab0b4d2b2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-11 02:41:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/921419/6/upstream-check/manila-tempest-plugin-ontap-dhss/e92591e/ : FAILURE in 1h 19m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/921419/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/728dac9/ : FAILURE in 1h 37m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"41072f204d8eeefce0d915e56801263184d86d33","author":{"_account_id":36951,"deleted":true},"date":"2024-06-11 07:22:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0c28d78b86245bc728e4ac004953830119d0fa93","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-15 21:55:54.000000000","message":"Patch Set 6: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fc86bb1454f73e08a553fa99ed58aab4c3b2095d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-07-16 23:43:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"360762a49e7480da5865925440e3824132ff1dd0","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2024-07-16 23:45:10.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4ac4431aa82b9d22e136e669e068cae004298dc9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-16 23:48:51.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"795f4d53ffe6962603bc992ed69ce2f03713b17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 01:11:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e50be96b5ae34685a05dc3c353635a1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76be4fbcd0642eca63cf0643e457465 : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7be9a569496d4969a13c2c8540515726 : SUCCESS in 1h 06m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb42292fff9f43c7ad41907c1aa3d72b : SUCCESS in 1h 27m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e5aee455b44845d4a26ff5333606fc54 : FAILURE in 1h 04m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6779e3bc4bda45bcb60289d608421699 : FAILURE in 52m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/64d9aed3fa21450ab001fab2a1363225 : FAILURE in 1h 04m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0fb2380e375f42db9ba48637f6414882 : SUCCESS in 51m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/652cfb6d125f4442a0fc6fffaa3ea290 : SUCCESS in 36m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f7064905c67440f18a87fa43fe3a878d : SUCCESS in 38m 57s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/e00059808d75499088271ae471f3f187 : SUCCESS in 1h 10m 33s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/2405ae0f9d924231a8932d8f19bd3484 : SUCCESS in 56m 18s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/4afe5121ed1446118034fe766fabfdbb : SUCCESS in 1h 19m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9eed83d981fe4e7bbde0f44a5229cfb8 : SUCCESS in 27m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2ad0d1e3a46949998165b128e255f9ee : FAILURE in 46m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/596c7e0259884f22b912d77f3183cf87 : FAILURE in 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b676c9fe5f039c2eb4bb23eaa4a364cd38a0eaa4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-17 01:31:47.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/921419/7/upstream-check/manila-tempest-plugin-ontap-dhss/de3f50e/ : RETRY_LIMIT in 15m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/921419/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/b7418ee/ : RETRY_LIMIT in 19m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"5cd9e61712471484974c16c77c09cefb63ce5d8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-17 02:54:51.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/921419/7/check/DellEMC_Unity_Manila_dhss_true/e6f07b0/DellEMC_Unity_Manila_dhss_true/1416 : SUCCESS in 3h 11m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/7/check/DellEMC_Unity_Manila_dhss_false/b1c81fb/DellEMC_Unity_Manila_dhss_false/1517 : SUCCESS in 59m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"cac9096dce7353b193e7a28f7d81a90473cd65db","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-07-18 22:03:04.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eb3ff5dc80f7025be34f17de1c46a0e1c4018893","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-18 22:32:31.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"57689cc5c97cd020282c78f05211f5377c78b7e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 22:32:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1574e3ad0d5462da85cac57588954c5d38b969db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 23:43:55.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1054e4d1fe14541a0ce3562dd3b39c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c7548f61d5463cba7a95359472ebb6 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/affbe1363c6d4ba783e492860558004b : SUCCESS in 1h 10m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/389e6ba6ea6842a5a58714527de99c6c : SUCCESS in 40m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/252af006b29c4105bed55b7c29400f83 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":7},{"id":"5a099ee75b26c43ced46a96e3c0adfcbb075ab51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 23:43:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"fc332d5ad9a9a4f09783438c4c77fd70e87fb51b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 07:19:38.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/921419/7/check/DellEMC_PowerFlex_Manila_dhss_false/1558d37/DellEMC_PowerFlex_Manila_dhss_false/1292 : SUCCESS in 31m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"caf8b9025c086f588d1d9340d4bc8cbbab92e868","revisions":{"033d36246149fb822865386d6ab3105d1dfcbdb5":{"kind":"REWORK","_number":1,"created":"2024-06-06 05:29:41.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/1"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:40.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/033d36246149fb822865386d6ab3105d1dfcbdb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/033d36246149fb822865386d6ab3105d1dfcbdb5"}]},"branch":"refs/heads/master"},"f452b7a197fdc4258ac006f186de367b357ca203":{"kind":"REWORK","_number":2,"created":"2024-06-06 20:18:56.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/2"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 20:18:48.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f452b7a197fdc4258ac006f186de367b357ca203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f452b7a197fdc4258ac006f186de367b357ca203"}]},"branch":"refs/heads/master"},"ac7edc9bbab803d90e4729c5458884ea1773f913":{"kind":"REWORK","_number":3,"created":"2024-06-06 23:25:32.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/3"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 23:25:23.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac7edc9bbab803d90e4729c5458884ea1773f913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac7edc9bbab803d90e4729c5458884ea1773f913"}]},"branch":"refs/heads/master"},"2aed9e5a9e8f39fb2abe408930bf291c7cb02fec":{"kind":"REWORK","_number":4,"created":"2024-06-08 13:21:10.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/4"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-08 13:21:01.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2aed9e5a9e8f39fb2abe408930bf291c7cb02fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2aed9e5a9e8f39fb2abe408930bf291c7cb02fec"}]},"branch":"refs/heads/master"},"9d077adff4fc2484cec4ded65bea86502c1ba90d":{"kind":"REWORK","_number":5,"created":"2024-06-10 04:08:12.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/5"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-10 04:08:04.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d077adff4fc2484cec4ded65bea86502c1ba90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d077adff4fc2484cec4ded65bea86502c1ba90d"}]},"branch":"refs/heads/master"},"732418cde54c31f36a8e33a56b1d467de45b312a":{"kind":"REWORK","_number":6,"created":"2024-06-10 22:57:47.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/6"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-10 22:57:40.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/732418cde54c31f36a8e33a56b1d467de45b312a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/732418cde54c31f36a8e33a56b1d467de45b312a"}]},"branch":"refs/heads/master"},"caf8b9025c086f588d1d9340d4bc8cbbab92e868":{"kind":"REWORK","_number":7,"created":"2024-07-16 23:43:10.000000000","uploader":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"ref":"refs/changes/19/921419/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/921419/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/921419/7"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-06-06 05:29:30.000000000","tz":180},"committer":{"name":"Volodymyr Boiko","email":"volodymyr.boiko@vastdata.com","date":"2024-07-16 23:43:02.000000000","tz":180},"subject":"additional configuration options for scenario tests","message":"additional configuration options for scenario tests\n\nChange-Id: I0f3e88d4322fecc83721e1e192e14d617281a3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/caf8b9025c086f588d1d9340d4bc8cbbab92e868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/caf8b9025c086f588d1d9340d4bc8cbbab92e868"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
