)]}'
{"id":"openstack%2Fmanila-tempest-plugin~922117","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-05-30 21:10:57.000000000","reason":"Change was submitted"},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2025-05-30 21:10:57.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-30 00:09:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-04-17 21:54:39.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-05-29 12:38:06.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2","subject":"Add CEPHFS filesystem metadata verifications","status":"MERGED","created":"2024-06-17 14:32:17.000000000","updated":"2025-05-30 23:10:36.000000000","submitted":"2025-05-30 21:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922117","meta_rev_id":"4deec69f642647630c9b06eebc95ba6ea39c51bd","_number":922117,"virtual_id_number":922117,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 21:10:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-05-30 00:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-04-17 21:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-29 12:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-05-30 00:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 14:39:29.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2024-06-17 14:39:29.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-06-17 14:39:29.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2024-06-17 14:39:29.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-06-17 15:05:52.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 16:45:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-17 16:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-04 16:32:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-05-30 05:10:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"33aebe52b604fe670d0308a37aba879620581505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-17 14:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21a66bbcf9d4b5ad615a305fb77713e7be11fc3b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-17 14:39:29.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fae8c7d142be297651a4d6fa6ce2b88dba1b1159","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-17 14:40:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a199dc6199507dced1ad8975757ca00d4b54370f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 15:05:52.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/922117/1/check/DellEMC_PowerFlex_Manila_dhss_false/436475c/DellEMC_PowerFlex_Manila_dhss_false/1181 : SUCCESS in 33m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"56e573fdd0536cd7a461507373c690dde062cef3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-17 16:45:50.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/922117/1/check/DellEMC_Unity_Manila_dhss_true/f961125/DellEMC_Unity_Manila_dhss_true/1325 : SUCCESS in 2h 13m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/922117/1/check/DellEMC_Unity_Manila_dhss_false/3cf500c/DellEMC_Unity_Manila_dhss_false/1424 : SUCCESS in 51m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6cb9d57189b8b5abcb86d95ecebe0394ee4167fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 16:49:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1614e54519d6479fb8bfc95918e9aa6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f9ee5b5baf4a85aca2fcbe91dbe809 : FAILURE in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d0e4c3bb86864abba30c133b038a5b96 : FAILURE in 17m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/95ca2c1e629442efa5963fcc39a34678 : SUCCESS in 2h 11m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f940ed6614fb4a7d87c6abc0b9f5214f : SUCCESS in 1h 34m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8168aff2209a42808d068ee6c6ae5f57 : FAILURE in 1h 06m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7fbf8f1ae1be47a88210cb931d6b419d : SUCCESS in 1h 55m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0340a3255e8b41cba607e961638ee5f3 : SUCCESS in 1h 03m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4970d4c342e94e138032e49c1817e18c : SUCCESS in 33m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ffd4027a3e264b058d1c9af69bc9888a : SUCCESS in 34m 04s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/ecffce7696b3496ca770a71d00ba7fcd : SUCCESS in 59m 26s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/ab8a461df62c4c08839ef8dcbc58c763 : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/c398dcda48194137b22e13d386548f58 : SUCCESS in 1h 10m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9f22f779aee94b5eb835cfa378b053ba : SUCCESS in 29m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/25106ff8d302493dbd626dad6ac7111d : FAILURE in 58m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c70dd2fb912249f09ed182fc2ad4e950 : FAILURE in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1f971361e9382b81eb4f585f86b9e232a91434d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-17 17:10:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"92fc7db7a7435b3afadfef7a3f29fffbbfc49ce3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 17:43:32.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/922117/2/check/DellEMC_PowerFlex_Manila_dhss_false/b421b21/DellEMC_PowerFlex_Manila_dhss_false/1182 : SUCCESS in 32m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"fe5f946c4763513316c2e23a481273096100fbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 19:11:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/506b4947befd464eb7e92537faf3a3de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402e6f0605944e98b1119eb7060c3da5 : SUCCESS in 3m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f5eb16beda3a4eac8b563be98a9ad40d : FAILURE in 27m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8655c227aec6416e9716b90f963f4d0a : SUCCESS in 1h 59m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e883a6f342bb4b619be17247d5e84cf5 : FAILURE in 1h 10m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/783e87926e994561ae0d40c45bd7cae9 : SUCCESS in 1h 10m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/75ee24ee0cfb4046a632546a7c8cc99a : FAILURE in 1h 43m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/78e5b4549ac8459cad3e4ee6c2176928 : SUCCESS in 1h 13m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0f0950fe235438b84a0f78c65333925 : SUCCESS in 37m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f2c4434ca7b2491d9ea0faf57280d23c : SUCCESS in 42m 54s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/127a820f9100462180aa3a7883cf9be2 : SUCCESS in 55m 27s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/5fa4961586ab4dd3801b6f2563dff083 : SUCCESS in 1h 21m 33s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/272e674306744ba48098a7eb911434de : SUCCESS in 1h 20m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c78b04b8f6b142ec91720b1edb7644a7 : SUCCESS in 22m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49c7400a5a284fceb15230672ccf9143 : FAILURE in 54m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9a567c823fad440fbeac42053ea64441 : FAILURE in 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8a979ba9f9777d49a99b55784e3e7f59a9d3c43","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-17 19:13:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/922117/2/check/DellEMC_Unity_Manila_dhss_true/c460a44/DellEMC_Unity_Manila_dhss_true/1326 : SUCCESS in 2h 03m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/922117/2/check/DellEMC_Unity_Manila_dhss_false/074a0c0/DellEMC_Unity_Manila_dhss_false/1425 : SUCCESS in 50m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e929a48b79c26a557b1afc7d377a302c40bef5e6","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-06-18 07:47:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2203066fe5d70d004e2709f8558d559e3b224ad3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-06-27 23:37:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82c0ecf07f8ba7dcd9c69788c3dee1540afa728a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-13 13:50:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c4a9166cc508ac2b6fade97baafcc8e2b52c23f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:58:43.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cc5bff8ed3c4edd8ae046c7b8885008\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3653a35c74c34e29b18c39577e1edfc1 : FAILURE in 3m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/25af4cfe3d0f423aac527cb89a78b530 : SUCCESS in 1h 08m 45s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/011e85f7ef114b178b7f80a847ace9c4 : FAILURE in 1h 04m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3d69d69fafbd4f6d8a762bf34346ab8c : TIMED_OUT in 3h 04m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/809e9aa82d274b82b528ebac853a167c : FAILURE in 1h 37m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cd73f1cf5e7048aa983e771e7b63264f : FAILURE in 1h 22m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8c25695da1194857be0ca7f6006bc626 : FAILURE in 2h 35m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fabb6da6f5d64883adc9ba1466cbbd1c : SUCCESS in 1h 06m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bab833982bc74b71a3e9e2b68553d819 : SUCCESS in 37m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/49c45f53527844d8ba6a02d629301a3e : SUCCESS in 34m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a1d12bf300a2439880649122c724cb51 : SUCCESS in 1h 03m 27s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3dd3371325c0426fa58ae92568c0b7ad : FAILURE in 1h 36m 55s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/bfb783dcb1464a6d92b3e93d092925a6 : SUCCESS in 1h 05m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3a8e837f07cd497c9e823dc68f424c66 : SUCCESS in 35m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/83a5dd16f2e345948f1b93f6b3fb7026 : FAILURE in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc8bd6b0b749532b19c8d263040de6a2c3734ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-03 21:21:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fd39fe5eff0b564051c0f0cec5ebca95c525444d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 23:13:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/189de842495b45358573d9d9da0eb42a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e29f4344eb744c4adb7b4f0caa39618 : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e87e363af3444cdbb5d558630d5c1efc : SUCCESS in 1h 13m 33s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/2dcaf0121b2a445482e9681af0bd43c8 : SUCCESS in 59m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/90ed4274c6d6486ab1c52b454a255f1f : FAILURE in 1h 50m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c4d1d3d9a2474d4da55940f0e984f1d0 : FAILURE in 56m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9c6ec61a72eb48ac99ddf3623a4b4b17 : SUCCESS in 46m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9b02d784ba6946d7a95f801856ea3d6d : SUCCESS in 1h 25m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/701647f3007044178d90e42f00aa614b : SUCCESS in 54m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/48e3793160b64edba4ed69691ef31b58 : SUCCESS in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4c7469afab1e43e9abb400f956a23254 : FAILURE in 27m 57s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4ad113465e1f4a6b8ff0a372ffc12926 : SUCCESS in 1h 10m 42s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/f9a7c05e8f4a4092bf67a35125f4517d : SUCCESS in 1h 05m 53s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/63cb790171774f45b8cd263d900ed81d : SUCCESS in 1h 13m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/27a27a9eb376453dadefe27f2f222395 : SUCCESS in 24m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7547efcd23294785bc518548ea025871 : FAILURE in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35dc970462c28f64b4558eceab6213e921749401","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-04 11:13:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1dd7969e505c025a38ed9ef800ab04a6d171ba13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 13:30:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db04475338f4cc39d19aa0e63b2e92e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27270219c1584c61bd7c2fa6157795ad : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5f1091ecdcee44ce9d46439a379a7ae9 : SUCCESS in 1h 01m 27s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/eed088885f134a57a91823075cc996cb : SUCCESS in 57m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ca42ebdbc9f1484aa52fe3e40fb318c8 : SUCCESS in 2h 15m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ef0ff709c4034460af19020b0acba107 : FAILURE in 1h 03m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c3cfa0b33f224babb603bcfd5f69d8e0 : SUCCESS in 1h 05m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f87c83d331d94df1aa81a9f9431e36a6 : SUCCESS in 49m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0c0d25891a644b2b4d26e5bc5d8127c : SUCCESS in 1h 04m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/608a19bc13fd450cbcef57a770f0c2d4 : SUCCESS in 28m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/896705e1663e498b8f1383776d7c54c1 : SUCCESS in 38m 06s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8f51a03c964242f38cce1f9b1da9aebe : SUCCESS in 50m 51s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/714832f7a169449cbaa4f021efd35d37 : SUCCESS in 1h 05m 53s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/830b52b575014945a4f907600b41304e : SUCCESS in 54m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7076c62b523746eeae151e3c8518beba : SUCCESS in 36m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/38892eb1ab4f425b8ea708d03f1e6ddf : FAILURE in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a6101405922e07f05bb8b29a0cf989df4da0d7e5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-04 16:32:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"4ad54973324e0f9418b7d909e8bd47745aa681f5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-04-08 21:51:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0642963928195705c7e672a6d8922a13f5482f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-16 19:33:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0a90d6da487bc50c7f9daa44a8d034b0381cde30","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-16 19:33:22.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ba7d988510978a29f7b6121f5f86a8bc8b408f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 21:19:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a470de569c849bd85768ab9fb7b9927\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfcda65482f43bdbad2d101cf6a2522 : SUCCESS in 2m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bb4863f6a8274d2dace6f5c54e70e440 : SUCCESS in 55m 24s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/5ba0ea732c6a4ae490d3c53afa74ffc4 : SUCCESS in 53m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a334046724f946028626582c07c95394 : SUCCESS in 1h 45m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/09851a595dd1419b8d82c23c2f0ae411 : FAILURE in 22m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/55a09712acff4f61afabcaab96617cd7 : SUCCESS in 50m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/194d84275c1348e190c49282567c0564 : SUCCESS in 49m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2bb9ba5027a74e0584e431c5a12daca0 : SUCCESS in 55m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/304af37ca02145dfa5a7b0d907e7a24c : SUCCESS in 24m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3f44998a1367426e8c871dd4ce024f58 : SUCCESS in 35m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4c690e33c0cd42f28013cf1e8bef8ed4 : SUCCESS in 56m 26s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/c85d975ba26245b4add935de422eaef2 : SUCCESS in 1h 00m 57s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/68993790c5884e539547f60893014265 : SUCCESS in 28m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2419825ea90a46ebbf83b247670150cb : SUCCESS in 19m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79181e629afc43d28db022639a3b91f5 : FAILURE in 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0ff4e1bcdc12b81b7b433c9d9c446002f6981219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-16 22:40:22.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fcac95311461d4cb461433871ec53548e5e46f9a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-16 22:40:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e5b1c5b3a7bf80cc4322a62a7e2f957469edbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 01:23:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b121a8575512496890e5890d4956728d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26c0b75b1a249aaa04b88e03278ad88 : SUCCESS in 2m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e0f655f1eec544f6b0fee5363c643443 : SUCCESS in 1h 18m 06s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/8cdd417ec1b9463f84119bdd4fbea90d : SUCCESS in 1h 11m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/473ac16830134574b640cb0c5d357d26 : SUCCESS in 2h 38m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/29cc3eda41694b9382b73982daea8992 : FAILURE in 1h 04m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cc6d2d8ffad44c058cd577359f0fd8bd : SUCCESS in 1h 06m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c88ec7491633465a870fd07b3d9482c1 : SUCCESS in 1h 28m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b2a8501c4abd4b4bbbb34ba9c560fe50 : SUCCESS in 1h 10m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7681ae77929d44a482edd672a204031b : SUCCESS in 31m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1bd4ff1aca40466899be5896f697cbfe : SUCCESS in 20m 59s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1649d2778de04b96bd2caf88f3009460 : SUCCESS in 1h 05m 25s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/dd2715dc1798435cb6edbe7a9c932b3a : SUCCESS in 1h 02m 05s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/98e657071e4145b79fa78146e190c7fe : SUCCESS in 32m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3c2c067939fb451594953014c3e3b7cd : SUCCESS in 33m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2e441831243a45ffaa89ee6eb7a06bfc : FAILURE in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a292b0578e9bcaff394c3eb2d0c02125ed2b2c4f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-17 04:24:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"ca3dfbdfef03f9b31c8fb4d76fdc0e51e1959432","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-04-17 07:54:40.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a889f986a0d5728b7fe4c249ab01adb86bb729eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-17 13:01:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6696b2c9d9402df6122533c19cd7db2a9dd43cdc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-17 13:01:41.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c62be1302a044cf54a025f240a07ecefb1c71cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 15:09:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c95841ef7354833b0bb4b48cac63d29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c422c4f3208b41549e2545d3c55c41fb : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/24c26550339e46e3860ca01639574843 : SUCCESS in 1h 11m 39s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/805f46fceacc47518c6ae43992ec6d6e : SUCCESS in 57m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/54bce58cb6174fdba2783ce935b4be51 : FAILURE in 2h 07m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e8f68a879e46433e83436e5ab4cdf098 : FAILURE in 39m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bf85871f6fb444d499b6b6d0dda81631 : FAILURE in 54m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bf35736abc514063b3e6b6e4620caca9 : SUCCESS in 1h 33m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e571d70f3bd74647ad3bed0d336f7c49 : SUCCESS in 1h 03m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3fcd25735ed04901a2315fde2ec90ef5 : SUCCESS in 30m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5278be67562e47c79040dfcdf986c3fa : SUCCESS in 31m 13s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/95484e475b044ff491802b9d6cb2eb0e : SUCCESS in 1h 10m 37s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/559e24c47bbf49f2a8c76644fee753d8 : SUCCESS in 1h 08m 18s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/c98e6f8551cd45f18daaaa4b3db74085 : SUCCESS in 1h 10m 46s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/fe3cc80452f64ee29a825ac85cbbf72b : SUCCESS in 1h 14m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/105fe827073643bfbea2c8334c59a0e5 : SUCCESS in 27m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3ed243900ec4416816eeef2e3fcf29c : FAILURE in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a439492f5a4b1f4dc302178d5bc8840a011461d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-17 15:11:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"8f5a94174ad286f3bb987b0b00d2bfbf0ac9d64c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-04-17 21:54:39.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5ee2d369ba999781eba0b2e78e924beb42536cf1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-04-21 10:59:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20c5521b07572616f9c3f8bb9e3bf3fbaf93682a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-05-27 13:28:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6307f9160032395bde93a9cee121caeaf11450fe","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-05-29 12:38:06.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d62592b1e88db2e168fc2bbae768f2a12ee93820","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-30 00:09:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"69ffc363e71d8d1f2a66f341f15a785442b584a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:10:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fbc0ca2901a8551808f57ba546cce17aa125c9a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 01:46:53.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d661367f3dc04800b69a2faa67ab3c3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c5e33ac3c645669051d5e7ec855d7f : SUCCESS in 3m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9f54364733fc4f7da5a29192766319d9 : FAILURE in 1h 05m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8e4f7a28039942609954838a2d952348 : SUCCESS in 32m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1769e3bae84a4b9a8e4704a552312a06 : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":7},{"id":"410eecad9c9f33afc9260c22d4925bdf8ed80ced","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-30 05:10:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d5326498df8e3485569117522fe1efd376f6390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 07:44:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f1b66b945e74efc837511a1848bb690\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a4f2930f8bc4d99b0762b8eea9b55a9 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e7d7a52127f748a18ba17d493127cc21 : SUCCESS in 1h 14m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fea4f43c0eab4419abb82927e1b66c42 : FAILURE in 2h 28m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/214b1c4773e44018b422cb854d4df675 : FAILURE in 55m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/57bb2b0a02d1448b96c9f081d34631b9 : SUCCESS in 1h 13m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/033e0ddf9b064eb784880c7ff4b7b073 : SUCCESS in 1h 24m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd726c913e9e40a4b8d8c1862fdf283b : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/14adeadf298d4a869aa8f245ea60636e : SUCCESS in 34m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e738b26cfbdb4e119e87c806ed93a737 : SUCCESS in 36m 15s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/c3e4a3f6610046b1869a8be4e24a5c0b : SUCCESS in 1h 15m 14s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/ecea6f99e95f4b9fb89dd62e81b3fbf1 : FAILURE in 51m 58s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a28048329c484b5c93bb3e9f2ba4abc0 : SUCCESS in 1h 09m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/562586b514874345a5a4b438c9b0d8cd : SUCCESS in 28m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1d0150811bfe49e286a4002207534665 : FAILURE in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e5f24905e3d8f380e63467935df29ce61632b91","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-30 18:19:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7c4c88d40e2de53db89201c5240ec830ffca67e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 20:09:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b63b8a08434bc3a2488c11f4156c71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49619dfd64fc412fa6a9bf954ccf2df4 : SUCCESS in 2m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c909f6b95e54ddd9f7b1b878d04d751 : SUCCESS in 57m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/952e2545ce054a7087f8c6eda6fc1b53 : SUCCESS in 1h 48m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4109fdad12049b0bde37c5520fb9bca : FAILURE in 42m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8b2b69691e864941afacb0b5f97342d0 : SUCCESS in 49m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f4b2632dbbb44e319c05b63ed9806a5a : SUCCESS in 1h 08m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44285722d7454e1981e17e4df88f9f75 : SUCCESS in 1h 25m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe44e256dd034f5db025c138685d106b : SUCCESS in 27m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/77bdc1ba4e914c06a6a27f5cba235e50 : SUCCESS in 34m 31s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/d5d108507c5b4277986ef1f7274ec9ef : SUCCESS in 1h 03m 54s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/89b4af891ef24363a98a011169b80cfd : SUCCESS in 1h 12m 57s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/87960e671b0e4aeeb31468dc405702c9 : SUCCESS in 57m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fb03a1bd8da346a28bc873d9cc1b529e : SUCCESS in 30m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/60e135e5adac47da886e410c0cebaf8e : FAILURE in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"06dedb19c6c55c60848a1db5bccb6475c5a8c9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 20:09:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"246e8f483886af1b4c80889fdde4bbb472811b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:10:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf95985082d34126a4e52357e56be40c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50ee91801d3412e99c77cf67f15eb46 : SUCCESS in 2m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a31aae39e1cb4f928fc4c15e83a702cf : SUCCESS in 59m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/685decdd904342f9ba188ea4aa5fa359 : SUCCESS in 25m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/de0c1fd18c744fe5b54fa7cefaba33f0 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":7},{"id":"6498fbe6abf740eee07857fb4dcb1cf5b063b204","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:10:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4deec69f642647630c9b06eebc95ba6ea39c51bd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-30 23:10:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"94c11fc8800720d87294283f25553113c6c7e4df","revisions":{"9c853b59093f00ae3a5df2abca6e675108a344cd":{"kind":"REWORK","_number":1,"created":"2024-06-17 14:32:17.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/1"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c853b59093f00ae3a5df2abca6e675108a344cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c853b59093f00ae3a5df2abca6e675108a344cd"}]},"branch":"refs/heads/master"},"cf3e0406e758ffc2235711d10a549d3d07d235d6":{"kind":"REWORK","_number":2,"created":"2024-06-17 17:10:36.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/2"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 17:10:26.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf3e0406e758ffc2235711d10a549d3d07d235d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf3e0406e758ffc2235711d10a549d3d07d235d6"}]},"branch":"refs/heads/master"},"8c5ae01f379b9fd1d199166751396b5af9494a39":{"kind":"REWORK","_number":3,"created":"2025-02-13 13:50:15.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/3"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-02-13 13:50:13.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c5ae01f379b9fd1d199166751396b5af9494a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c5ae01f379b9fd1d199166751396b5af9494a39"}]},"branch":"refs/heads/master"},"5c196fc8af8d473be4cc289ecb0c7690a51b50ae":{"kind":"REWORK","_number":4,"created":"2025-04-03 21:21:58.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/4"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-04-03 21:21:49.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c196fc8af8d473be4cc289ecb0c7690a51b50ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c196fc8af8d473be4cc289ecb0c7690a51b50ae"}]},"branch":"refs/heads/master"},"04e9e0f95bd4741ce78144460aac500356892bbe":{"kind":"REWORK","_number":5,"created":"2025-04-16 19:33:13.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/5"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-04-16 19:32:49.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04e9e0f95bd4741ce78144460aac500356892bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04e9e0f95bd4741ce78144460aac500356892bbe"}]},"branch":"refs/heads/master"},"95fddb0e9fc044fecc5244d9ab90db95997e318b":{"kind":"REWORK","_number":6,"created":"2025-04-16 22:40:22.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/6"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-04-16 22:40:22.000000000","tz":0},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/95fddb0e9fc044fecc5244d9ab90db95997e318b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/95fddb0e9fc044fecc5244d9ab90db95997e318b"}]},"branch":"refs/heads/master"},"94c11fc8800720d87294283f25553113c6c7e4df":{"kind":"REWORK","_number":7,"created":"2025-04-17 13:01:30.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/17/922117/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/922117/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/922117/7"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 14:27:24.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-04-17 13:01:23.000000000","tz":-180},"subject":"Add CEPHFS filesystem metadata verifications","message":"Add CEPHFS filesystem metadata verifications\n\nRecently, we have implemented a new feature that allows the share\ndrivers to set metadata to the shares on the creation [1].\n\nThis feature allowed the Ceph driver to report the filesystem name\nwhere the share was created (only when using the CEPHFS protocol),\nmaking it easier for users to get this data before they mount\ntheir shares.\n\nThis change adds a tempest test that will check if such config\nis present in the shares.\n\n[1] https://review.opendev.org/c/openstack/manila/+/909175\n[2] https://review.opendev.org/c/openstack/manila/+/909916\n\nChange-Id: I2ecef30de76c40b98accb5ce04b6c9eefdf7fea2\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/94c11fc8800720d87294283f25553113c6c7e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/94c11fc8800720d87294283f25553113c6c7e4df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
