)]}'
{"id":"openstack%2Fmanila-tempest-plugin~922143","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iceeb41ff8c34fe86f2825089140801f983a96a85","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-06-17 22:26:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceeb41ff8c34fe86f2825089140801f983a96a85","subject":"Temporarily remove neutron dynamic routing","status":"MERGED","created":"2024-06-17 18:15:17.000000000","updated":"2024-06-18 00:20:32.000000000","submitted":"2024-06-17 22:26:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922143","meta_rev_id":"168ea094046c577270d95c32a321c5a42a37acf2","_number":922143,"virtual_id_number":922143,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-17 22:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-06-17 18:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2024-06-17 18:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 18:46:34.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 18:48:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-06-17 21:19:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-17 21:25:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-17 21:45:33.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"ff72c1f998684125694e701af7835e7583b3c05f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-17 18:15:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bed69609731aa2c5237e2fc892ff019b39fd4a2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 18:46:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_PowerFlex_Manila_dhss_false/c17413a/DellEMC_PowerFlex_Manila_dhss_false/1183 : SUCCESS in 31m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"633c1108522760fe29d8f8403d4d24d6b1243cca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-06-17 18:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39ebda357d39db462c2fc7d054126510318f695d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-17 21:19:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_Unity_Manila_dhss_true/5b3fa43/DellEMC_Unity_Manila_dhss_true/1327 : SUCCESS in 2h 05m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_Unity_Manila_dhss_false/e495e6e/DellEMC_Unity_Manila_dhss_false/1426 : SUCCESS in 50m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2c640f1a60bb01e561ddcb546e6bae7e2cfbe22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 21:25:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3458015b80df4edbbc54c65f59eb7b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c79ddfc76946039bcdf9dc3e34a89a : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/04addb7a5b3843ed8595ddf458a38c9b : SUCCESS in 1h 13m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d6ee6fac18cc4705997e562b1f3f819e : FAILURE in 3h 01m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/933872ec22ac4621a573cce7608962bd : FAILURE in 1h 32m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8c928a12d5624419a09ca4660761ade3 : SUCCESS in 1h 24m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/01f69ff0f94c43e0a8b37b06d0143387 : SUCCESS in 1h 28m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f0ee94b598fb4b9cadf76f8fb5a1fb10 : SUCCESS in 1h 01m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0189f99fc38640dbb33405dd9eab4b43 : SUCCESS in 29m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af083ceb26fb45b49206461dc508fe26 : SUCCESS in 41m 50s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/15a713f509904b048491b868103a367c : SUCCESS in 1h 15m 55s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/44a5f110d40240bdb017174d4fa7aeeb : SUCCESS in 51m 57s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/30a1b49d47cb4047b51da29a21d5106f : SUCCESS in 1h 01m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9e6969a0c4214fe9a54e3a3c9c6dcacc : SUCCESS in 31m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/37a4e979e3cc4d0796461088011af120 : FAILURE in 47m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6d664904b6184b978509de84a9350762 : FAILURE in 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26ab50321f0358b2e1ac36d0f752b2fbef87c407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 21:25:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"964af1e83aaa4d7b783d160b7c99ca7043f68d0c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-17 21:45:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/922143/1/upstream-check/manila-tempest-plugin-ontap-dhss/3ba03bb/ : RETRY_LIMIT in 8m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/922143/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/99c7a44/ : RETRY_LIMIT in 11m 58s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b60b57e33cf0dd6d65c7f661464c0316fa1edcf0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 21:56:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_PowerFlex_Manila_dhss_false/4ad6bae/DellEMC_PowerFlex_Manila_dhss_false/1184 : SUCCESS in 31m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"acf17d72aed55dc17da34311379921a130eabdba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 22:26:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d947bfb6ad4c55adf4da0ce2f44c4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eb1fb2a42346439e8b004b55392ed4 : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c9b80fdd1a4e46beadc4e8195b08b106 : SUCCESS in 59m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/37c8433a53c445d6b3fef8d7b202c3e2 : SUCCESS in 30m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7ee994b1cd0b4602bf1c39a95eaaf3d7 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dbe06a5687eeb3e29fa1211a2b3e3fb9c0e1a6bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 22:26:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"168ea094046c577270d95c32a321c5a42a37acf2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-18 00:20:32.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_Unity_Manila_dhss_true/aea7089/DellEMC_Unity_Manila_dhss_true/1328 : SUCCESS in 2h 54m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/922143/1/check/DellEMC_Unity_Manila_dhss_false/20aeb52/DellEMC_Unity_Manila_dhss_false/1427 : SUCCESS in 50m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7928e6d273e530110ccb53518c5a5a46011bea0","revisions":{"f7928e6d273e530110ccb53518c5a5a46011bea0":{"kind":"REWORK","_number":1,"created":"2024-06-17 18:15:17.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/43/922143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/922143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/922143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/922143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/922143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/922143/1"}}},"commit":{"parents":[{"commit":"c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c","subject":"Merge \"[CI] Fix cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6e0efeb6bf074d8035f1ee0c2d2c28a06e0885c"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 18:12:18.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-06-17 18:15:09.000000000","tz":-180},"subject":"Temporarily remove neutron dynamic routing","message":"Temporarily remove neutron dynamic routing\n\nNeutron dynamic routing can be temporarily disabled, as it is not\nbeing used until a known issue gets fixed [1].\n\n[1] https://bugs.launchpad.net/manila/+bug/1940324\n\nRelated-Bug: #1940324\nChange-Id: Iceeb41ff8c34fe86f2825089140801f983a96a85\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7928e6d273e530110ccb53518c5a5a46011bea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7928e6d273e530110ccb53518c5a5a46011bea0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
