)]}'
{"id":"openstack%2Fmanila-tempest-plugin~922867","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I777e8137c5a2b8f2a03667faf413ae65b1170d48","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-07-15 16:06:51.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-07-09 23:41:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I777e8137c5a2b8f2a03667faf413ae65b1170d48","subject":"Reduce test resource consumption cephfs-nfs","status":"MERGED","created":"2024-06-26 19:06:33.000000000","updated":"2024-07-15 16:06:51.000000000","submitted":"2024-07-15 16:06:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922867","meta_rev_id":"9bcb69218f76f719a3cd393e2bee5ff1991251e7","_number":922867,"virtual_id_number":922867,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-15 16:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-07-09 23:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-07-15 14:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-07-15 14:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 19:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-26 19:37:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 21:40:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-26 22:45:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-07-10 01:57:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-15 14:42:04.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"d344449dd4d5ae7d15c1c3fd488ae0efda63ec92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-06-26 19:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f83f3ea3ecc261ffd81b2aabc1ead67483bbe700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 19:06:55.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  while parsing a quoted scalar\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 496, column 30\n  found unknown escape character\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 498, column 43","accounts_in_message":[],"_revision_number":1},{"id":"c9c85aed616b1e1e37262c72bdad8c54ca52b7f8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-26 19:37:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/1/check/DellEMC_PowerFlex_Manila_dhss_false/0bf4484/DellEMC_PowerFlex_Manila_dhss_false/1224 : SUCCESS in 31m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4219d7072d0ae9f0af3b2249148a561672bc8f66","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-26 21:40:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/922867/1/check/DellEMC_Unity_Manila_dhss_true/88cdaab/DellEMC_Unity_Manila_dhss_true/1357 : SUCCESS in 2h 33m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/1/check/DellEMC_Unity_Manila_dhss_false/e4cf1f6/DellEMC_Unity_Manila_dhss_false/1456 : SUCCESS in 1h 00m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7abe5ed8351743834d1f20a906fdd8f5889b505d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-06-26 22:45:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7d28cde5a15f7e92030e4fe9012da34311d7cbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-07-09 20:23:20.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ff316b63bbb952c631128dec44f44ddf7249fdd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-09 20:51:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_PowerFlex_Manila_dhss_false/8643e1f/DellEMC_PowerFlex_Manila_dhss_false/1266 : FAILURE in 28m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a11fa52a09780672ca369b8b54f5ee03bcc9b533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 22:32:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f763c6d5b7e347e4b44e094977b5c210\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f20cfff06cc4575b25793777490a362 : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/109b4af44f8545168cb8d879a4cb3451 : SUCCESS in 1h 06m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/27b9821f9dbd426e98d84f9d014b9ad8 : SUCCESS in 2h 03m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ba755805ec0441e4ad9f6d74ffb847b7 : FAILURE in 57m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ce2cfb9724a444488c3ccb1a16a4a8fd : SUCCESS in 1h 02m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/15d27dfa1bcd412e81b6bbeac41df6c0 : SUCCESS in 57m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d6f72998805c4534b0a954092e0a4a38 : FAILURE in 1h 07m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/50d8305b91024cae93973eab0ab0febf : SUCCESS in 28m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1fbb7898a61a4f4fb149788109190975 : FAILURE in 30m 55s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/9ec974c478fa45e6b2e5fac4912c2c2e : SUCCESS in 1h 09m 49s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/0e7d68da507644d6baccd32ff47bab84 : SUCCESS in 1h 06m 44s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/b8c24f7e43f9409aa7f551afba12cd2d : SUCCESS in 1h 01m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/137e3983470b42d29a58e91cb0062032 : SUCCESS in 31m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/380da670019b4fbaade850020502160d : FAILURE in 33m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/66a24cef6fa8434d9a7a7c7c3bd9a637 : FAILURE in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18b61b603351a6dba8e94ac16d092a83441537b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-09 23:00:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_Unity_Manila_dhss_true/23d3f2a/DellEMC_Unity_Manila_dhss_true/1393 : SUCCESS in 2h 36m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_Unity_Manila_dhss_false/fc9df04/DellEMC_Unity_Manila_dhss_false/1493 : SUCCESS in 53m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d78f25c2f1fb03180914816f00f203ad2b08fad8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-09 23:41:22.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53173b68cbdc455ff3bb7b6d134c91a72464f300","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-09 23:41:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"891d8f5e4e33a68ac1edd5d09f1a094d845fe31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 01:27:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6333559ca98543299cf8d30041043ea8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c4af8efb6a4adab1152b823bb8c85b : SUCCESS in 4m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7cf49b5f24264373b4cd52dcc3a680db : SUCCESS in 57m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f9dc2b7b425413daa429498090326c7 : SUCCESS in 1h 43m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6ccf3e5f645040ed86193adfce9ce283 : FAILURE in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/efcc4f61ecd24446aa94c2e6da421f14 : FAILURE in 58m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/31feb7a1718142f690a46e3e0cc1187f : FAILURE in 1h 44m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dad740bba149480da794a5e78e354984 : FAILURE in 1h 06m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/388696b6798f4579977c560ec447d6f3 : SUCCESS in 28m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc065c2fdc494f32a57e23de1f6efe25 : SUCCESS in 30m 50s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/16436cf134514b2da6bc89d7dfbe02f3 : SUCCESS in 1h 05m 55s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/ef3b4bfaf5df416bb2be0c0de127e43c : SUCCESS in 1h 20m 41s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/3e1be1fbb5b34e16b4262cfa586c1b12 : SUCCESS in 51m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/02f335a7fb924055ae959c1700c646b8 : SUCCESS in 23m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e3cbdbce34304de09d25cc4919f21548 : FAILURE in 43m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/593380957127414ea048e207f83da1ee : FAILURE in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40be19bcf17b1e6522769d7c3f63b49e093a5326","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-10 01:56:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_PowerFlex_Manila_dhss_false/abbb84c/DellEMC_PowerFlex_Manila_dhss_false/1269 : FAILURE in 29m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5a410f9e06f2a523bf48507f819349b527119ce4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-10 01:57:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/922867/2/upstream-check/manila-tempest-plugin-ontap-dhss/8428bd4/ : RETRY_LIMIT in 8m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/922867/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/3d7989d/ : RETRY_LIMIT in 12m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e39691cac79b676480428b7747ba78fbf11865f6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-10 05:06:04.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_Unity_Manila_dhss_true/4e662ef/DellEMC_Unity_Manila_dhss_true/1396 : SUCCESS in 1h 58m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/922867/2/check/DellEMC_Unity_Manila_dhss_false/a5a10f0/DellEMC_Unity_Manila_dhss_false/1496 : SUCCESS in 59m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c8b45bbd5c3a0a6611238f038fbb8ad99c94dcfd","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-07-15 14:42:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60995debbf963c4ddb962d09e8bdf60188b211bc","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-07-15 14:42:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4cc808be9e9355cdda959d778d72dd16ec1dad55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 14:51:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"767b5f1a3299ccc446da7bb2fedbe8057acf53dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 16:06:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/620b245249d24059adb08c47735810dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fc0b7458b1473395ffad1197db6116 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0cec707d2b2a409d90c1633de8e3076f : SUCCESS in 1h 09m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e00fd7f792e142d5863720e1f5826fc9 : SUCCESS in 39m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd10fe6f4d554bd8853b6bc356b7a062 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9bcb69218f76f719a3cd393e2bee5ff1991251e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 16:06:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b615b26716ce922bb0b7f48a0476e28f9ae8a071","revisions":{"85f747cb02deb9055852eb18e0b92dc9c5e22e7f":{"kind":"REWORK","_number":1,"created":"2024-06-26 19:06:33.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/67/922867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/67/922867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/1"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-06-26 19:02:07.000000000","tz":-240},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-06-26 19:05:17.000000000","tz":-240},"subject":"Reduce test resource consumption cephfs-nfs","message":"Reduce test resource consumption cephfs-nfs\n\nReduce test concurrency to 1, and limit the tests to only those\ntagged with \u0027backend\u0027 to hopefully reduce the resouce consumption\ncausing timeouts and flakiness in cephfs-nfs standalone jobs.\n\nChange-Id: I777e8137c5a2b8f2a03667faf413ae65b1170d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85f747cb02deb9055852eb18e0b92dc9c5e22e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85f747cb02deb9055852eb18e0b92dc9c5e22e7f"}]},"branch":"refs/heads/master"},"b615b26716ce922bb0b7f48a0476e28f9ae8a071":{"kind":"REWORK","_number":2,"created":"2024-07-09 20:23:20.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/67/922867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/67/922867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/922867/2"}}},"commit":{"parents":[{"commit":"cd451435c1d79dd50145fc649aabb44bdc1e08c1","subject":"Merge \"Fix racy DHSS\u003dTrue share groups test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd451435c1d79dd50145fc649aabb44bdc1e08c1"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-06-26 19:02:07.000000000","tz":-240},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-07-09 20:23:09.000000000","tz":0},"subject":"Reduce test resource consumption cephfs-nfs","message":"Reduce test resource consumption cephfs-nfs\n\nReduce test concurrency to 1, and limit the tests to only those\ntagged with \u0027backend\u0027 to hopefully reduce the resouce consumption\ncausing timeouts and flakiness in cephfs-nfs standalone jobs.\n\nChange-Id: I777e8137c5a2b8f2a03667faf413ae65b1170d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b615b26716ce922bb0b7f48a0476e28f9ae8a071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b615b26716ce922bb0b7f48a0476e28f9ae8a071"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
