)]}'
{"id":"openstack%2Fmanila-tempest-plugin~926563","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I64a82f0909b48bdfabdf47137cf1a3163d623257","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cmount_path","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-10-24 21:12:14.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-24 21:12:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I64a82f0909b48bdfabdf47137cf1a3163d623257","subject":"[DNM] Test standalone job","status":"ABANDONED","created":"2024-08-19 17:38:13.000000000","updated":"2024-10-24 21:12:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"280f5911b605ebc007adaee2803af38330b8debc","_number":926563,"virtual_id_number":926563,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-21 20:25:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 17:47:01.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-19 18:14:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-19 19:09:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 19:29:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"707ac62811d0e6830bb59a44af94e75d3edad82e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-08-19 17:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2f092ab8698baeaf40bcbb38fe0423bb97790d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 17:47:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_Unity_Manila_dhss_true/4eb3c56/DellEMC_Unity_Manila_dhss_true/1555 : FAILURE in 8m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_Unity_Manila_dhss_false/68459b5/DellEMC_Unity_Manila_dhss_false/1666 : FAILURE in 8m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"49ac72c741960ed8ae18237c05fbd289d4dfd867","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 18:14:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_PowerFlex_Manila_dhss_false/975ba52/DellEMC_PowerFlex_Manila_dhss_false/1487 : SUCCESS in 31m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"91f404f7309f1243ee22605c6efbd41eeeaa711e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 19:09:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbaf8c0331f94dc38cf2507a0c7c78a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54813c53b6f248f6b9ecad389e31dd94 : SUCCESS in 2m 35s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d6cc78dea49e4404a40e82ad6d1a15aa : SUCCESS in 1h 21m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8037f1911b7849299f059d5cf8cdf052 : SUCCESS in 35m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7e8e40b6ebe0409abf121e5fd00702ae : SUCCESS in 44m 36s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/80195cf463164f169f741f619920ddf0 : SUCCESS in 55m 52s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/a67ebee25e44411d89d03b3f4e6b7546 : SUCCESS in 1h 08m 45s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/1998c73fe4df4be3a64f5f299160489e : SUCCESS in 1h 23m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0345f68107314957a7f5351389ed01f5 : SUCCESS in 30m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1a1b1bc5307e4784a947f8c4b12861b2 : FAILURE in 34m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c676f342e9614c6f9218c0ae9f427438 : FAILURE in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4f562d3cb1f5f61b950e518a9c4d79c325ef12","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 19:20:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_Unity_Manila_dhss_true/51531ba/DellEMC_Unity_Manila_dhss_true/1558 : FAILURE in 10m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_Unity_Manila_dhss_false/9db025c/DellEMC_Unity_Manila_dhss_false/1669 : FAILURE in 10m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9f04cb3ae0383e1b2ced5166a031ff88a344c55","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-19 19:29:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/926563/1/upstream-check/manila-tempest-plugin-ontap-dhss/8e4d912/ : RETRY_LIMIT in 8m 24s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/926563/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/dd135ef/ : RETRY_LIMIT in 3m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"563e701faa812327a734a00365ec1c90fd10aadc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 20:11:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/926563/1/check/DellEMC_PowerFlex_Manila_dhss_false/71d1cd4/DellEMC_PowerFlex_Manila_dhss_false/1490 : SUCCESS in 31m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1940e1430c49381982f19e89ec0d4482956ba579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-21 19:12:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"09aae3ecd88d84005aaa15229e88e1a2dbd483e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 20:25:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c64d1abbb40344128547a528c552eb9f\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/70c4082fc9c74f289cf3d4a13b6d52e1 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5ad89dba9570277bef7124f2cc4601e394e81da1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 20:38:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/926563/2/upstream-check/manila-tempest-plugin-ontap-dhss/6e593c7/ : RETRY_LIMIT in 9m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/926563/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/0ceb8bf/ : RETRY_LIMIT in 11m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c22304f84c4ebb24ec242448ec567cdf749d587b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-22 02:19:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/926563/2/check/DellEMC_Unity_Manila_dhss_true/8db2b6a/DellEMC_Unity_Manila_dhss_true/1575 : SUCCESS in 3h 12m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/926563/2/check/DellEMC_Unity_Manila_dhss_false/656230e/DellEMC_Unity_Manila_dhss_false/1686 : SUCCESS in 1h 14m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"280f5911b605ebc007adaee2803af38330b8debc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-24 21:12:14.000000000","message":"Abandoned\n\nNo longer needed i think; please restore if you\u0027d like to continue testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d05b97e12a7bbec886b5584a862b1eb282210113","revisions":{"b0666de3063c5ce9d1424348b1222729bfc6cae6":{"kind":"REWORK","_number":1,"created":"2024-08-19 17:38:13.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/63/926563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/926563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/1"}}},"commit":{"parents":[{"commit":"b758fb49cc8c9d56fa83f910f3462a845eec1b1b","subject":"Skip test_share_extend in CephFS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b758fb49cc8c9d56fa83f910f3462a845eec1b1b"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-08-19 17:25:38.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-08-19 17:31:20.000000000","tz":0},"subject":"[DNM] Test standalone job","message":"[DNM] Test standalone job\n\nFor running cephfs standalone job only.\n\nChange-Id: I64a82f0909b48bdfabdf47137cf1a3163d623257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0666de3063c5ce9d1424348b1222729bfc6cae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0666de3063c5ce9d1424348b1222729bfc6cae6"}]},"branch":"refs/heads/master"},"d05b97e12a7bbec886b5584a862b1eb282210113":{"kind":"REWORK","_number":2,"created":"2024-08-21 19:12:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/63/926563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/926563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/926563/2"}}},"commit":{"parents":[{"commit":"b758fb49cc8c9d56fa83f910f3462a845eec1b1b","subject":"Skip test_share_extend in CephFS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b758fb49cc8c9d56fa83f910f3462a845eec1b1b"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-08-19 17:25:38.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-08-21 19:12:14.000000000","tz":0},"subject":"[DNM] Test standalone job","message":"[DNM] Test standalone job\n\nFor running cephfs standalone job only.\n\nChange-Id: I64a82f0909b48bdfabdf47137cf1a3163d623257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d05b97e12a7bbec886b5584a862b1eb282210113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d05b97e12a7bbec886b5584a862b1eb282210113"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
